)]}'
{"id":"openstack%2Foctavia~752864","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I477d2d12c64c1e38e3917333e5d47f612cab93b0","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I477d2d12c64c1e38e3917333e5d47f612cab93b0","subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","status":"MERGED","created":"2020-09-21 06:14:21.000000000","updated":"2021-02-23 11:40:52.000000000","submitted":"2021-02-23 11:39:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752864","meta_rev_id":"540bfc824eb1644324f5df341a4f218424c84494","_number":752864,"virtual_id_number":752864,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-10-22 08:32:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 11:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":2,"date":"2021-01-28 10:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-14 21:49:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-12-03 13:55:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":1,"date":"2021-01-28 10:33:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-09-21 06:14:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:14:21.000000000","updated_by":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"reviewer":{"_account_id":30657,"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","username":"ppeereb"},"state":"REVIEWER"},{"updated":"2020-10-22 11:13:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 13:55:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-14 21:49:30.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-28 10:33:51.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"9e138a0fcb43567630b0ab55268842abbd9f0e01","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"509830a5a0ca440ac4c516d77719912e5e913290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:32:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e01895686ba469cbe249b57e06164d9 : SUCCESS in 8m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8fac38244b84b2da11c51833840f56a : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9dcfe09bec475286efdcbfeb5247ea : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/345d367527c74373b54bc6772bb31f8c : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a25d5fd144544faa4690efedb2a5250 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dacdc20006ba4220b18bf552bc81cfc0 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3ea0553e664276a94c9ef7f806dedf : SUCCESS in 23m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb3c5dc7f4da481a9150459ed0d888b3 : SUCCESS in 5m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/75905d7aef4e42998ecc368850f0e645 : SUCCESS in 10m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a40cdf55101d46e5b7294459a127df55 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ff7d505974540a2aaf0beb4a699c9e7 : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06dc386903d0407f9184fdacf0a3a718 : SUCCESS in 1h 39m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dc9686886a8d49b3beea61d3a9e49252 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e8a341045b44b6a9bdbd08bf4efccca : SUCCESS in 1h 53m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87f5294f7c6f4ae1a7f7c4e7567abe7a : FAILURE in 1h 24m 47s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/346e9472792f4929993fd8b77153c25a : FAILURE in 15m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bbbf09c8df5a412292ceeb539e087c0e : FAILURE in 38m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/08d901cf85294c2a9000d4e86974ad9f : FAILURE in 43m 08s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69b7f5e15dcc4f448c807bcaca3ebf09 : SUCCESS in 1h 24m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fa2dd66d712340d08b587d8c566ffe46 : SUCCESS in 41m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e28e75f71c054adbbdf10f3444ae6074 : SUCCESS in 1h 47m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/055402f9c5b94dca8943bfa898e93cd2 : SUCCESS in 38m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/88e21ae84a144d17baf9d47d9da293af : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9d2dfbffd8264ed3bbd04d099dad799d : FAILURE in 12m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4364f8e13533a13df975b3abb34b1056d8f657e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 11:51:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c49c4d31d3f3f4916930bbe4dced632a8c4161bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 18:34:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efe795af612b4862a24486e08f8d5ac9 : SUCCESS in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32f6e225909740c792560ef6ba34c3fa : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ccc4908451345be9c6b8fc59113962f : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f75f63cbc534a4084d003bdf3270fc1 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d03ab946c0f343bc81f4a86132c01ee7 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5ac99beece840baa6fb9f0a6db5a05c : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7f65086ac6642d396727ed44c288281 : SUCCESS in 17m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/35ae5459872845659a6b691823711165 : SUCCESS in 5m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8214420c94814111b7344d237abbd1a0 : SUCCESS in 7m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a93d078c3d314f4f824ab4f1946c07b4 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5eb9bff388554ef1813fbd72f1e91791 : SUCCESS in 7m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06b874f366db4dfdbd853ee4c488fb79 : SUCCESS in 1h 37m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/13b5e842c4b4447698ab0ec95b9496ad : SUCCESS in 1h 55m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71ef98bb93604f4aa32a1b06d05ccdf3 : SUCCESS in 2h 19m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4ab6c4605f924728af6f4d62125f79e6 : FAILURE in 2h 01m 31s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/76ae5ff83292480dbb54ef2e878c7341 : FAILURE in 12m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d951afb08d50438b964330b46989e304 : FAILURE in 28m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1c442539525f482d859bc7c828cd8159 : FAILURE in 47m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a674e901f6804a47ba3eae67298519c2 : SUCCESS in 1h 16m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72f2167ce5cb445b99e1883324a7a12d : SUCCESS in 43m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/11cc88f26e96469987b61d27a60741db : SUCCESS in 1h 54m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/344bdcc5840443c8ad4bbe93eae4b9e2 : SUCCESS in 41m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e8ab11a82dcb4c119dceabb6e8e2f669 : SUCCESS in 41m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6bb405ab516140f68e1f5e3afd859db0 : FAILURE in 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4be02170014cce17f7c19152b14f5d8e888d02d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:27:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"30aa80df607d63966e6a750c5612db37f957307c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:56:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91f35c11eeee47298fb7d6e66bc74a01 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d540db365e134b8f958296e234174bc4 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df02513c471e4c65b64fba6c6252ba72 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2052727ba49d4c3e860f2de174868fc7 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547bf732c7104524be7755edb674551d : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab188b2fa9e54195b0d82de66a53cbf7 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26feba25e4249d3a565ef5084ec5018 : SUCCESS in 17m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7c54e0a705bd499ba6547be4da5ed3ba : SUCCESS in 6m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/23e272e81af24084b13e816e3514fdfd : SUCCESS in 7m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1e8662740859408c917ff3031d8c191b : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d426a3370106426194c9e0e264627839 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2117ecac30e14ca7aa473a11d52fbdd2 : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/26c14ec5dacd40bd85a0a9d7d3b3ad76 : SUCCESS in 1h 59m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dd7570d6de4469cb10489302606db47 : SUCCESS in 1h 41m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6ec7f5b1b14e42738bdafc8f53aee9ce : FAILURE in 1h 39m 44s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a3929d98427e4d89bccbf113bbdbab98 : FAILURE in 11m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/da765b9df502484b8a86de0c5585c6b5 : FAILURE in 28m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a72aa1a426e046c2a3bf4e7c85398907 : FAILURE in 1h 01m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/22bc16862ba643f59aa2ecdad42d94af : FAILURE in 45m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2b3341fb4924d91b7da4805c13f6496 : SUCCESS in 41m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c2a13f50961340bc98f7ac3c005c3444 : SUCCESS in 2h 22m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35eedcc2d3664e4cae4942874836bcb8 : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d62de84d61b49d29aefc89c73cfd93f : SUCCESS in 42m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/295d8d6c66304c6fbac9481162611802 : FAILURE in 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"302bcf3547e01334d6cb16648998455a522c96bc","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:32:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1e74b4296946576cec2944e0aa0bd8ddaf03e52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:13:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4833b3ce3bfd4de99c1104a125acb69a : SUCCESS in 9m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d77b7084d547d4a6f70afcfab9be68 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6eb4acaaa984396b011e70d78d0216b : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54ebf3a3edd943328308cc636bc5be63 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248066cb4c614354b74f5c1876a3b1dc : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6471f0ab5c9435e800aca10954fcf70 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4153c73e93a4053bfe3dd36add04321 : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e96b1afb3814f3fbfaffe0e2684a0cd : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1947381286e141459036ee17070ee96c : SUCCESS in 9m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f83c734ab7fd4aaf8c821a612a7e30fa : SUCCESS in 10m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07a75c396e004bf79144d484a01041a8 : SUCCESS in 8m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/31fb38d646b6449e80f0e30e757d615c : SUCCESS in 1h 41m 27s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0405e9e9b60e48639067a0bb74f8d8bd : SUCCESS in 1h 40m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ff2060bd04e49fdaff2af0f67870f97 : SUCCESS in 1h 34m 10s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8850eba427404db4a544ddc30a2c809c : FAILURE in 1h 57m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/81ad169e7d654669996aea37fb862335 : FAILURE in 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c4a418f98154fbab8d2b32bbdba011a : FAILURE in 28m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f056591f8dcc44079fe2e49dbf94db89 : FAILURE in 46m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce4d26591a9f476490bd5fc0d1c6881d : SUCCESS in 1h 30m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bef3d0333b944656a6d439ccc62e5bd9 : SUCCESS in 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7bd59bb350bb4434b7b040c928462769 : SUCCESS in 2h 32m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d089a720b0fe483c882638510cd9563f : SUCCESS in 33m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7866226787cc4ec29d26c9ad9a1fb993 : SUCCESS in 39m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/21fe0eebff724f68b5c6af666b037a38 : FAILURE in 14m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e534e5c9489ed7df9291bc9bc7accf86fef8850","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-12-03 13:55:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2f655d45db850744ce381636a1062fdb5f5d35e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:49:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f481fd5ce397666febb71719163e7f0071f1d740","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:33:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0082a6ad39e58c5422b6616c86a606676c50260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:34:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5318a40242e2408092b1242217761dc28f8b6340","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:04:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98ef3d87ef45433489042f5e05670d7a : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d493a26e49448079ab85858daa9efed : SUCCESS in 9m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c75e7c4a57445f78d5e529495143da6 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2edea45e3b6a4a33b7d82925ae2f80ae : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e1a82777eb547169e6b109454fa2e97 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c486d7c479b146a49b07f37bef696ac2 : SUCCESS in 9m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b863092023c490d9bb93a0e4ba58c97 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c8e571e21954879bc7087d15bd4c939 : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8330192a8b9a412aa27a364bffe17169 : CANCELED\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d6d6d77288d1446ebf5bbcc234ffa5c2 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e41109f82f474e1faff674415d7e68ce : FAILURE in 25m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b804d78bf604d4ba42582934fd78790 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"86af23007e0a75e3a2994d093f0a26466ca02e3d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-23 07:46:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3fefb26011a545407af1782b4b66facdefc4fbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:56:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/452d58df32d945cc8190b570863cd6b1 : SUCCESS in 11m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2c226e1302a4a88b1f4242081925fd3 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9473b984c8754e9e9397af17aa1282de : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c76de7bd595428387f2cfd1b7c49865 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d1913d044674dc39b471046f5ddab6d : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/269cc88603824fc6987df234698aecff : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0028e111dc443018d2be6f38c1683f4 : SUCCESS in 10m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36a0db35b6e3451eab1f1e6e345f3442 : SUCCESS in 7m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/517c042557944401b677373a4015c05f : SUCCESS in 11m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/794018640a6146738e8336236ff0366c : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c578aee2be6f4d78a92484806fa154bd : SUCCESS in 8m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/92a2b45e13f04242b7b82d632969b5f7 : SUCCESS in 1h 31m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d9f5e56d0ef042f68abb9e06921f9427 : SUCCESS in 1h 32m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c95eee19b5b44ad1aa1d85e4ec321e23 : SUCCESS in 2h 06m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7cfc9c96f29b435ea451c337cbfc9d99 : SUCCESS in 2h 05m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c21a38a796df410e9133edba020b017a : FAILURE in 11m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40265f17e8ab4e16a224081dc2f5a1ea : FAILURE in 50m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/890a6f95018b4111b31f609e7f63d0cf : FAILURE in 50m 36s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b807db32700f4d47abc8a36de5f84964 : SUCCESS in 44m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b47dacf272df4d1e9c00776cf0765df1 : SUCCESS in 2h 01m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a0f9a5f082f74ef0a58d397f08a8f772 : SUCCESS in 36m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/158b741db0c34aec8e14965a56464dfa : SUCCESS in 39m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/300109e4f4334a289a359530f708f987 : FAILURE in 27m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"459b213529ff0ae68124d4fd3e5e40e27b0c49ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 09:57:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"552335d035527a8609d8caca9fe2c21a2145e200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:39:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed6a8cd3b37c40018927fc04fb2b9b35 : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b030e2f771477bbe11be1743f6647f : SUCCESS in 8m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad6e620920e04dc8bf54924e901de5ff : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d22aa42fa5eb44999d114012e7a22490 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d40b140df20b4ed297501034968b5332 : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ec7c50f0b34586a34afe26a85c875a : SUCCESS in 11m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1fb0ce54ffe54145809d106d04375f92 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb638345fc37434dabfa869fdf994bab : SUCCESS in 9m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00eff0510859452aa109fd1fe756891a : SUCCESS in 1h 17m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/57c769855bfa4c9288341629645936f4 : SUCCESS in 1h 16m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/590742c8c91146358c52c2190c84f5ba : SUCCESS in 1h 40m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3341c2bc309845177838dc4397cab3b917cd8729","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:39:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"540bfc824eb1644324f5df341a4f218424c84494","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:40:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c48e033145f404bbfd287d1e364ab5a : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60c1c48bb5cf5f5d0139823da460c1f875aeada0","revisions":{"60c1c48bb5cf5f5d0139823da460c1f875aeada0":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:14:21.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/64/752864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/752864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/752864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/752864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/752864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/752864/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Paul Peereboom","email":"paulpeereboom@hotmail.com","date":"2019-07-05 13:43:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:14:05.000000000","tz":120},"subject":"Fix user permission for WSGI configuration Task: 35692 Story: 2006172","message":"Fix user permission for WSGI configuration\nTask: 35692\nStory: 2006172\n\nChange-Id: I477d2d12c64c1e38e3917333e5d47f612cab93b0\n(cherry picked from commit 6df1ce8e28eb76cb782dc06f913598cfa2ecc156)\n(cherry picked from commit 966adf0f84ae3c67c91f2e5840fa059bd4995c8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60c1c48bb5cf5f5d0139823da460c1f875aeada0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60c1c48bb5cf5f5d0139823da460c1f875aeada0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
