)]}'
{"id":"openstack%2Foctavia~752865","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ia9d9d844b325296401577f5617d2b89cf1a017a7","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ia9d9d844b325296401577f5617d2b89cf1a017a7","subject":"Add some details on enable_anti_affinity option","status":"MERGED","created":"2020-09-21 06:15:34.000000000","updated":"2020-12-02 06:22:45.000000000","submitted":"2020-12-02 06:20:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752865","meta_rev_id":"65bedea5be5263b998a009ad4e62e5f671c94a9e","_number":752865,"virtual_id_number":752865,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 06:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2020-10-19 12:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-12-02 00:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-12-02 00:17:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 12:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-19 12:27:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-12-02 00:17:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e11160c57c82c9dfe90d47a9347e33b6e335ed6c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0e3cb053d2a1bb4b8cedaae9fe6bd54aeea6758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:22:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c48aead1684a62974d96c582127078 : SUCCESS in 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5042998683a04221bacfadd94e14bfae : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad7584fc78b421490ec877cf3520f3d : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70b90bf723ba467585fe3c0e412622f3 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a637387bfde64cca967af2c6c6a792bd : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33f8c0e34ec4af0b893a0ad5e7f1d7d : SUCCESS in 16m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a7eb541d408f46a9a9fcbd7f3a9105c1 : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/284cb60ce61b4ae995af30b6412fe73f : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc69b705be1a4e13ae8ea75688ad936f : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9025b7efb12e4d7c8bbe2d5dcd66426d : SUCCESS in 1h 36m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5043b99f05a4a6f93730a59c6b3b87f : SUCCESS in 2h 01m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0fbf8446f0db4512b26afcd181639475 : SUCCESS in 43m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a73677725a034285bb0da0bef62e485e : SUCCESS in 45m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4ba01e32e9d748709f67aeac94955da6 : SUCCESS in 58m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2242191774a42de84dbe29202ad6cd8 : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8d1706a2ffa145638a6a0437b203fbae : SUCCESS in 2h 14m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96faad8a2b5c425e9036cc8c38d0e4dc : FAILURE in 48m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45fd7403ed204f3ca07b25d6422caf15 : TIMED_OUT in 2h 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b8fc24c014865447a630b85192d5b13e3066a9f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 12:27:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82642679e7cb93e4cd7b4493bcee174048f65832","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:17:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSure","accounts_in_message":[],"_revision_number":1},{"id":"80a2049a394c94dc5b521e8f395533f857dcb28a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:18:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95535dd6028b516bddddf70373cf456d48d63f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:20:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c7a252c8ffb4f5a9518b85fd4cc0b66 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66306191ae404620b1d372c2d8fadec3 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f12022ece341539c4817e8d21a907b : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6f4e9a16888449b94fc7d8298acdff5 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55f68adbfa4471a9ae9e69ec4baa96e : SUCCESS in 11m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a166ac05e7ca42c187c53d417559367b : SUCCESS in 12m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6557127a6cac4b6caa4fa8fd8ca6c134 : SUCCESS in 1h 17m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5b2300274544c8a87990dac5219b2c0 : SUCCESS in 1h 44m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0dac187cf7df4f73affd9e32a642fcd7 : SUCCESS in 50m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/92b4cad0c4ca482889f7194cb3a0d693 : SUCCESS in 41m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8f5b048d51f4711887cff2396e688da : SUCCESS in 42m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1eccbbb7c8da48a985d0778fac25687f : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"906ebf35256e30de4db76eae3326e6f00179f190","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:20:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65bedea5be5263b998a009ad4e62e5f671c94a9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:22:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d95380f45a42d8b162572ccc4be624 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a1121420f8e60648dd2b6be37af38e8a725a1d9","revisions":{"6a1121420f8e60648dd2b6be37af38e8a725a1d9":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:15:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/65/752865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/752865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/752865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/752865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/752865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/752865/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-23 12:49:19.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:15:22.000000000","tz":120},"subject":"Add some details on enable_anti_affinity option","message":"Add some details on enable_anti_affinity option\n\nExplain that \"enable_anti_affinity\" affects only amphorae in\nACTIVE_STANDBY topology.\n\nChange-Id: Ia9d9d844b325296401577f5617d2b89cf1a017a7\n(cherry picked from commit 259f19ec152fe64777ea9bdb658929e22d59427b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a1121420f8e60648dd2b6be37af38e8a725a1d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a1121420f8e60648dd2b6be37af38e8a725a1d9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
