)]}'
{"id":"openstack%2Foctavia~752866","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ia9d9d844b325296401577f5617d2b89cf1a017a7","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ia9d9d844b325296401577f5617d2b89cf1a017a7","subject":"Add some details on enable_anti_affinity option","status":"MERGED","created":"2020-09-21 06:16:06.000000000","updated":"2021-01-14 01:40:03.000000000","submitted":"2021-01-14 01:38:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752866","meta_rev_id":"60f279dc7cdd8a6d43f678718113558f7024529a","_number":752866,"virtual_id_number":752866,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-11-23 08:06:34.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 01:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-23 10:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-10-19 12:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-10-23 10:49:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 12:28:24.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-23 10:49:51.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-26 11:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 21:09:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"3bdca912a58c201e69d43a0ceff5cce957af664f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d59bc0818e205fcc6610e477c43d82ac6d56395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:31:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69f6f43d9633419fa9eca02b3ea2dcec : SUCCESS in 12m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6045440e9abd49b8b4e1d1cb31a0c318 : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251d49abff9341d8a7dda8fde5032b0a : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac5e4802d123436892dad22451ff270b : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/597d41337d0d4e7b8c5a8f41e61328d4 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/779fe65741dd4507977ceade410c5d5f : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2ba14ba38b491c8c06b5550adcb1bc : SUCCESS in 18m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/21549a4cd5544d64be512fd691e0147a : SUCCESS in 6m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/816960cac83f4ee998bcab69c15048d6 : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/106bf403d59f49feb97f4c0cf039ec13 : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0afcd43b3724ab49f956a82f7fb704b : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21dac2f239b0462aab3160097f16edbe : SUCCESS in 1h 34m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/bf2dbd634ce4471a850726bac5962930 : SUCCESS in 1h 55m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e895a2668484fb48d2f7cfa138e398b : SUCCESS in 2h 19m 42s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/870d897710b241e4a8e06fb0b8e52dc3 : FAILURE in 2h 24m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/79be93ba8b4845d1a1ce15d8a9f7d6cf : FAILURE in 15m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7589f35df5074e648c46d2804af16c85 : FAILURE in 27m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/827d07e452bd4d95a2a9987acb02f1b3 : FAILURE in 44m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6041ee0aac6e48ddaee21fd4f31c5769 : SUCCESS in 1h 17m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/339b60b4e9734ed89ca1fd44128e3c03 : SUCCESS in 42m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bbab336fe5454e5fa7be9e18d3902a66 : SUCCESS in 2h 01m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1b8e69c332b4e3197139cb6e5311777 : SUCCESS in 40m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/71bffa5563734ea9b578d4556d6cdecb : SUCCESS in 1h 23m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/894a5fd2e04145b087a3488d71f806e8 : FAILURE in 13m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9731a94dfd8971ae8b8fb21a3e57c948bcf58a24","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 12:48:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a599f54c2a124e6224eee80eb37c3ad5256e612f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:01:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/449279c70ebf417693917e4139137051 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/429aeb69e0094f15ad08efeaf3b790af : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c86fcae36484aae8dfd75ca7f0af450 : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7eaf59888bdf405a98782ccff78ca38d : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348bbfeca64c43e3b5181df622302ec1 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0e15ffb72f14af4a9f73ea64ad72b7b : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7db1a30b35489fbd21609b9524afac : SUCCESS in 19m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14928a18a80a44bd9803a7b9771329c8 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/718d2d469b854ecaa15124ec1ea4e1a9 : SUCCESS in 9m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b909898ef874464797963ce70beeeabb : SUCCESS in 9m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43448e10d7df481985a480ea0fa8d5f9 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eee7620341d44727b02d1c5fb4d8c824 : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9d207f410c7748fa95f8619ce1938b95 : SUCCESS in 1h 44m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e1e86da5285e4c81850782b8743b6c71 : SUCCESS in 1h 53m 32s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7a0692ea8f344276a10656410e5e0775 : FAILURE in 1h 30m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8940a57001594327b901170439686080 : FAILURE in 14m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66cdd146404440e6a051b109353b6d7d : FAILURE in 57m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/55d50eb94968440eb21207c4e1491fcd : FAILURE in 46m 22s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08c58bcbba2f4ab69fba5c2ed0257cf0 : SUCCESS in 1h 23m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47ad33ededbb4aada87c22997caf035e : SUCCESS in 42m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0815a036e7014c4c886b8ea2c0a7ca90 : SUCCESS in 1h 45m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02c235048df34d938b0937a6391af126 : SUCCESS in 46m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1ff67d014b864576a0874f8d7b3e1849 : SUCCESS in 42m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a204636060a54c4cba5a603489268519 : FAILURE in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b5f185edf4a0374d87135edfa3e9c0c41463d3a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:27:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed3ceb7ecf7d633b7bce8aa4955dbefc08e0902c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:56:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b33063309f94b10bce4e3a4aa5640b0 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c9a087833594b3e9d6e92b7f6b419e7 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/234a7bde17334d6692af76a27c97ee6a : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d35bff307a4d4a74b1f657f8de145bcc : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c22bf97007ac4fd3be3e894f325f5cd0 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c35b28124d3f4309b238f0781fe6f968 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41fbb68a828419e94dbfca0db23cd7e : SUCCESS in 17m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c05b1ec5a5844eb2bef67b77419e0ef9 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bd203e56a18b40ab9a311867b1294c42 : SUCCESS in 6m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/45fa367048ca4883bc9acad625dc6727 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad9f47ab2b7e4defaba3d21119d4cbb6 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2fcdc305d77437ba5a495977a214bed : SUCCESS in 1h 55m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c3bfecaf92f142a5a359f9cbe2b23812 : SUCCESS in 1h 41m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/354a95d97e024fab9f6756e4ebd5c8b6 : SUCCESS in 2h 12m 21s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8220ec67f99349c498063c6beafd77e3 : FAILURE in 2h 25m 01s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/29e5b94967b54910af6150cc6c7a66f0 : FAILURE in 10m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f56a7e1968374517b5ec2b96066b3a0d : FAILURE in 28m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e30e53f04ce4805971b042eec3259fb : FAILURE in 40m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/220bb1ac33524ee7bc03f220546e571c : SUCCESS in 1h 38m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8388503957d4a209b573a69ea46ce07 : SUCCESS in 48m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a524234813954dfb94898186131cfadc : FAILURE in 1h 59m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/221acfaf44b64d6486f4fd5baa6f4e20 : SUCCESS in 39m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a633a775869341d8b073019881c604a8 : SUCCESS in 41m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/96b26d3083a04fca9252501cc5ca1a89 : FAILURE in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3689c180974561dc90e0dbac5dae887a3ad09ea6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 12:28:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6e5bac48013d14caaf25b8ce13939a541d1e467","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:33:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5999748d711236cbabfd281abe58f3bb2dc4707a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:36:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2b6c265db04450faa026084b2ad4da6 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/231c5cfbe886428ba556bf6389a8f87e : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bbbadbda97447778666c75cfbf09713 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/53f6d92c5cfa46dcbf110c00136fba16 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/262d18e245ba4f818d81a33ec03180d9 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55c13bb186834f09b57d2fae2f3a9bea : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc716943405465691313d6bf2193034 : SUCCESS in 13m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5d186c84469d4f6bb1f97428b066197c : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/70f11110ebbf4a6ebe78b8cb1c386748 : SUCCESS in 8m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/328d57c76a0049fdb3d710f9f9ea2508 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74fce9f495414bd7955020b1fd0940a1 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b108d5205b742e888410821715d51b9 : SUCCESS in 1h 49m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2e589940d2314cb4871ce0c30a08be7d : SUCCESS in 2h 09m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e9fbeeae15b4beba74162fad93f2d0a : FAILURE in 31m 25s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9f6d44d278f84683b77e240a5f47eb5d : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a2d0291b02464fcda19fbdd62412b2ff : FAILURE in 17m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f641bd83c5e4e888f1da802e77b15fd : FAILURE in 30m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3cce0c33d0d442d082e888663b1893e2 : FAILURE in 1h 02m 14s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00c64ccbf84d4c6cb69318dd06e700e1 : SUCCESS in 1h 31m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/12627d9d607e4352806c3c0273e1d689 : SUCCESS in 57m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfeb909baea2463780d3bf928d0bb4de : TIMED_OUT in 2h 35m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b421722a6adb422c9fdd0c0cc016e8ab : FAILURE in 32m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/247ff1ab6e0247fb840d2b07ea75e54c : SUCCESS in 1h 00m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/04ed251cdbc84a58a8679cfc98c3887f : FAILURE in 18m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92aec1e743032a6531a5f337ba03e15e784609c7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 11:52:13.000000000","message":"Patch Set 1:\n\nrecheck\n\ninfra issues and -py2 now non-voting","accounts_in_message":[],"_revision_number":1},{"id":"ab0430afd9f2322a42bc56e46f40e2d97615a21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 14:30:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/693362b2340d45c8b6f05c9d7687301a : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b052f6fa85714210a59d054a94e67f3d : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dce17ea324946469cc6021fe98e0b55 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ea7ef7c5e474f0797524fe274fa3d23 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ad8d6f1d929485fa977769465573baa : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7db6a85cead14c94a8a65c65b1fb1c9a : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4369247ecd6548718d874d99a506435a : SUCCESS in 12m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb8e2b22506145f7954d8bb7b1a9dc02 : SUCCESS in 7m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fa4321c32d034e5290a0855daa69b9b8 : SUCCESS in 9m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e1df019a05844b186e35101da9e960e : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/09febe0c196242b5bf81cac593a96825 : SUCCESS in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85ad90dc3fc54eea811fd76673a476c2 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/85b530f86d5745efa86d859bce9e134a : SUCCESS in 1h 50m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a757502a4254d8aab6012d7d239433c : SUCCESS in 2h 15m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d2fde8a537f44bf9a841e18fa7ea16b1 : FAILURE in 2h 30m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0360984f817545a6b5399240b998aba6 : FAILURE in 13m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/32fa3907bd32444db267326c04d0fc90 : FAILURE in 29m 57s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/672f817f2495452d86c9ea4fe6157064 : FAILURE in 48m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/088dcfc3ee054e9e871804ecbee8102f : SUCCESS in 1h 22m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdb2679518914d34a6cfad65636d7564 : SUCCESS in 46m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ab2815a2751a476e98eedc98ce551484 : TIMED_OUT in 2h 34m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce901062c9c344a7836c1aba4cc19d6f : SUCCESS in 45m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/76b37c6f748b4ad68b5bf66a85e205a0 : SUCCESS in 44m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8a42e6de529a44fbbe6b7e1e641d5e92 : FAILURE in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29744afecfcd6a18a5b70c17667372c334656a79","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-23 10:49:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"742d4bf6e4e98516d7917f6476c58fd4eb85f53f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 10:50:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc3707099efc592938414ea3dffb77a8d1bc8de1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 12:50:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0394d8aa50a4282af742e99be9786b2 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44feda776ce4167b56bcd087d6735b4 : SUCCESS in 8m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/15d04ff343934f2ab5de41a45223c6c2 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05b397210464d738ef4e4f5da82ca5b : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a705dd272f642dbac7e107bfa0ef8ee : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fba112cceb423594d2e3f949331013 : SUCCESS in 14m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/67125a8995d444f8aab8afd4e7414185 : SUCCESS in 11m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a59915eef814d6ea0bb11c5a14c5fba : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3f5f07447de94b9fa103fe42669aa34e : SUCCESS in 1h 39m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/491b72797bad4d88afb09fcbcfe158a5 : SUCCESS in 1h 48m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d54764dfcd543ba8d2f7f156a2d1200 : FAILURE in 1h 50m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9c50812692fb4b6c874f3ef49442639b : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ed6a4d89cb139f36a4e36d2c1e2a4759c97b0943","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-26 09:00:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"636d37c154c0f706e67c9d246d0f0ba7710e9521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:30:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a4afca96e2442bbf36c7f14ce7c919 : SUCCESS in 10m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1bccea1ac404f5bb3f66713f85f0463 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd450b8f38fa4a58bba064336f096109 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57da94c22a1c42648d85c8ebbd4b989f : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e237bf110e9d418a9a447f99fdc6562e : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4afc1f4af4a4477fa59b1322bef84b55 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c277e712d644b9baca6b185d1b6a100 : SUCCESS in 19m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/35cf705743b248a6ab4da7ab5cec480a : SUCCESS in 4m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49b59eea2a5041cfbae315f04335990f : SUCCESS in 6m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c7498195eac147cb836c9215d7023142 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c12a03bcf436454488f2fedfa2bc5b4e : SUCCESS in 9m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/288263d231b5479ca5c55d1455c04187 : SUCCESS in 1h 47m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9fe96a1893f447f4b062385dccdb1cfa : SUCCESS in 1h 52m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec4a8a6eb6274462af3e329b37222024 : FAILURE in 2h 21m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f8d84dd47eb74fc5be6e0d09c927683f : FAILURE in 1h 25m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8a08dc98489545f0bfa8a4340a0deb3d : FAILURE in 10m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4cea42e55f2d4129a2a7e8d626d90633 : FAILURE in 29m 06s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/82da5249f7044f6bb9f320282ffe39cd : FAILURE in 45m 56s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ebc68fc68224b1dbbc4c8f5be96ad29 : FAILURE in 2h 03m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd662f7975584f54b290091b6da2ea10 : SUCCESS in 34m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/be4af096ad244e9c92410621ab749581 : SUCCESS in 2h 08m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f8f28c9bfdd44c1aee5cef3b6bb0517 : SUCCESS in 33m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c646e4b47bf346f8a915e45cb97b65e7 : SUCCESS in 57m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/00ab6d3c72974f5d8d3b07fea37efef8 : FAILURE in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd2b3330be6e5c545f7b321acb3acd3bdd365af5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-02 20:33:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c42f71410cce30a0e2e805127d8cb7d12a5fd1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 23:06:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e69f7ab1a524eb4b19cfb0c5805df3a : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a4823c6076b4c6690b7e39c21c1aa54 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377e31537b4b4cfe8c81c41bbc2da213 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a441fd2f498463ea1c50d5c8a3365a9 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b69baa14e6e4b6fbc27c502ce0df0aa : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a04223277b842bc8393151be4f46edd : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c977cbf0a0448b1a24d4d3c4fc92c33 : SUCCESS in 10m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cbe709cd677f43b387184657055bc12b : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4558aed693b46d7b3358912f0837add : SUCCESS in 7m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca4828bcdbb74a089e01c1b158eba3f6 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6abc6c50ee2d420ea7db83bd5964e2b4 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b8c083012b1457e909d231b3d9baaf3 : SUCCESS in 1h 19m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/18afa3d349204d6eaf716e9f8f900837 : FAILURE in 15m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ec69a7fbad5d45cba837fd305e54783a : SUCCESS in 2h 04m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/58714d61501f4414a7a839cce744e2c5 : SUCCESS in 1h 56m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7388bc513c4749a99076696b385c5169 : FAILURE in 12m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c19faabe487e4345959494b138d2e731 : FAILURE in 1h 00m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5f877e51e9747e7a535b291434858f8 : FAILURE in 16m 40s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8bd49726ba5349818331f09413164bf4 : FAILURE in 1h 23m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/230274477ba04bc3844dbe565e3acbac : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e1bcbddb105d4c4c9504b34406288d75 : SUCCESS in 1h 55m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd7b170537c64bd98970d5cef2b8822d : SUCCESS in 41m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6e80e7b0390f4d0f8865a72fba8b918f : FAILURE in 15m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1bf2169380314b6b8996d6f212f31fd7 : FAILURE in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c140292e2d1c94b7b57ffe83179e9005816e0a8d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 08:01:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8204eae1cb49297eeb6fd59b4cdfc49b18c20c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:03:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1537e16d95844386b91ba52fdd0be2b2 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eb4c94d6e8641ddaf977d1ce2e45d3c : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d656da94cb714ef0b0a0f963ba4dce4c : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85bb1b9afac8473093a79f51bc6de050 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf701006ffa4421bf218ecb30b663ab : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4c1b8342ba64c23a6960c6d01c6db14 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50fef13f260d49ed8d1331b1c90a9a2a : SUCCESS in 10m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c38cab5931a54461992eebbf364fe60d : SUCCESS in 6m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/379222d1814842b1876f58d7fefffed7 : SUCCESS in 6m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0dcf44ff2522473cbae7ee4599a5aae3 : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c119f19a0d354f15973870f238c53a7f : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb36804f9ee748488aa982f600f38861 : SUCCESS in 1h 20m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f746da28d0f046a1aa6adb2215cf8ad5 : FAILURE in 16m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/61fd8df21217428b88614db6081fc97b : SUCCESS in 1h 47m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d0359483319a4214b5ae8f0901692a38 : FAILURE in 15m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2b5444f3042a49c5bd239fbd6ffe01ca : FAILURE in 13m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2dd9ea5be2b5436fab0863d3324f2c4a : FAILURE in 30m 40s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ba9ece997bac41e1a06bd642d0351a50 : FAILURE in 20m 15s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3623562b4f2a45d69ae017f8b1d52031 : FAILURE in 1h 24m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eaa30384c68c4c518c64c507e2e8cd40 : SUCCESS in 46m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/459be7fc23a344038c3fdf3319d3fba4 : SUCCESS in 2h 02m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f290efee6744bb8a7656558e09b4aa8 : SUCCESS in 47m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/961d3b2f8ac04322aa192da0fe474b81 : FAILURE in 17m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/496c2ae9a51442698c34637f2d036d42 : FAILURE in 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3aa28521a2af06e7fb81debd9b04cc8f9fe8d141","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-23 08:06:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d2716894346caa6e7e4c31ce7114de3f79ec623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 12:57:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e99242f5fa5449d2a3b3baa0b62ce593 : SUCCESS in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b237d4844c28498198213fed21411cf4 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8da430046a44d35bf16779c90c6a8b7 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bb3a04c8680484a9f488af859db9b50 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba473ab1b0044c1a347132200e27c80 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cf6ac43d07943868dd895a02bb00a26 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cd2de9ec2244a186f67e7df1a89fa9 : SUCCESS in 11m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8242ed5434954b97b93103d22ba4e44d : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1dfc57d31ac843369b9d83894cb1e706 : SUCCESS in 7m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/b6c4223f03e24c1ba6c0e02cbadd74d2 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1385347157124b8cbb8dc8e44f078e33 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37487a792ff241e6a6ee9a925d79d171 : SUCCESS in 1h 11m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f14a00899635474db61cd2892f7a2c53 : FAILURE in 21m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e289d57ca2c48b5877cdc13f5831c4c : SUCCESS in 1h 55m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2490d96f7ab445d7aecf9c5ae1e4674a : FAILURE in 17m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a902d62cfa544915b2c96f9d645c9c60 : FAILURE in 14m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6169415bd6534923920cb6d4e2f6b840 : FAILURE in 26m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b23d36ba2ef464dae394a1a81868cd7 : FAILURE in 15m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b14d19d60044c9f8f52c3b1bde141f3 : FAILURE in 1h 31m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f5ff162af034e6ca12e51b7669764f1 : SUCCESS in 38m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a853ffb549554f66939df645c5c1ecd3 : SUCCESS in 2h 09m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6c544b7fbba4768a4b0711574999ccf : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7897ca6e23014fecb2ded15e82567b9d : FAILURE in 14m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b90eb639b9c14b38a19eb6c52270f781 : FAILURE in 10m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba9bd863a0b457be09841da7e6f1220da49f56a8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 07:55:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be86b44216eec1ecd587e5a47a01f69535a9dfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 11:35:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e7e143a239b4996b11ef9a54ae0c806 : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd419e4e77a347a1827b00f639b0310f : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849411afd6e7499385e868b2c7d75fe4 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c050dc36a87142b68346081ab3f25e8a : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d4cd16606c54c5abfb97f4ea83660ed : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05edec1cef0e4d0bb77eb9aa2584b0eb : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815b827cdf3541a780cd2f80be3ad445 : SUCCESS in 16m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bf458c7974d6486f93d20bcab1439aed : SUCCESS in 4m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62de8db8647941e69e118a1459818f30 : SUCCESS in 10m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d02e6489a9f54189846e854e4c737244 : SUCCESS in 10m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7949aed2742d4a4a9eaf56786396f590 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de2f28f9c7054f9bb21db5d2ae1fd3b7 : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c495cbe7d7ed449ca87fe19199b3c2fc : SUCCESS in 1h 23m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/889222461fc340d1a9e4a94e6c0ff4d0 : SUCCESS in 1h 46m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/726e820cb4e04b3c9ea9104224eebf44 : SUCCESS in 2h 04m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a595fe2c980642bfb0536c4ab881dafb : FAILURE in 10m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/213653f73d0f4204b570c07b5401a0e9 : FAILURE in 28m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab3759c57e4a42baaa0d0823fd5fb466 : FAILURE in 55m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ee9e6e85137462b9a2662676a4c0f42 : FAILURE in 2h 01m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/75f6a64c7f2d4a04a6895947f96a6b23 : SUCCESS in 44m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a40ad2858b3741c2913eb7d94080d235 : SUCCESS in 1h 55m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7faddebf7d0d456184d97fc63eb8f858 : SUCCESS in 42m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f04d5e27069d4057834b00c6999d3c06 : SUCCESS in 42m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/45d883c4d07746958a5b92fb7860674a : FAILURE in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"227dfd2b336f878a3e20476c094d0193e512b759","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-10 08:27:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"574c144593ba379358971e0b06bb417cd5d177e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 10:48:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eda1aeaad7754e998be5d88f4acb37cc : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5da9f2ae5a43ab9761040847fc3122 : FAILURE in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec3c2abd1d648489b2ccb011cae3567 : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e031db635704ce3819e655f8a03265e : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee4f9703d11548a493e7c0f747f979f6 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9dd6a7d729f24ad9be3e0c7a84dc4951 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f281810df5f49e5afcfb8e61726d19c : SUCCESS in 11m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/14cba2882b92444c9db1bef49502b611 : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5e769a16840a43ce840f20608a6e0aa4 : SUCCESS in 10m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca2612248bb84f46877872204384c70f : FAILURE in 3m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c02c6599171042098f3cc070d63baaac : SUCCESS in 8m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4e3186570d8c4c6f98e9a6651dea7207 : SUCCESS in 1h 18m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/2ba7f093859945fc87f314ceafcc2c29 : FAILURE in 18m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d54bdcb7dfd448597b26691e654b669 : ERROR Failed to update project None in 2s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6e55f3e0426a44cb8101713d90412c15 : FAILURE in 16m 58s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8ae9bb97a1ac420983a77e4e591c19cf : FAILURE in 19m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad1523b2f77d4856a30b67538fdb6260 : FAILURE in 31m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ce48fb8f5464c72933bc935079f7cb5 : FAILURE in 13m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c4f6838480f44aa8a316a081b5cf085 : FAILURE in 21m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8904dc026d1747b1a11d523646216132 : SUCCESS in 45m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fd9a56fa1756414ab70da0e9b97b7ef9 : SUCCESS in 1h 35m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/238229ce217f452b9be5803320d18f10 : ERROR Failed to update project None in 6s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c553f6a47ba64cf2be0bc943069a7146 : FAILURE in 11m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4b84bd254bae415bbbc6b68c424ed5cb : FAILURE in 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2d6e1165538c3c1d29cf4bbfbcd95049d9ecdfb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-13 21:09:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09077f957dfe6ec8d98869b5451c5ff24e06a12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:39:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9b00e35bfe48ad86af3cda57db1d26 : SUCCESS in 14m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/311a6b97870d4617a6ef9bd902bcea01 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4656dbd5d9f446d48cc0215e14269a6e : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/914ba08817ce45ea915ecf236ef30bad : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0773cfb72c5a4309b2f3c02ac967531e : SUCCESS in 6m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b27ef2c24f1446f68f0edf3cc5a91a7f : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c7f5ada41e4beebdc4210fe584b800 : SUCCESS in 11m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5bf01c2d640c4b80ade509712534f05c : SUCCESS in 12m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b9003da4a13d4b8cb157118f933e653a : SUCCESS in 8m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7af164cdd3e34dbea3825285026846b6 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8842c6e1e334f83a27c315fccb117f8 : SUCCESS in 12m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/861684a767194f3b8a7cb5f0974cbf1c : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f0960dcaff934d8090b90cecfcd62973 : SUCCESS in 1h 42m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/96d9a9a070074f83962519c426175916 : SUCCESS in 1h 38m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8b5b2ab9db2e45beb34aa88cbed1c187 : SUCCESS in 1h 41m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1166cc9787be4efe89a183eef791feb5 : FAILURE in 8m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f3973e3da3d648d3833b0d9e432e116f : FAILURE in 26m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b55e8f05ae7c4a4c904ea2190599ec7f : FAILURE in 43m 10s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d822230cd1fb4d49bc17559904009f0f : SUCCESS in 1h 48m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5465e7543cce48119ea83cd4011ffbdd : SUCCESS in 39m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e6b816cffc7444d7805f17fd3a4012ab : SUCCESS in 1h 51m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02f1b88d2f214f3e93315ed17962bd72 : SUCCESS in 50m 03s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/08fd6a6d655649c5b01c863b7212aa41 : SUCCESS in 33m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4fcc093500e34925b86a27d79bc9ce78 : FAILURE in 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fae26c37efcbde0f9634920069fba1052c5e52c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 23:40:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c253d0c88908f17c709daac28e7e096026984778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 01:38:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8856e9dc21e74196bf7fa7d8ae9380c9 : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3327d41f4b4d44e7a146d8a03850b2eb : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db6ee1f9f2744bb59d304066badfcdce : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ab7a78a80b243ad8bc22a995ad52b61 : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db658cb0acd548b88cf32eb9c665011d : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e45fe57c0ee465ebfabe6679b94b1c7 : SUCCESS in 12m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4dffdd650f4e489fbc170a3df5910f33 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e28636256e424dbc85c8fd660b749461 : SUCCESS in 8m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3424a35d37e44e7f9a45b90964724946 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f659c483f860477d8a72771b4703d977 : SUCCESS in 1h 03m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d918ca22ec1e47268b92e9d92456f90d : SUCCESS in 1h 54m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/385986af345b46c6b9a2911d45792df2 : SUCCESS in 1h 43m 38s","accounts_in_message":[],"_revision_number":1},{"id":"86852fb216747b1d9f6ad0edd5b190db31b2c20d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 01:38:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"60f279dc7cdd8a6d43f678718113558f7024529a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 01:40:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9581cded35a648b9b70c3cfbec38d385 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cad144e0967c859ef25bc106e76c7c4ae90536e","revisions":{"2cad144e0967c859ef25bc106e76c7c4ae90536e":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:16:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/66/752866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/752866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/752866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/752866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/752866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/752866/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-23 12:49:19.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:15:52.000000000","tz":120},"subject":"Add some details on enable_anti_affinity option","message":"Add some details on enable_anti_affinity option\n\nExplain that \"enable_anti_affinity\" affects only amphorae in\nACTIVE_STANDBY topology.\n\nChange-Id: Ia9d9d844b325296401577f5617d2b89cf1a017a7\n(cherry picked from commit 259f19ec152fe64777ea9bdb658929e22d59427b)\n(cherry picked from commit 6a1121420f8e60648dd2b6be37af38e8a725a1d9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cad144e0967c859ef25bc106e76c7c4ae90536e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2cad144e0967c859ef25bc106e76c7c4ae90536e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
