)]}'
{"id":"openstack%2Foctavia~752867","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ia9d9d844b325296401577f5617d2b89cf1a017a7","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ia9d9d844b325296401577f5617d2b89cf1a017a7","subject":"Add some details on enable_anti_affinity option","status":"MERGED","created":"2020-09-21 06:16:35.000000000","updated":"2020-10-21 20:52:28.000000000","submitted":"2020-10-21 20:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752867-1603313548450-af1bddda","meta_rev_id":"ad47a8a69b17dc82abcf25cbdb06c0b25ed3d31e","_number":752867,"virtual_id_number":752867,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 20:52:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2020-10-19 12:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2020-10-21 18:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2020-10-21 18:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 12:28:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-21 18:18:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-21 20:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbb07c6c7786648a346c83ca6b6daa43ca3104d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:16:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"860b634741643a17f5380c4b363159e1d005d899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 11:48:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab0b6a30d4524b0bb2b6b7981411bb94 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdcd72c9828f49cea9baf322c44bf94e : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5364f2ee74340749255144a72e4f7db : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70e979c0b7474ff2b5865bdfda5bd247 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/58d1bce3eb8d424fbeb3f00503a057f6 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b91cfdd7694a3d8b27973d8e90e9ba : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f02474f24374d6ba8e1f39c3a2270a1 : SUCCESS in 10m 45s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d733c1b6d7b94f15966b4a5e5e5e5a39 : SUCCESS in 1h 05m 53s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a88449921d33444ba7059fba99d8c6c6 : SUCCESS in 1h 44m 39s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/01313bda4e55469a8fb396d1522083ce : SUCCESS in 1h 43m 18s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/992d89e11b33416693909251880d1d0b : SUCCESS in 1h 08m 29s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c0a84854772b404391c392dca75682eb : SUCCESS in 7m 31s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b4f2d7957a0d44a78e7bc7e82232ea6b : SUCCESS in 8m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7be350f21ed347259572461257abc015 : SUCCESS in 6m 51s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/614eb5a3f5bd40b09d63d4a13cc27cd9 : SUCCESS in 59m 07s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/99f6058ac55440888048562c7a79e21c : SUCCESS in 1h 03m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4fe937dfd9c4bd8853c81ff72af7f23 : SUCCESS in 1h 22m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b67a89b696ba4565a9f1533fa81dff92 : SUCCESS in 1h 35m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd6acca4ed5f4e379293d02cdf95d6e4 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a8b234a4d0d243da87b9c15ed7ba414f : SUCCESS in 1h 16m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/aa1172faa7a54b15b56e557befcd78dc : FAILURE in 10m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56810735fe2545529d5cbdf0231bcc0e : FAILURE in 47m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/dffa4cb3f70d44bb820718e8b395e405 : FAILURE in 1h 02m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/20df5a80ef5342f1a26373a570e99934 : SUCCESS in 1h 20m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ec55c31badd4972a634def3034a95e2 : SUCCESS in 1h 07m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f3945e1838c23728f9ba3f578d0062bdb0056cf","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-19 12:28:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6985f37086c6af8a0a5aa114cada7825d8bb432a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:18:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSure","accounts_in_message":[],"_revision_number":1},{"id":"1ad1a934dbc39c2a5a67c7b9e82a04599065291c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:18:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a88fa5e84031258dcaac50b181aea4676e6cf8f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ad47a8a69b17dc82abcf25cbdb06c0b25ed3d31e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcd4ec6d472944f3a5f746f630f9c5b9 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c92fe704b9548eaa2721b1cbe0019b4 : SUCCESS in 9m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5df14b81d1384c4db62aaa643bf5acd0 : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08a96bec7edb4556adbc8b253533c5b4 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e23bd5cefe35428b8f4f79f87a80b80b : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc6254850504563aaf9bf0ab85d2a71 : SUCCESS in 10m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16e9e381e8da4254a84506df42ac95db : SUCCESS in 11m 12s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0bb7f81c2e134f4d926da8f6fe581cfe : SUCCESS in 11m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17dcd6dda2344cb9ae98526aebb36a7e : SUCCESS in 9m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f840ec0f07a4a5aa5e3dd83b7234fb3 : SUCCESS in 1h 03m 15s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/63b0943599a0498c9e71bc7a3b874e71 : SUCCESS in 59m 13s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c6ec86830c034e35b418f491a7571676 : SUCCESS in 1h 25m 17s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8a2401acc2be43ab8b231343d01f82a2 : SUCCESS in 1h 26m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b2aabaadd6942248829830406c192c3 : SUCCESS in 1h 30m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7374c153ec8e432fb23d5f1a0617d8e0 : SUCCESS in 1h 26m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53507e4a51c747bfa7acc53d77a2d771 : SUCCESS in 1h 36m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4bb64e31a2f4209c5441598cfa908acea6537d6","revisions":{"d4bb64e31a2f4209c5441598cfa908acea6537d6":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:16:35.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/67/752867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/752867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/752867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/752867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/752867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/752867/1"}}},"commit":{"parents":[{"commit":"2ac913cd1946218a42c3209df29cc5e74e811634","subject":"Merge \"Fix API sort key for complex columns\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2ac913cd1946218a42c3209df29cc5e74e811634"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-07-23 12:49:19.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:16:25.000000000","tz":120},"subject":"Add some details on enable_anti_affinity option","message":"Add some details on enable_anti_affinity option\n\nExplain that \"enable_anti_affinity\" affects only amphorae in\nACTIVE_STANDBY topology.\n\nChange-Id: Ia9d9d844b325296401577f5617d2b89cf1a017a7\n(cherry picked from commit 259f19ec152fe64777ea9bdb658929e22d59427b)\n(cherry picked from commit 6a1121420f8e60648dd2b6be37af38e8a725a1d9)\n(cherry picked from commit 2cad144e0967c859ef25bc106e76c7c4ae90536e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4bb64e31a2f4209c5441598cfa908acea6537d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4bb64e31a2f4209c5441598cfa908acea6537d6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
