)]}'
{"id":"openstack%2Foctavia~752869","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I400dde533f78bc14ee568bfd9714eafac97e8a39","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I400dde533f78bc14ee568bfd9714eafac97e8a39","subject":"Fix operational status for disabled UDP listeners","status":"MERGED","created":"2020-09-21 06:23:42.000000000","updated":"2021-01-15 07:34:38.000000000","submitted":"2021-01-15 07:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752869","meta_rev_id":"65a4dba8ea45b9b8df22c1e3f4b2064af7263a03","_number":752869,"virtual_id_number":752869,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-09-21 12:47:49.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 07:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-14 09:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-02 00:08:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-12-14 09:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2021-01-14 12:27:25.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:23:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-21 19:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 00:08:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-12-14 09:03:46.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-14 21:51:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"0c0d04be9ba51d1171db5f82457fd2f56187c14d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"517f90174e2db3ffa849283cb809098d50f97d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:40:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc2bc19f414f400c8c6208c79ff4527b : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b02471fd7bfc41a7859a027455bf5c62 : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bfdbc0cbc043f9a9d5c796fd9bdbb5 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a18d16dcd3a64e9fb36e0af599cc3906 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1ebbff00c1447ef987dd56e9bea7f2e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/964f76767e044ebbb4db13b9f382154d : SUCCESS in 19m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9c5a609987144a3b9ceab9fa98e7029 : SUCCESS in 5m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/254318fa77fc46e8bad7a7ae1457a7f3 : SUCCESS in 6m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a24a534232f44171ba05fb6c67a19473 : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/de3b77dcae244b40ae2414c8c5156fcd : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b95a5f78a0b54ebb97fa1bb9a4d010e9 : SUCCESS in 1h 39m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5ad6a0de043f4b6ea761a70c9b89af26 : TIMED_OUT in 2h 33m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5e9b21c1000c4fcd93fec09a02699f34 : SUCCESS in 39m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6f2033a42f664292a868f22ec0884175 : SUCCESS in 37m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84d4066386c049daa41417cd6e440438 : SUCCESS in 45m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/866aaa5770d84dbdb4a591e0efbc9ea9 : SUCCESS in 1h 24m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/730973a2e5fe4b3397bde8fe968add77 : SUCCESS in 1h 41m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d12a8e013f5c4a15bfd07366931a486b : FAILURE in 1h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ab34a5673f14a6eac604fd2560dc926 : FAILURE in 1h 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e17fb1da6fe1f171e7a1ba31fcad063aeaf8cd6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 12:47:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"115e0f7e6da0499dbf8bd892c230200a3b70b5b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 19:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9aa1db4be6148e8b2cc124fff1e2833 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3669837c9aa0470bb4c2daca99b597fe : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56946bf3df8491194527ae25b6b5ec8 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab45636a19654f00bf2329d2c9a9249a : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60997c8d28484fb7b23ee5817a5d97a3 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb716f2b2474aeb938c5d2d4cfaf061 : SUCCESS in 17m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed06dc2ecc8741a19b3c0cb96922af0a : SUCCESS in 5m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/00c3c784fb104e5a9e7ee5bf1fd9d285 : SUCCESS in 6m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b65bed152a3461f904ee8b496c41a5d : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bd9a35839e04b2e8e757b6ee9f8c78c : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d9d6f051d7764781aa9e0edc52051889 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82f1e71e3bcb4913b0cf72966cf72d7d : SUCCESS in 2h 07m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1e72544f389d4d0a8f9260118bad83eb : SUCCESS in 42m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c565b1772f8642a68b9c8e97276302b1 : SUCCESS in 40m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2aa33d5b8dcd45308f915254f2306c42 : SUCCESS in 39m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/65bd100976af4380a72319b6bb969128 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ffd18a8b52848a09c03383272935685 : SUCCESS in 2h 27m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b1f69fdea6b4a4ea447563d9efe4745 : FAILURE in 47m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f7704ce8cac34bbc8ff3f553abc0276a : FAILURE in 1h 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46ac770b02836de1647a26dc1a79b984a4598db6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:08:21.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f6210501e3810bf6cd6260aeef493c6ce623e1f2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-14 09:03:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d86fc28937401e35d4dd39dec23e7221d7cd4a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:04:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4fb2bbe59354203b6c2d80cb7b38e3fcd97b891","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:13:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98446f3b71004e839a112cb642495d65 : FAILURE in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d41de660bf9746ddb04103ff07d82fec : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d56d9d0c1aab443cb028e96e7660c6bb : CANCELED\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c4827b423a84223ba75bb89281e6c31 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da1f7abd268e4bd1bbde94fec60d9021 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5926dc75126f4014a74dff932bd98139 : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cd673f3b21045849991156d376e4c00 : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/114b9c16e647466cadaf17398da9470e : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4cf004fb6b543eba64308e69fee0274 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d08b425a956447ab786e0a59124154f : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf8d6cd116dc4c7ba79ad6c0b830c063 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a174acb347584dbe8662dc6abab3536d : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/None : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"fe6065ea9b429c13c1c62223752f161d755ab2d1","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-22 07:25:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3bab300f1763cbf54d3e54a6f42ad4c62b800b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 09:31:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13eccdec62544c2fb5ec280d4ec4707b : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/721e06ba60554686aec46c728326ae32 : FAILURE in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96341f6ec6bc4fd6878e55bdc52dc98e : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73661738b28e4c95a656a24f2905335d : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7296ce38a0cf4de6a34462b8faf6906f : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946109d00eb0444a945a7f812bce6ddb : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/416e08b584994761bd6d993c85e430af : SUCCESS in 4m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b86330515abb4d2db7ddd4557b11dece : SUCCESS in 4m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/848589efe742459cbf1bc95bbde61281 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d7352cd55a14bfe902c2971855d0037 : SUCCESS in 6m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc2968fbaeb040dbb87c8d3808ecc742 : SUCCESS in 1h 13m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2c9f7eb91ca44e7a5caa6cd569058e8 : SUCCESS in 2h 04m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/84e945634d3c4d33856b7eb9daeb51c0 : SUCCESS in 36m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6c230bdff2da49e29269b76449ed9105 : SUCCESS in 41m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d451765b9f4543eea62be29d90c1172e : SUCCESS in 42m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a260eed22ada4c3496d11d0e27909853 : SUCCESS in 1h 12m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d2c81b74a5a44ed98d3805843645760 : SUCCESS in 1h 52m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73dfa8cee4e7444e912f52a6daf014a5 : FAILURE in 44m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7328f0f3e1d04f35b6f0ce034c033a78 : SUCCESS in 1h 59m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2db7afcd40840784fb3921ff43109861bd895cf9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 12:27:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4076eb9e641216d969d6e89c2291784d072a788b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:22:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7354bbb51200450492f7dd17349cf944 : SUCCESS in 12m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9883ac8122954873add7b7a2601881e1 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ddc7e942eb4ec48f508228de70a50e : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5f3445bca634103bce2e14d7013e7cf : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a0e0d6caa0949ae944014851fc827e0 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ac3ecf39be426fae3e4a395959d415 : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98370d3b1f65496f8a17fd6280d9fb74 : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/909d66e1d8014dab93224a998af17d3c : SUCCESS in 5m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5386094a866347d7a8b9679b99ccc272 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf41e11bca3b4415af0837b139477c73 : SUCCESS in 9m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98460e8188104255a039c3c2a2ac9fbb : SUCCESS in 1h 55m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14db01c6543649c8b11fb13f1e1c5403 : SUCCESS in 1h 23m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2929f2d037f74110bcca62228c2a79e6 : SUCCESS in 40m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dde03d88828f400e8763414ae61ee06d : SUCCESS in 46m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1018f5b0e4a948999247315f2b254fe1 : FAILURE in 1h 07m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/03317ca5c5e04b78948e5687669280d1 : SUCCESS in 58m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a7e08270a72f459db216d52b3d83bc8c : SUCCESS in 1h 56m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2da7530b825b4e8c97bd8bf9705add41 : FAILURE in 1h 19m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/563233c7a1964a849d57afc239203573 : SUCCESS in 1h 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"959bd3d5f0d7146c8e8028398bc1bca2e512675d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:51:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ecd5ae828c7db83f5fc342778b86d63ffd25102f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:21:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47632753caf48beb87fa05bffddc3e5 : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfd3d417fe8d48b5a57b22f5338f7452 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2021fbfd4ec64c508a7575a9f0eb31f4 : POST_FAILURE in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6037ffeabc747bead5c3e1cf1ce3a66 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c6da6a831b74f3398f856523d4b1b55 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad181129f774d8eb25fb182ab2a586d : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4bdecc0a58e457d98b331adf34d7f92 : SUCCESS in 4m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9ec477bba69c4fdd9b80d007fbc57fea : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/96355ac7b3374f50bf45d141eced957d : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05968fa5aef34d80a31133fa89fcbd45 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f26081fb26345019c59aa1455537b95 : SUCCESS in 1h 15m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4199e7a93d2f4ff7946bb14128f3ce14 : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4fa5d3a28404a34b220f76cc8152767 : SUCCESS in 45m 56s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d2cd9862661d42eca7223d556e95aebc : SUCCESS in 46m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9ea6c5174fb4e85b9e7ea55446dc90c : SUCCESS in 41m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b4d3cf39cf304b679fba7fe9ec7175fa : SUCCESS in 1h 27m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a7e7febd5c641f1a29c46627ef4d627 : SUCCESS in 1h 55m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1293d3777eb34cdf9573c78f5573a658 : FAILURE in 54m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ade505ae923c4572a06b0f647a1ad247 : SUCCESS in 2h 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df7ab4ed6ff3a7e91897eeaaa833f9c96c0b87c5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-15 01:43:40.000000000","message":"Patch Set 1:\n\nrecheck POST_FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f2fa49d469e833fe01455b7a39e81a79c75ce5aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 05:22:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b5a28164c9d463e9e12f5721adab1e1 : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f75efe65c2dc4e279603cc05ecdf821f : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3734587f9b49f791c4a2315ec0545b : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ce7db0a9ed34d13ae7d516e926be56b : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ce98ddc704f4d8eafa190a7355b26a6 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d243e9ad0948d3ab90ca21804511a9 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfb8e2bb496e4754b9480082635b7dc4 : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ea6e15856e894e69a71e4152b2970352 : SUCCESS in 5m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f04b04ae526a49748da56642763b3844 : SUCCESS in 9m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b369258e2ab4807a00701de195b874c : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/637e7d814d01440c88f164b50a439ae4 : SUCCESS in 2h 00m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15eb5a438c7b4a8eb411bb7d53b981fc : SUCCESS in 1h 59m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c37aca6ecfb4422fb847284c168305da : SUCCESS in 39m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7e47890689544197b918bfaade1d3363 : SUCCESS in 36m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0491832102ba482fbd8374c5cb15fe00 : SUCCESS in 38m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48e2efa43eba4ef4a3f1c46987e2924a : SUCCESS in 1h 07m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9432973f6f614efcb1d92f337074f385 : SUCCESS in 1h 57m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef24f8641b914ffaa6c8266be72f3f45 : FAILURE in 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cea740b1fc644b6e9f9caad08db3a75f : TIMED_OUT in 2h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfbb0e14d1f4f4e68efd249c4643904ad2a3261a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 05:22:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84147310480fa97cbb95d1fb6329fab1ca393237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:32:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccf796ed114543aea2837705dce17002 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a7773ea29342e19cb9fdd24b59420e : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2b5dc2f0a634187816c03a3cd4ef879 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffd8e4b00aa9491dae84db8d9020a869 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f1ec18aae2401e82ee99cc6ea67219 : SUCCESS in 16m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff5f6a5b8ebd454a80df4dfb7fe78fd2 : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1312cad487c84d06abcc39e0d331c473 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/62250227807341d9952a83a350c33369 : SUCCESS in 1h 26m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11a4fb21e8e246b7965ebe080493811c : SUCCESS in 2h 04m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d69cec4213fd442a9453e68b7da1867e : SUCCESS in 34m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/68fcd8f17ef247ba9fe64550b83a4429 : SUCCESS in 39m 50s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33f0649e9f4346d68f718bd63401f4ae : SUCCESS in 39m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fa3ab9768c1a4fa2a9cbdc7aa7b7d856 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d90b19b896f89704a000e93bf3860a11126f8208","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:32:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65a4dba8ea45b9b8df22c1e3f4b2064af7263a03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:34:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7a576d86f44696871087e2e14f5578 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62e63c83cf664d8cac2584d589837a1c : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fe7cff8dc3f95c229f3e7a1608c414a495e515a","revisions":{"8fe7cff8dc3f95c229f3e7a1608c414a495e515a":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:23:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/69/752869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/752869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/752869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/752869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/752869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/752869/1"}}},"commit":{"parents":[{"commit":"43db34778771ee1af758ca56e9f862724c03e86b","subject":"Update grenade job to run one smoke test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/43db34778771ee1af758ca56e9f862724c03e86b"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-08-05 06:14:24.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:23:10.000000000","tz":120},"subject":"Fix operational status for disabled UDP listeners","message":"Fix operational status for disabled UDP listeners\n\nDisabled UDP listeners appeared as ONLINE, as the heartbeat message\nalways showed listeners as OPEN, even if disabled.\n\nThis commit fixes the configuration file for UDP listeners, previously\nenabling or disabling a UDP listener created the same config file. Now\ndisabling a UDP listener explicitly adds a comment in the file.\n\nBased on this comment, the heartbeat message doesn\u0027t include status\nupdates for disabled UDP listeners.\n\nStory: 2007979\nTask: 40605\n\nChange-Id: I400dde533f78bc14ee568bfd9714eafac97e8a39\n(cherry picked from commit bb4f50f798dd5619185cebc5f3db9a55e7f6122e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fe7cff8dc3f95c229f3e7a1608c414a495e515a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fe7cff8dc3f95c229f3e7a1608c414a495e515a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
