)]}'
{"id":"openstack%2Foctavia~752870","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Id9e19375ebca6a720e6a85006f5e8948d3aed760","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Id9e19375ebca6a720e6a85006f5e8948d3aed760","subject":"Fix invalid DOWN status when updating a UDP pool","status":"MERGED","created":"2020-09-21 06:23:42.000000000","updated":"2021-01-15 07:36:37.000000000","submitted":"2021-01-15 07:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752870","meta_rev_id":"5b13c814f0818f5355c64aed11e9d0e9fc142b59","_number":752870,"virtual_id_number":752870,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 07:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-14 09:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-12-02 00:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-12-14 09:04:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2021-01-14 09:24:55.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 06:23:42.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-21 12:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 00:08:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-12-14 09:04:18.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"b24738ed65191c99732d16168a9ad8901d5dddf5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:23:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74c907c6e1ceca1ea5d051c4c61bb39ea32b25ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:34:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f4b280b18af443b9851422b47201566 : SUCCESS in 10m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f666ae47e28b4c18a0529a1c1134ee6f : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24eff456043b4c5988ba0cce53b6d462 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afd23b88ca364ccb8d3855aedce1e0a4 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecbb11ddc7974a0592dba159b07b013c : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5d661dc53f4c18b745235d9c527ea4 : SUCCESS in 28m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f212fb496244e359052d2d54e1ed8e7 : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fd53718b59ef41058c4c2287991ced5d : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5569bb3b70c34c99bfd6bd9c0216eebb : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cce7004f47040e8ba5665f17c4bceb0 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c153fc2212634961bff18583729055a9 : SUCCESS in 1h 50m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca6e23d9f5a0443e983febcab9b877bf : SUCCESS in 2h 06m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b115735e514341ee8177640c8e4e7089 : SUCCESS in 43m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4613c1bc6e148c3a6522d013bd2e99a : SUCCESS in 49m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e135e00bd6074d789eb6eab8b943c665 : SUCCESS in 33m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ae6462279e624697a730e8f6846b161e : SUCCESS in 1h 17m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b27afd5cf78452c9e80de385a16181e : SUCCESS in 2h 10m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c02609ce836a4692afb66eb73c61b83c : FAILURE in 1h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1ff7abb5c3fa49168072b75cf0ede0c9 : FAILURE in 2h 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54978913a7dbd3b31be58804f7fec8e9a52aaff7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:08:41.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"ffc7c8d873e6404e7004cd028cd0993bbb08f4fc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-14 09:04:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1912035e8d493cf6a4502212b2d89d1e063d331c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:04:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf594259b19e774b369a75189c7bd94c8dd849b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 09:13:31.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"95dffacc3f933417c718efe12d3a3541f0a2d0e1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 09:24:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8bb31700610c15cd2f0a91bafbe57a0d4fc06e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:21:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f1204b0f8e94483b299b100295e8351 : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4e27153250848db8397a0814b656335 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a068409e2364615bf142a7fadc4d144 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c85c60c22e784a52b86c2da570c97372 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7222d1caec4491987dcfb509d91cb03 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572878d204ef4f44b0661e36d97110fc : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66ebce3dd16c4a2b8ec179fd80724bfe : SUCCESS in 6m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/02fb9017f76842798755b612482e6ddd : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ab6ac10bc3014b75942f8bbaa04a00de : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8171f93f06b04da4afbd7127794383a1 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87bfa4af2a594b58b57b99e3aab33b22 : SUCCESS in 1h 11m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d85233997ca34953836d319736b2e176 : SUCCESS in 1h 40m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e4f55ab2b7654dc4b67a18ea31415720 : SUCCESS in 53m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/3f3c4776d39f4d6ba32e18f8809d0280 : SUCCESS in 48m 12s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e8672eefe504560a777939184c5ffd1 : SUCCESS in 42m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c5b80d78209b4cf88e95a397084349e4 : SUCCESS in 1h 11m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7479e19d120742ce821c06a263c25f79 : SUCCESS in 1h 49m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d18b4e861a34359b7c15badb04e41bc : FAILURE in 47m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d7b1123c5fd743f98cf3a388d0807fe0 : SUCCESS in 2h 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2971d3ee0dcddb7802cb5c97d59e32e018f0733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 05:22:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9455353291dec422dabfcefdff00f7f74816199b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:32:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd104c60cf6044ac9d311f09af6d60f3 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e50876718f4eba9f07f9aae1e8f8b8 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a08c88913c5548dc8ee49513327eae49 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/107df7b6176f41e787599d9e41815a73 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3617ae1fc9b46fdaf57eb42127c3a4b : SUCCESS in 13m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e57440d541b4138a87e57ca3c871c76 : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7df343c004e24d9db4b8b7b8f784a229 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d55ac8f007d649fabfebafde7659d600 : SUCCESS in 1h 03m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/af3f1ff0abdc4e84a983b2aa222d2020 : SUCCESS in 1h 59m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d21313f66bfc487aa556a2bbfdb18975 : SUCCESS in 43m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/755f200481f74295b521abea65045df4 : SUCCESS in 36m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97555565ee9f445eb7a3beb01e51ec49 : SUCCESS in 38m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c775716341c44c208c39c7d7f4384df0 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1479d6f1ac357b7dd554d94012757077f1d57373","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:32:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5b13c814f0818f5355c64aed11e9d0e9fc142b59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:36:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc9a330575b41f2bc2c956b65cfc7ce : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6e3379e099468a8144901fe729c287 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea47a0efadb42d9c052fa7ae114e8f82a4be2775","revisions":{"ea47a0efadb42d9c052fa7ae114e8f82a4be2775":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:23:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/70/752870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/752870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/752870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/752870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/752870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/752870/1"}}},"commit":{"parents":[{"commit":"8fe7cff8dc3f95c229f3e7a1608c414a495e515a","subject":"Fix operational status for disabled UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fe7cff8dc3f95c229f3e7a1608c414a495e515a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:23:26.000000000","tz":120},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n(cherry picked from commit 9fb58eb9f4d2574106a48e3cd3c439d5345517f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea47a0efadb42d9c052fa7ae114e8f82a4be2775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea47a0efadb42d9c052fa7ae114e8f82a4be2775"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
