)]}'
{"id":"openstack%2Foctavia~752872","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Id9e19375ebca6a720e6a85006f5e8948d3aed760","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Id9e19375ebca6a720e6a85006f5e8948d3aed760","subject":"Fix invalid DOWN status when updating a UDP pool","status":"MERGED","created":"2020-09-21 06:29:01.000000000","updated":"2021-01-15 10:49:20.000000000","submitted":"2021-01-15 10:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752872","meta_rev_id":"80a5f7fad0f9f74b4b9ac7ebe150341c9f2d8e70","_number":752872,"virtual_id_number":752872,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 10:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-15 10:49:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-15 10:49:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-15 10:49:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-15 08:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 15:48:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-22 11:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 10:51:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-29 17:53:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-02 20:16:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d074447491bc4e011b117cc036317476d74e53ef","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 06:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e003d97f16f1d7ab3e464f2329ad488554fa52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 12:54:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf75dd038f0f4d9f9d2d28c6401d3946 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e748827db844432cac3686e4bb2c7483 : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744b1a5dc32c4f858556b02a58138cc6 : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6028353e87b442c0b93e860605767a1d : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3084612f2a8e4ce780ee87d2746c1650 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7644454c1dbe4c5d9cdffd4aae03053f : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0975476cc524469bbd80049dd4f343f : SUCCESS in 17m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aec9added6d497397de57e7a8dd8da5 : SUCCESS in 5m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f1abfd779f2e44a69b24bee2fe758190 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b4321dd2979442b3962594213990cf8d : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1c40fcebe5eb4d9c92579d2939f366f1 : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19cb3eee9d9249a480b7f9d4b448c7a9 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8479af4647654d4a98ed4b1d41adeeeb : SUCCESS in 1h 37m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6f9aa255508e49fea6df1bfd77f14e0a : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de15065b6b6040b9a950cb0bece12956 : FAILURE in 2h 01m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc113a1f804c41f2a3d832aff8451f6f : FAILURE in 1h 36m 54s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bf5ce32e4c3e4e849140c5f8ce2582d9 : FAILURE in 12m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ef282e7b5f5941a1a66cfdf6b7ed49ec : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4e046b9556874514b2a2da7d3c85ccd1 : FAILURE in 1h 04m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f6bb515de175416aae3fd859d7ab723f : SUCCESS in 1h 29m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22c522b827ba4e9e9b36cdf96f9920f4 : SUCCESS in 48m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/77d6d8ab2b9345cebae24347c6d87ce2 : FAILURE in 1h 58m 07s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25de501a19674dbaa90094007676279b : SUCCESS in 37m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a3a076db692141438a3008f023ef68b3 : SUCCESS in 38m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/964bb9c56c4f4391a38ea62b908f0515 : FAILURE in 15m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"353f25ecbaef2f15d51dffb9cb24f2eb4f2fb835","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-21 15:52:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b15849921118a8c8f58eb4a3612fec5514641b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 22:36:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de2b3ea772a4e2ab9a19fcdd2fdcd68 : SUCCESS in 11m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bbcd6190b0b43e899a85d4add158815 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/325dee0a2dd24f58a1ecde3b5344f0fa : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf2f871d18d74f57aa639e0f4022edf6 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13c821af9cbf426cb7c9be62cf12b118 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d79f5b0f346444fb9353a5a8f74fea86 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8b6743c1d14bf3b92c83cc8cd45348 : SUCCESS in 18m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79f7a907eec148b49cd97b81d1d52c1c : SUCCESS in 5m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dff08c96e37d404ba91cea864e3ba365 : SUCCESS in 9m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b13daa86689b489284c6bddabd4c0e67 : SUCCESS in 8m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be455dd78ef5427e9b899f6a5423b0fa : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b51ab12c2af4488b0338e47308dc401 : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8a3100007014a5b85b129799aac5e9c : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/65ae5aa7ad9e4911bc43c173611cbc72 : SUCCESS in 1h 43m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c4a808ec8824319854f47881cf50f02 : SUCCESS in 1h 52m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2e36f5c1020e43a8992a288bd22286db : FAILURE in 1h 38m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f9db30bab7d8407b932c0a7492661f6b : FAILURE in 14m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f0a5fd902b34412bb10719ad3bd9a05b : FAILURE in 28m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ca0ae4723e346aeb4d939a2e32b1017 : FAILURE in 47m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8fe19212cf154ab0a24f7f7f40120e9b : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8452a1fccafa468181f0f5f310c75b6f : SUCCESS in 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f6b8126cddc4796929f18b243d6115f : SUCCESS in 2h 14m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb9e68b88966423281afc4142d68508f : SUCCESS in 44m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a55efc227bee4b32b24d548dc2a2e2bc : SUCCESS in 42m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2a868476074f4d57a823ab2435d4ace0 : FAILURE in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34833accb3e536b4a72b15cc96029a3cf91a6257","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:25:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f4e6304c59d90aca43c02f72535899471aa7fc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:51:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abc3a48246234f43b6ce323d611006ac : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e739ea43e4e744a397179ae3bcd732c6 : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7ab816f9714a329554cca58f5f77c2 : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1c6ab10bb584b8aae88ce9570e541d9 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10d7355389174bafac9e7c7eabb476c6 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19639afd51e94dda91d18ed486059daf : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd6f2f77981467d8343a5c18f91cd14 : SUCCESS in 19m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65d485254b2a469cbca68f4559129ccd : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1867d24b96b540b78779b60d38d58968 : SUCCESS in 6m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3fb4a60c64184b78bb28703784030b15 : SUCCESS in 8m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4fbb996fcd484552ba2b89df34d6121a : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/52a969d8d165441bb2d7a49ec425c10e : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7b3d532cbc5479bbadf8e77b0b3b33b : SUCCESS in 1h 36m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/705e519672454e738498cf72ab121ed7 : SUCCESS in 1h 36m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e29184977a94469ab16824e84361ee92 : SUCCESS in 2h 01m 05s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c357ff14565e4235937b9bec3531340d : FAILURE in 1h 30m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a6c089884dbf43e1a05ad8f824fe4d92 : FAILURE in 15m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6dc7f3f6fd414a19bd51ace32c14011c : FAILURE in 27m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/eced6fa0b908453f94c39330ae95ab39 : FAILURE in 42m 37s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f9e9379bd4bc482ab4372296ee8c75aa : SUCCESS in 1h 31m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a634205252f40c6b6b92dfe7dbc29f0 : SUCCESS in 40m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/75b170cb6b554cb19676115bae64617f : SUCCESS in 2h 23m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2e47f9912a604cea96894fbb68bb9a50 : SUCCESS in 39m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bfff33060ab741de8369f4b812663a6d : SUCCESS in 40m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/59a1cf1d82a54e6686e17ece37998bd9 : FAILURE in 11m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e020e9dce12163f859b75eacbccb6607e89cd7c","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-30 12:43:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62ccb84544a2f19f8d6b54059efac8809228061c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 15:21:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/294e6efd0a514b6296f126ba12f4bc8b : SUCCESS in 8m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed20b4a08e414233af094427c12ae31b : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6291baa76504bdf9974cee0dc9f010e : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e7f33b82b974455a52fae431962b680 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcd4f1ad7b3641a5bf3a7a745b79ba86 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/150c2c89bacc49149092732b71847862 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe49345b55b41fcb6c3a5510ccec118 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e20d1e4188b42a5bba0aa868244eeac : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2727ae231b3a4eed8c0d510c0965abc8 : SUCCESS in 6m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a7d584dbc6764baaa441c0096781f48c : SUCCESS in 7m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/886821a441b64368ac0a377e560380ae : SUCCESS in 9m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b63d907add244e3783305ec3c636bf63 : SUCCESS in 12m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3f34c6cee0f4dbd86f38041b77e9fdb : SUCCESS in 1h 32m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/143e6207680b42f3a2f3aaafa89281f5 : SUCCESS in 1h 42m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e4e4df609e24e81ac9aaebdcd785ff1 : FAILURE in 2h 29m 22s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc2ab237aa194f5fb5300fbe00b471df : FAILURE in 1h 35m 42s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0ee7d1ea093140d6b3ec523f5646b31b : FAILURE in 14m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e00ef4f00d324730a7b4141f388718f0 : FAILURE in 29m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d8fca9d149b4eabb70673b850bf82ec : FAILURE in 45m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7d3ee95bd5b84c7c9ed5835abc32373e : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da03e694f29441e3b32f54e526c6a33e : SUCCESS in 42m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4eab28fde3a94954b53c5d9b66ec1d2a : TIMED_OUT in 2h 34m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33b8f8a0fcee43aaae89cf6a239a8b8a : SUCCESS in 1h 01m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5955b46f253243c3961fbf1ecdc2bd83 : SUCCESS in 35m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/09bf1b6ca4f740ab9b7b6660ef70af83 : FAILURE in 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"940b355b2bf0240b6d0fe2878ce7991f7124c0e1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-30 15:48:49.000000000","message":"Patch Set 1:\n\nrecheck\n\nComputeBuildException: Failed to build compute instance due to: {u\u0027message\u0027: u\u0027Exceeded maximum number of retries. Exhausted all hosts available for retrying build failures for instance efc522b7-846f-4801-83c0-5b434a89d96a.\u0027, u\u0027code\u0027: 500, u\u0027details\u0027: u\u0027Traceback (most recent call last):\\n  File \"/opt/stack/nova/nova/conductor/manager.py\", line 651, in build_instances\\n    raise exception.MaxRetriesExceeded(reason\u003dmsg)\\nMaxRetriesExceeded: Exceeded maximum number of retries. Exhausted all hosts available for retrying build failures for instance efc522b7-846f-4801-83c0-5b434a89d96a.\\n\u0027, u\u0027created\u0027: u\u00272020-09-30T13:27:11Z\u0027}","accounts_in_message":[],"_revision_number":1},{"id":"86ec369400c91497178b2229af6adb068c1437c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 18:34:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99e602dc92234817a507513af6d1fd4d : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/186d59624a7b466da458fecc845a692c : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560810c6adf846e7bda132e225b63e0f : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c4dd942372a4bc58bec624159827206 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf529a5827d4d0496b66358201eb617 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/065c75ccb502490e958961e05e59eef5 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03dbaed746194e99a6ecf04a4f0b12d6 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/746aaac01e9f4976823794283e9a0a72 : SUCCESS in 4m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/93f70aa213764469b7399736f24b369a : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d296d412eee64431a6d5e1b45a84c3f6 : SUCCESS in 7m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48c90667789f42c1827177e17d8750b0 : SUCCESS in 9m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b62683310ba14ad5ae736fc9ff7e0b11 : SUCCESS in 8m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93f5ee7449824a2b87a9957642779cef : SUCCESS in 1h 42m 25s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4978f4836990429c9509ec98fa66b207 : SUCCESS in 1h 59m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3a5c4e918036486fa51dd84c822a1495 : TIMED_OUT in 2h 33m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6bbba5591db6454a8271e75296ff1a40 : FAILURE in 2h 28m 07s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d22f3a800c444c4a9501d655882f179b : FAILURE in 14m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e76eacedc584cabb0baf50bb83dda62 : FAILURE in 48m 08s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cf2ee479495740feb00816884cf8189d : FAILURE in 49m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/198993e627054ab6acb2ab9940d5bc2c : SUCCESS in 1h 15m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1989a18018c44781b7de7992c85ba921 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4e44ef98bb9743ad903f37b0543e944c : SUCCESS in 2h 01m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d20a8bd4b8894abca99fa797dc220963 : SUCCESS in 1h 05m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f3e19daf0acb4be78de46d0ce0f9ba11 : SUCCESS in 57m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c96828ef55a345b783bd4abb03370b02 : FAILURE in 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93e624599aa9a765f214eb54b585bd7907ed9f4b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:32:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"32e793482dab6da98fa440e52aa93f75d038da1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e90ac8b9ce354feba8759e729474401b : SUCCESS in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6172a1e75361480b85a6cd32168a81a2 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11aad6ce4b634d2ba5357c37e4b3ef08 : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c13ed00cf154b31b8415c0892a6df84 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e4d0b77e6c845df968a2f0b068afb07 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df762f3c7f7740aea7c9688325eea18d : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5e089ffe36477aaf2dc32b71cf4ce6 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad4464abedfd4c8fb3831d0d8dceab12 : SUCCESS in 6m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d50313cc10c6446cb121c1f4879da422 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/93da5e5726344a94ab8df61f3dd22c38 : SUCCESS in 10m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/077a7fd1130d4dd29b9afbd5df50d6b2 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e922af4316a34499bfe626790db3e1ed : SUCCESS in 10m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e3f5cc6aea344b3f9246d640a70fa22a : SUCCESS in 1h 39m 37s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/443bb18c9a184694adc8026114ed55a0 : SUCCESS in 1h 56m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04a0a588495d4ed8b615fb004941e91e : SUCCESS in 2h 22m 24s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/712f0350a6e34b2e95c72837cfeeafc6 : SUCCESS in 2h 20m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/13ddaa2ce9f447dba8008cd2a34b91e1 : FAILURE in 15m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13c74e37278c4e98b8d76d9794e0c012 : FAILURE in 30m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/01b637c8a6b14f78b424ff715dc2c222 : FAILURE in 50m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0d855cae1c94c5d869d9b199df8dbcc : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d49f5e9d636549708be20f99f8c9e076 : SUCCESS in 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/49c59d241c66476e8093aaaf39b24793 : SUCCESS in 2h 15m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68cce7e270df41a883a67338c98beaaf : SUCCESS in 44m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/761baa6aa19e455ea08494c4aa68c9cc : SUCCESS in 40m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/dc61c651489d4c468c1cba9ece26b5c7 : FAILURE in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50397f7803ff4cf88d68b739174e8aeb7b6a2253","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-23 10:51:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"18d3face0dd294ee614b61c382e89aa7e0b2a1df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-29 17:53:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7a3dd8bfed9c4b0d2b00e9350d8f91ed36287dc9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-29 19:22:14.000000000","message":"Patch Set 1:\n\nAnd since it\u0027s a backport you can ignore all my comments...","accounts_in_message":[],"_revision_number":1},{"id":"aaf0173090045b3e869938c85a50b72c09205d2f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-02 20:16:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"273f4c296abeaaf281c7e85095b68abf534a62f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 08:42:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3fca429e4464d486f1d4c9983f5ed1d10e8e9c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 13:18:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3db27d7b1fe54d868fcd3c4bc6dfd145 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5397ed1c368c406ca4dbc791751975ea : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e47d7eb41742cda738f91021b7e040 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e45b513c90841d2b99d7281d1777ba4 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/261a97c7b1134792888c88f36983e498 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b845c185c7434025845a70277dde94cd : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385f8df3627547b391242f352346d844 : TIMED_OUT in 31m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46510e754bd5406496c9943cbfb6aafd : SUCCESS in 19m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0b24bd1654a546699f78e7fedfd71505 : SUCCESS in 17m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/aff92de4da6b405095f3f2fdbe5c97dd : SUCCESS in 8m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2b45bc06465d440fb2e12d62e5ad7ef8 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2939517a6291422386d3249394abdf5f : SUCCESS in 8m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3f786bd75554a25b56164e062cb48a3 : SUCCESS in 1h 23m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0669ee8bdba54df981919005e2a077fe : SUCCESS in 1h 17m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acde19fe80a54a34809ecb91cc9e407e : SUCCESS in 2h 08m 41s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/02e97cc83e3646b0a85e43bf0311bccc : SUCCESS in 2h 02m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cb8368e768a14ad6a0fc4eea216edb5d : FAILURE in 10m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4df5082df54d46b6b0cea05b30985a51 : FAILURE in 28m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/a764eaac7cf84d158dbf9fdc31b91af5 : FAILURE in 52m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97d19105846646f7938256d87cb04bfd : SUCCESS in 1h 50m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6c1249d5e48744b6835c1199bc6c84b9 : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b66c7dcb95c240b69549a4ffde783597 : FAILURE in 2h 15m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7134fa8cf0147e380c7759f7e72e9d5 : SUCCESS in 42m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/98a51e343f494543835a2939080d0cf0 : SUCCESS in 58m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/e26ce3fcc3ae48f9a77ab8238d845e16 : FAILURE in 9m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"90cec087a9f302e8e26f3e1db3d32c71c5e4ff1f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 16:27:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"16572ad0b9bf133ca3cda75dd7bba35ef3356a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 21:34:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb9a11e9e0043ec8cf3f608a7a99bf1 : SUCCESS in 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5155f11afe9546d1b146ef405cad4487 : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277fc7b665a5492a9772f5ed67f35a2b : SUCCESS in 8m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66c0a1870ee748d29738988f9c619e37 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4b6c363ab24ef89c8d2076e8e024af : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dded3466241469e8dcd73d76d531ca6 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9de92c1cb054b488e95fd308eeba60b : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d955d726e0ea477781b0af9abfe49cab : SUCCESS in 4m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1157a49ac0464876a5432c3a3760ca65 : SUCCESS in 8m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f4f01b1b9079402e9df68ce4a68a4c79 : SUCCESS in 9m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1b2dbf459e2442ebbf9b3d654395a077 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aab7fb42d38f471f8ba2d9964cf5d2ff : SUCCESS in 10m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56eb8352b3d74c1fa57a5084572458ff : SUCCESS in 1h 18m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/7bed31ead678464aba5cc83b0970e60e : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30cdef8981ed4cf0a17f3dd3b2df79bd : SUCCESS in 2h 03m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b69d9d9057e14fabbcb5fbcb8cbda59e : SUCCESS in 1h 50m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85d8d1382926404b9032c7e1a7c25847 : FAILURE in 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa318299669e4b66a36a01f1c29d939d : FAILURE in 31m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/efee21c1720c4d8dadb34cc280557057 : FAILURE in 59m 49s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c86a9ae6ad4a401588b28f830367784e : SUCCESS in 1h 17m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dc88fa5b05d444e9b433aa2661681d7e : SUCCESS in 38m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ab64464994e4042aa6df7b7bb40e6ab : SUCCESS in 1h 59m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f746b6cafea4f8d85b19b6c99c0f6af : SUCCESS in 40m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5cdb03f3d2f8456da5b46e0c89555c17 : SUCCESS in 32m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d95251a057fe44cc9ae8d63ca7484989 : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17c2ebc25991ea9d962c378685b9de389664a4de","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-15 08:49:55.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2739aab078f2302e85c3c1381bc0075c731d3ab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:50:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb58fc9c37a670a641bc945a8ee1251d4b4f0c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 10:47:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87b03c6d791a4a3487d5b3c20abf91ca : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9971f0904234559bf27c8522e0caf4f : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c249d60e3744941a49492ea72129a2d : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/809363ec125347f3bba9df8b3bd9558a : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b4891d65c404abba9d10fdf68c86c77 : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a54feef57a487cb5c8ead0abb9c5fc : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/979492dda5ff4da684c0b5fa0b4d7f6f : SUCCESS in 8m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/954f783f8a204ec7a9a2716ee131e4cd : SUCCESS in 9m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4ed88a2b89448ed81b1f03172fa8502 : SUCCESS in 11m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a78833f706ac460d930f5738d96e83e5 : SUCCESS in 1h 19m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/97ef324036b74978a58cfdefa026f1aa : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/247dfbc443114aa9ab73002c1adeab2a : SUCCESS in 1h 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61040ac9808944319efb0a388d80d814 : SUCCESS in 1h 17m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c23b85a63e43ed835bba75c1e5c19d9673ceb59c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 10:47:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7fd73041cbbd43f0bed34f4761d02231a5cc371b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 10:49:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20bb8f3db7d43379dd458883b654383 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d94659fb579a48a4b1c9996a1b73aa76 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"564eec76d1f1059b7ffae3f0dea1f44c50a86a89","revisions":{"7ec6d2c0b16de151b79069404a7823c7586c6c22":{"kind":"REWORK","_number":1,"created":"2020-09-21 06:29:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/72/752872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/752872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/752872/1"}}},"commit":{"parents":[{"commit":"11d5096babc776c6c2688aa7155c5367419c8034","subject":"Fix operational status for disabled UDP listeners","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11d5096babc776c6c2688aa7155c5367419c8034"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-21 06:28:35.000000000","tz":120},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n(cherry picked from commit 9fb58eb9f4d2574106a48e3cd3c439d5345517f9)\n(cherry picked from commit ea47a0efadb42d9c052fa7ae114e8f82a4be2775)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ec6d2c0b16de151b79069404a7823c7586c6c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ec6d2c0b16de151b79069404a7823c7586c6c22"}]},"branch":"refs/heads/stable/train"},"564eec76d1f1059b7ffae3f0dea1f44c50a86a89":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-14 08:42:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/752872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/752872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/752872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/752872/2"}}},"commit":{"parents":[{"commit":"e561409b12e92a3ebd68b35b9c5ef1e2c6ad4b63","subject":"Merge \"Fix operational status for disabled UDP listeners\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e561409b12e92a3ebd68b35b9c5ef1e2c6ad4b63"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-06-18 16:27:10.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-14 08:41:52.000000000","tz":60},"subject":"Fix invalid DOWN status when updating a UDP pool","message":"Fix invalid DOWN status when updating a UDP pool\n\nWhen adding a new UDP member or a UDP-CONNECT health-monitor to a UDP\npool, there can be a race condition in the first heartbeat message\nsent to the health-manager service.\nThis message might contain a DOWN status for a working member that\nhasn\u0027t been checked yet.\n\nThis commit introduces a new member status between the amphora-agent and\nthe health-manager: it indicates that the UDP pool has been updated and\nthat the status of a member is a transitional state, preventing an\nincorrect ERROR status.\n\nStory: 2007792\nTask: 40042\n\nChange-Id: Id9e19375ebca6a720e6a85006f5e8948d3aed760\n(cherry picked from commit 9fb58eb9f4d2574106a48e3cd3c439d5345517f9)\n(cherry picked from commit ea47a0efadb42d9c052fa7ae114e8f82a4be2775)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/564eec76d1f1059b7ffae3f0dea1f44c50a86a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/564eec76d1f1059b7ffae3f0dea1f44c50a86a89"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
