)]}'
{"id":"openstack%2Foctavia~753169","triplet_id":"openstack%2Foctavia~master~I88e64fd5e8a4c709f735be85b85139dbb52e4acd","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I88e64fd5e8a4c709f735be85b85139dbb52e4acd","subject":"Fixes API list handling of unscoped tokens","status":"MERGED","created":"2020-09-21 22:24:44.000000000","updated":"2020-09-22 19:28:10.000000000","submitted":"2020-09-22 09:28:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753169-1600766892025-fe4ec94a","meta_rev_id":"cf7c5d6aecbc7e06ee175f183fe4ad3c4e14143c","_number":753169,"virtual_id_number":753169,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-22 09:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-09-21 23:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-09-22 07:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2020-09-22 07:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-09-21 23:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-21 23:45:18.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-22 07:34:19.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-09-22 09:28:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2305a1e16564fb9efcd620abb5972fad55605bd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 22:24:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7181d5495a4253f740409afa7b1de53b38f4234","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 22:33:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e96cfb7c07144acfa32e4c33d3fce05a5d01b219","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 23:12:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2eb559b03eb238e30abe6cb2ec322436a3a0b199","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 23:17:39.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"f11c23e5c6d38e383eda665a6c4b2efb0a7eb1e6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-21 23:45:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6295c0a3b3e31a73fad7c4567eafcd74a1a96b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 01:39:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5471dcaff24c49998f4b7cda1a20da9b : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e4d09e898dc4bcb897d4cf29d920ae0 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07520fd1130d489fae377417da00ef9a : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3177745339d42eda3219210e7d42d7d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b0ad7187c5a4a88995331a7525b33f5 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c77292a94f4f51b2d4fee39ee78e8d : SUCCESS in 19m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/375338ddf9ca42a6935678aa0611da1e : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b778b4ba7dbb4cf2b551cb1b4326f173 : SUCCESS in 5m 12s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/05c752b21ce04ee190b2add2e265198f : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a62e866e4a4458880fdd1c50da9f851 : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d0c6a80b99b4bf09c057ac18320d543 : SUCCESS in 1h 37m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d953bb5962f641cd9502d5aed418e4e5 : SUCCESS in 1h 46m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/be509ddd123c445d806063dad8026865 : SUCCESS in 39m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/61e4004effc740e0a6da2f8d15e59493 : SUCCESS in 37m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e8cfdbbb02740ae84a77aeadc21dc41 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4838f3c043444f1ca4a069e359c073de : SUCCESS in 54m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c6f67894e40d4454bfef20735c5b861a : SUCCESS in 2h 21m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6aca2323b4254aeaabcc7d47a0221e66 : SUCCESS in 1h 42m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/47578aa447dc4b0c9fe5506554ebf1e1 : SUCCESS in 2h 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/34d7739cb4824ba58c95039b1c85c353 : SUCCESS in 1h 47m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9a009c6ca48049e59159036a71f06ca2 : SUCCESS in 2h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2a04ebc6334347e694df4d3fe2801964 : SUCCESS in 2h 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"602f002ff002ab8c1d3fd173a3816a82e9395da8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-09-22 07:34:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5126d206633f1be2f017f241badaf0cdfd801a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 07:34:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9cf80c54aad9a0a60e2a94e772fc9120f7eb3986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 09:28:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/030c9f53bc7c467eb30d9933b278dfe4 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da01307540d64eaf9431a0526726e4fa : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9838db5dfb1b42f3ad57691bf493aa4b : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b39944ceb3474edaa579b51a38cb60b3 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6628e843134230a5a5092ee078c0c1 : SUCCESS in 22m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2f0e9e28a9544978348b156e703a10a : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/16a56da5f9fc42f29a5ccbcf159aeb40 : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5a743700bca45f6a0826cc5989cfede : SUCCESS in 1h 45m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0ce58aea83f430f9adecda55451b387 : SUCCESS in 1h 48m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f80d18fa963e44a7a249962b4ff352b6 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64f07fd79da44b8ba8ce06fd5c07ad23 : SUCCESS in 42m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b9bbce3d58e417089cc1ef500650b8a : SUCCESS in 41m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b4602db5e13c4c1ca1cfa85d6499cdca : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a68d518b9db870aff92f5616294c19c4a5ca05b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 09:28:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3fff835786a666ae4bc1e268ff63b8c86a22dcf1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 09:30:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792c6455bec64c869d97b232f553f64d : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17a8436fee6b4b5eb01d6de2c794a330 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2},{"id":"2557c2dd67eba4316f14453fdd33309ba0fb8c9f","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2020-09-22 13:39:06.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nimo checking project_id is not what functions set to do which is \"Check authorization to list objects under all projects\"","accounts_in_message":[],"_revision_number":2},{"id":"2384c0ece478ecf073ec5c02dd705aa0868e53e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:32:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99af9cb5b4ab5faf327a695272f6868e475deac4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:07.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 8de7e1ec037caa082a0d9a5dc29d23a8bda3713e","accounts_in_message":[],"_revision_number":2},{"id":"cf7c5d6aecbc7e06ee175f183fe4ad3c4e14143c","author":{"_account_id":26647,"name":"Vadim Kuznetsov","email":"vakuznet@cisco.com","username":"vakuznet"},"date":"2020-09-22 19:28:10.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nI see. Thank you.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9453701fb46f0ad23dae00b3a3792bb379978d0e","revisions":{"a175123ca1d7f49e353a04f47cc9add47ed902c2":{"kind":"REWORK","_number":1,"created":"2020-09-21 22:24:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/753169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/753169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/753169/1"}}},"commit":{"parents":[{"commit":"e45ecc2111efa7c860299809c8d9d629dd9e47ee","subject":"Merge \"Fix requirements.txt for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e45ecc2111efa7c860299809c8d9d629dd9e47ee"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:24:39.000000000","tz":-420},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a175123ca1d7f49e353a04f47cc9add47ed902c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a175123ca1d7f49e353a04f47cc9add47ed902c2"}]},"branch":"refs/heads/master"},"9453701fb46f0ad23dae00b3a3792bb379978d0e":{"kind":"REWORK","_number":2,"created":"2020-09-21 23:12:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/753169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/753169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/753169/2"}}},"commit":{"parents":[{"commit":"e45ecc2111efa7c860299809c8d9d629dd9e47ee","subject":"Merge \"Fix requirements.txt for Victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e45ecc2111efa7c860299809c8d9d629dd9e47ee"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 23:12:17.000000000","tz":-420},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9453701fb46f0ad23dae00b3a3792bb379978d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9453701fb46f0ad23dae00b3a3792bb379978d0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
