)]}'
{"id":"openstack%2Foctavia~753193","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I806d220236ad741b638ad9315537334f2d923031","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I806d220236ad741b638ad9315537334f2d923031","subject":"Fix amphora failover when VRRP port is missing","status":"MERGED","created":"2020-09-21 23:52:25.000000000","updated":"2021-01-21 11:36:20.000000000","submitted":"2021-01-21 11:33:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753193","meta_rev_id":"b79f1160bd7668b5ee31799be26a615ad8e82afa","_number":753193,"virtual_id_number":753193,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 11:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-21 09:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-15 09:11:24.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-01-14 16:28:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2021-01-14 08:57:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-01-21 09:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 02:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 09:27:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-14 08:57:37.000000000","updated_by":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"reviewer":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"state":"REVIEWER"},{"updated":"2021-01-14 16:28:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-15 09:11:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-01-21 09:29:48.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"5172b86384a5b38d62674c42555733be5d605b33","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-21 23:52:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbb77ac0958bf1e674a9079cdb597fecd46924d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 02:19:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f223cf0cbf6a4bafaecf38af4efb5b53 : FAILURE in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fedeba4599e740869f76c3caaf9f02d1 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df98172da51f495cb52a64bc262aeeb6 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c360f1d1e4f54058b7970fb97a763e00 : FAILURE in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf42bc8036b04a3ca7985ab03f1aa806 : FAILURE in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5637fe30749481795314800eec20f44 : FAILURE in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6742b3a2fb8e46a593e64cd6355ff673 : SUCCESS in 19m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5144270f1dc64704bb0295cfae7760d0 : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/299b9cbe3c32412da7c47813deb1664a : FAILURE in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/54dc8da42d8a4608b0247909b8bac238 : SUCCESS in 7m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c141f940adce43b7be2e608e0db04029 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/867a858327f44e7fb959a35d57dbad20 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6a6e5bdaaa684086aa8f36c559655169 : SUCCESS in 1h 56m 49s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5414bab7ee4349fdbb6e14db7a5d5234 : SUCCESS in 1h 48m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/483abb80ab55421daaf057c6a40e3737 : SUCCESS in 2h 21m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/624b279a427845a78821286619bf4aa7 : FAILURE in 2h 19m 58s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9a1cdcc4580843d6a46fa3400ec5520e : FAILURE in 11m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e4135a22a56b4e36b5649ce586b399fa : FAILURE in 26m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/56e15b87761146e4819175f5aab1b34d : FAILURE in 1h 01m 13s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce5338ace3cb4bb9970bae0cfbf22436 : SUCCESS in 1h 30m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d2af2a1d4ae45cb834a6627f33b668a : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5e00b4201c4d418682e6439dc4c93485 : SUCCESS in 1h 55m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e3558c55c71345a18d358d06401b7bb2 : SUCCESS in 41m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2fc11be46135466589308682ffdcd11e : SUCCESS in 38m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c6687865f71149b088946158a8679f5b : FAILURE in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e705fd79b77259894ec6bbc09c90db472b5d5682","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-26 09:27:09.000000000","message":"Patch Set 1:\n\nrecheck\n\nresults are gone","accounts_in_message":[],"_revision_number":1},{"id":"e61a81250312c375c60f8688c514a14eb0435583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:53:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f6300dad134104b67105b708d15e56 : FAILURE in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f29fca2a46ff4caa98c2f255b0abec28 : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4947c9c7bc5545bda2f54d528add454f : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3bca533d7cd04b4cac3aeb67705e5a46 : FAILURE in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5191ee33ca37459380de47c4dee5149f : FAILURE in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ec1199eba7f42bbb29cdb8b3b50e574 : FAILURE in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3abe47edd2444b3e8efcc62b3e78ea9a : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/086a6a42822f4ac3b1241ae983dfff59 : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e5a109065a6c4305b175065b2f1ea4bb : FAILURE in 6m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/66928be9b39d4ee9958fa6b30244a3b5 : SUCCESS in 7m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c9272674b4144e04b902eef88925d5a7 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2fc157eae5ce45ecbf210546aaed3c10 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a7a775f6a22643179b9b1eeac9818b1b : SUCCESS in 1h 47m 39s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/998738c8584f4915a9b22004fa6f630c : SUCCESS in 1h 52m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac1fcf02cf8f4834b900a3a2915f3003 : FAILURE in 2h 16m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bf6b512a075a47778258ac5f51ef48f2 : FAILURE in 1h 39m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/770681fcd04e4e599856ce6afca928ca : FAILURE in 12m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b575ff45617e4597bdf84cd8f5162746 : FAILURE in 24m 22s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5383ab9a7ee747f9b70b3ac819d9da18 : FAILURE in 1h 21m 21s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de61ba2d9a7248dcbd44edd988251d91 : SUCCESS in 1h 22m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/46a13c8566b54c4a9c2a4525616748f8 : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bb789d90c7e6431f87a6fab3b9b0109f : FAILURE in 2h 04m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d6ab01b2f9240de836674d463eb5f1c : SUCCESS in 47m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eb4403b20d8b4617a3f6f074da566411 : SUCCESS in 54m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3a0b5f42530e42f9b81df7aec0ee1b41 : FAILURE in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d678d8e3c7b93f3a65939c2abded27dae2948d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 07:38:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa665656407ec6da2d8f8a343947cb40cac06ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 08:40:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba964b62fb9580d313f237a79cb8ef642f029fc4","author":{"_account_id":30636,"name":"Chung Phan","email":"chungphan7819@gmail.com","username":"archiephan"},"date":"2021-01-14 08:57:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9b07d3cf4579eb976e24a379e7ffdc7cd2ce58e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:02:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44e4a3761df64880a40a927b70e5bf84 : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9836b626876428eab118355b7f7d069 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b0dfb467a9482a9dce1ad9871888e2 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea9adab65cbe44b29bfcc2e32a52b6e8 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98862b1fba19463c9d3b170f239bc198 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4784d48bf9d046a0be8d11183ce52dff : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56bfed95b26b4b46b7a15a6f43467136 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b478893efe947488e04fcab1aec2752 : SUCCESS in 5m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8b35f2547b124fe3b2e982c6d0aea6a7 : SUCCESS in 7m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/39b408d627f04002b5ad78570a36e1ac : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5eaecda97e6c4db0b19e440ff950d88e : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d2cea49004444e294f79b3b1aa7e106 : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53c31b3a22244091825273a0cf53c09e : SUCCESS in 55m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/30ddab24805a4bd7b61e6c5ed3f14c89 : SUCCESS in 1h 45m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e9a02abe22ee4dd48fb77ae56286d806 : SUCCESS in 2h 22m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/772a89269c6d4c6a80a64ef2a3a247e6 : SUCCESS in 2h 07m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/eb4476e4e0b54078b3d9ac95a8d51254 : FAILURE in 15m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b9e081adb3148a69551e532d32cd843 : FAILURE in 27m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b13c3451cf845d682e504d57d17321e : FAILURE in 50m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f24965a66ada4b2b942591f7873ec0d6 : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9eec6785378d4a53919fec3a04940d8c : SUCCESS in 42m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6feda742cb5541c2ae8ef4c48e058d64 : SUCCESS in 2h 08m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cdbf59ff74ce470c902fcbb2268db42a : SUCCESS in 41m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/ffac2457e9484b34bc89d2e972e997f1 : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/80b8a154f29b46adae773844fa796e9c : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b86627c0878df549aeb3df733e9c96045a5e14f7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 16:28:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad3bb12f8b8c209ac72922e1a74075dd8c568e0b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-15 09:11:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81439c76b3551a1dd3a654c68fc78cb27de90dd8","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-21 09:29:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0bac800d8642610d17edb9839d8c96c797c75a25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:30:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"33dbae0748e1a92ff14e631dc65eb46cb7dd41eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:33:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19c23622e0484a4abac5c01de4332105 : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d7ee0455704844a0a94ae2611ea9b1 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aca328e0fa8646e79de4cc7110027e0a : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9be9411e0c854315856d2dd77ce9c094 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b7ce96b65974103ac1540c11c618407 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0f4aab8b9241edba8c386f939851ba : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5706d9487aae4c47a09b84a3ebef70ab : SUCCESS in 4m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a9bc302989144666974215d8aac7f72a : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a29be11812d949638e1050b9b3a2904e : SUCCESS in 9m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c1c97af39494ac1bb7564f5809566d4 : SUCCESS in 1h 13m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d6328eaa85e04c1cb0b5e276b11928c1 : SUCCESS in 1h 20m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5408f859ea9e4ab391cfdf73ee7550de : SUCCESS in 1h 56m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd08890cdccb46c4bfdea9a802ab6d74 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":2},{"id":"9925a75e976231d037bea5253bae3657d4a5f440","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:33:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b79f1160bd7668b5ee31799be26a615ad8e82afa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6babc3ec5d4fa7b847b955025fc38f : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1c91e49638241549ed6e7eec760098f : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea7351efe96fcf49ad42b4a62ebe7df70cbc9902","revisions":{"73a1785a43e097d5787dd3adc46b1a5574f953c1":{"kind":"REWORK","_number":1,"created":"2020-09-21 23:52:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/93/753193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/753193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/753193/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:54:09.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 23:52:24.000000000","tz":-420},"subject":"Fix amphora failover when VRRP port is missing","message":"Fix amphora failover when VRRP port is missing\n\nIn an amphora failover situation, if the VRRP port of the other\namphora in an active/standby pair is missing from neutron, the amphora\nfailover may fail with a \"port not found\" error.\nThis patch corrects this issue by allowing the amphora failover in\nprogress to complete even if the other amphora is also in a failed\nstate.\n\nStory: 2008127\nTask: 40851\n\nChange-Id: I806d220236ad741b638ad9315537334f2d923031\n(cherry picked from commit 66f3a63f442bb790dea5c529aa56a462d9515363)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73a1785a43e097d5787dd3adc46b1a5574f953c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73a1785a43e097d5787dd3adc46b1a5574f953c1"}]},"branch":"refs/heads/stable/train"},"ea7351efe96fcf49ad42b4a62ebe7df70cbc9902":{"kind":"REWORK","_number":2,"created":"2021-01-14 08:40:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/93/753193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/753193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/753193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/753193/2"}}},"commit":{"parents":[{"commit":"e561409b12e92a3ebd68b35b9c5ef1e2c6ad4b63","subject":"Merge \"Fix operational status for disabled UDP listeners\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e561409b12e92a3ebd68b35b9c5ef1e2c6ad4b63"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-10 21:54:09.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-14 08:35:18.000000000","tz":60},"subject":"Fix amphora failover when VRRP port is missing","message":"Fix amphora failover when VRRP port is missing\n\nIn an amphora failover situation, if the VRRP port of the other\namphora in an active/standby pair is missing from neutron, the amphora\nfailover may fail with a \"port not found\" error.\nThis patch corrects this issue by allowing the amphora failover in\nprogress to complete even if the other amphora is also in a failed\nstate.\n\nTrain-Only: this commit also fixes the order of parameters in\nv1/.../amphora_driver_tasks.py. It was fixed on other stable branches in\n\"Update amphora v2 for the failover refactor\" (From\nhttps://review.opendev.org/c/openstack/octavia/+/738292).\n\nStory: 2008127\nTask: 40851\n\nChange-Id: I806d220236ad741b638ad9315537334f2d923031\n(cherry picked from commit 66f3a63f442bb790dea5c529aa56a462d9515363)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea7351efe96fcf49ad42b4a62ebe7df70cbc9902"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea7351efe96fcf49ad42b4a62ebe7df70cbc9902"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
