)]}'
{"id":"openstack%2Foctavia~753382","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I88e64fd5e8a4c709f735be85b85139dbb52e4acd","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I88e64fd5e8a4c709f735be85b85139dbb52e4acd","subject":"Fixes API list handling of unscoped tokens","status":"MERGED","created":"2020-09-22 15:33:07.000000000","updated":"2020-09-24 06:08:44.000000000","submitted":"2020-09-24 06:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753382-1600927565882-7e3b2230","meta_rev_id":"0b454ed8145566c87e93da09e0cbdd7b667c2946","_number":753382,"virtual_id_number":753382,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 06:06:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-09-24 04:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-09-22 17:54:28.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-09-24 04:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 17:54:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-24 04:07:06.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-24 06:06:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f383bbbff49ef2bf45fb6d47344110d62729e4b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"18962d06ebc138ab721a4233521ec7a3c05e5718","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 084be59e49961eb689832d19923a673736ffb2be","accounts_in_message":[],"_revision_number":1},{"id":"af3721e949cf3dd0516d92b70027e2ffd0288981","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 17:54:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2df02700ef016808637a26ff0d6dc50996707057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 17:58:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b8b9fe37c6343c89b001a79814ab7bf : SUCCESS in 8m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74c4d4b681e457394bc6d4adafa1986 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9e1b46e68a4b5fbc8601af6137f837 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32be1666a1644affb19b1b20ab0df0b3 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/752d81c333874634982c3683b03a3e04 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7f74ab9d0e408ab56756e267c76785 : SUCCESS in 22m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bb6191aa6cf4d9698a556efc70a14b3 : SUCCESS in 6m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8cb2f282c71549aaaf42e4c179a570a1 : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6af902c3c6bb4877913b9e1fed3fbaba : SUCCESS in 8m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13fcafe23a264730beb4510465ea9430 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78dd353f694e471a9a704220c0635a0e : SUCCESS in 1h 41m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/16649973cec84ccabb1a29ecb5a063df : SUCCESS in 2h 09m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/337f1eae0972488e84fdc3117f36db96 : SUCCESS in 39m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6fd463f52a304c288dc75508257ce479 : SUCCESS in 37m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/97c88c4136b647ff9ae4b5056c31b77c : SUCCESS in 39m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dff891803dcc41c69ce80aef9b8bb5df : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/13e4de1f1635431d90d68d1f03801c13 : SUCCESS in 2h 15m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90ef875a0f444b52a79884387c57405e : FAILURE in 48m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4f2f6b0339d74a4ab5b33481d9884d3b : SUCCESS in 2h 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29da9d59dd836216b8c8f573375dfe158e347092","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-24 04:07:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"017cbfbaedcf35c44fe968da295cc13f0448705c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:07:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"231835ffabec3c31e0c635d401988f9288cd84e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:06:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"023450407087b9fbcd1d7f8406c27e6ee441349b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:06:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a45786133a924c3f9aaa734aab635980 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fde39a040aed490b8c8b1ec75713de69 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e4cb20018d41c5902d96dbfb337100 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c404844ae7df45a78b73b6be825973a9 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367d350b542d43c985805b67481e72ad : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e8a106d73c843789edcc6761bf8e220 : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f58891f1b7b4f8abcf918d1933cc502 : SUCCESS in 6m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc2bfce99dcc45eea3abd659eb64dd26 : SUCCESS in 1h 19m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c57379c472f440eb9616ef4c207c3f13 : SUCCESS in 1h 52m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f327e088fac7473bb529c8fb867df65f : SUCCESS in 41m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bb1c54e2743f45bebe70035d9a5880a9 : SUCCESS in 37m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40ee3761c7f54398837a2801feb31021 : SUCCESS in 40m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1c69afca936422b91ff022083e3bd38 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0b454ed8145566c87e93da09e0cbdd7b667c2946","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:08:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff409d1b9b84a15940db2daf62eca08 : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/533b9087abbd4f33bb3386a05a974c09 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8de7e1ec037caa082a0d9a5dc29d23a8bda3713e","revisions":{"8de7e1ec037caa082a0d9a5dc29d23a8bda3713e":{"kind":"REWORK","_number":1,"created":"2020-09-22 15:33:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/82/753382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/753382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/753382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/753382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/753382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/753382/1"}}},"commit":{"parents":[{"commit":"2dbd8f696d79671a11f7a5a43e282720f8983f31","subject":"Fix amphora failover when VRRP port is missing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2dbd8f696d79671a11f7a5a43e282720f8983f31"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-22 15:33:07.000000000","tz":0},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n(cherry picked from commit 9453701fb46f0ad23dae00b3a3792bb379978d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8de7e1ec037caa082a0d9a5dc29d23a8bda3713e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8de7e1ec037caa082a0d9a5dc29d23a8bda3713e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
