)]}'
{"id":"openstack%2Foctavia~753383","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I88e64fd5e8a4c709f735be85b85139dbb52e4acd","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I88e64fd5e8a4c709f735be85b85139dbb52e4acd","subject":"Fixes API list handling of unscoped tokens","status":"MERGED","created":"2020-09-22 15:33:12.000000000","updated":"2020-12-04 13:39:31.000000000","submitted":"2020-12-04 13:36:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753383","meta_rev_id":"6d66d30638de751a0793102f7933923203e2d18d","_number":753383,"virtual_id_number":753383,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 13:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-12-04 13:39:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":2,"date":"2020-12-04 13:39:31.000000000","post_submit":true,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-12-02 00:05:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 04:04:04.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-24 10:37:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-10-05 11:50:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-04 15:07:02.000000000","updated_by":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"reviewer":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"state":"REVIEWER"},{"updated":"2020-11-04 17:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"258ddbbe6af4fc281e8edbf47a7c8a68f35fd6e3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"79feb699bd008a3a2d336ef82d43582053298bc7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:19.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0a6011f7784303f7b1d2f55ca801f2e27d146419","accounts_in_message":[],"_revision_number":1},{"id":"44f09fafa7fc239cdfe53c237f30d2ad1e6e91a7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 17:54:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8950942fbbd90270db8e8e15b336a99287ef69cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 18:00:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c2be176d0c34e3a9495488b86f30c17 : SUCCESS in 8m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6daf7f9640d0486189ad075d08893e57 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1208b2b984e4ce3b1556023b1f344f6 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ed1d3135a35451c86f0745559ff85ba : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e464bc7dfcb4cf1b0cf5ae2b4644255 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/331b62ee7aff4a0d96bc7c05274f90b3 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b330c7236f24e9fa9967971f4516fac : SUCCESS in 18m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8565b6dd2280414fb43636bfd318c642 : SUCCESS in 5m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/03b0703e5416423399e13ce90d88902d : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a73f5df67f7447a9b5577893e34a51e0 : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/303240f7ae9d4317a603adc5f53642e4 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f034736883bc428abf1a7b9d2df24c7c : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aea86ae034a24732b50dedc6bd7e8962 : SUCCESS in 1h 56m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3131e033b17748ba8acd47dbcc9227ae : SUCCESS in 2h 11m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f05cd54a9155451987b9fa3d794b202e : SUCCESS in 1h 41m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e75fd35cbef441c6ba439cb2544da4d2 : FAILURE in 1h 44m 28s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/85dbead44af1460d88b647f523c98245 : FAILURE in 22m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/efaa38b51a9a4e0b9ac843c6addd6d04 : FAILURE in 34m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/96c7143b684a409681c920da38954c45 : FAILURE in 46m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0051ea4b9f1c4b0991aa43a84aba2e94 : SUCCESS in 1h 37m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8608df625529492a8c4d0d487edfbc79 : SUCCESS in 41m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9ce132897ed542ac9ff6cc982c49e475 : SUCCESS in 1h 57m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50ca6229a97c4cdea6d392e3b0906baa : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/23ee8186929d4f11af7d36e923fad6b7 : SUCCESS in 34m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/172dea9f8d714d1796bb03905d630836 : FAILURE in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24e7933b0f007cebfec1b4f07b69b40bab4686fa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 18:38:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f0339a9079405ba351835cc006f9d7daa4430db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 20:49:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34aa321736e545d4a87f0ba86c1dc399 : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2184346979054999b537f3b31c6336f7 : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2f2732b63445a287f7c9d49c2962e6 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de542635a5134611abd8618e9ac004ed : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d3c259fd26349fca1b530ae77749962 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c582f6fc9cf4e84a88126be36834624 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/893c3615406541598517f182bb8d465e : SUCCESS in 18m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3faa56e5a10c4ea3944a06c200d1d2c3 : SUCCESS in 4m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3704578c74c44cf39ce3844712114339 : SUCCESS in 5m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4bb61cdbc96143faa0ebd2a82a3df844 : SUCCESS in 12m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e41818055ac24712bbd188cf31c54096 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bf0afe575b24e698749238587d771b1 : SUCCESS in 6m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/973bd3d95d4145a0b39ad55d32455782 : SUCCESS in 1h 35m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/dbdde960a7eb4937bb3b746fbe3860fb : SUCCESS in 1h 40m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d474c4579a54597a9fcaa9c3de883d2 : SUCCESS in 2h 08m 12s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/23bc1932b1c4437b98f57b2b43513324 : FAILURE in 1h 39m 05s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d76200c5e3ff45b9b3ca0f466e589e69 : FAILURE in 12m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fb9d67f4de5d4d5498faee811a223dc4 : FAILURE in 30m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d3ae60aba514af9bdf0ab7482ae9ac3 : FAILURE in 45m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3410891d268c4f0baec103fbab5526cf : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/06075215cf604a26a3469461d510e493 : SUCCESS in 40m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7c0b5bc234c5417ab7c3b7099f46ef4d : SUCCESS in 2h 01m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2007931176f543d7a8da067e13832d96 : SUCCESS in 39m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9c58faa714a0475ab339612dc7f1173d : SUCCESS in 42m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/03443889582f4c698f49fa36c63d18e5 : FAILURE in 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9c41f6cc3284af6bb0d60c5bc3cb5dfba73a756","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-23 19:54:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4884aa0735fe8f85595f4ec95058dd469538da3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 22:55:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1598cbcf9c940c78684d209f707f5aa : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33965fae8f624425a0033ca253ded1af : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab836a9a8e2d4ea6855c851d488a69e7 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fca0a8c063ff4e388a228f9aecafeb2b : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7884c433611144e29864b6e09e32a370 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/878f11f49a404ed2acb787805ab9aebb : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5719c2065574b57bd787efcdaa26d35 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed179a7ceb544df4beadbd6a166c147d : SUCCESS in 5m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/25da71a2552647cc92c59a775bea9d7f : SUCCESS in 6m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b9dd6758536a4aa6be6954262bd2deff : SUCCESS in 7m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0ee463cbd80a44b29a9c05fb2cd783ec : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a0db24859ccd4fca9476438c3be06c92 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9d37acdde4b34338bb01f1513a5d3937 : SUCCESS in 1h 42m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cf7b051221cd4477b3a6b2449d7f0f45 : SUCCESS in 1h 47m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69160988ee414a0fbdd29d846428a650 : SUCCESS in 2h 09m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5cbd2c008d79439f8b8e74335cd5cef5 : SUCCESS in 2h 29m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/212486f800dd49ebaf0dc6ac35616c72 : FAILURE in 12m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0adf47d622394c978512618746574922 : FAILURE in 27m 43s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8714676b2ecc44f58411d6aad687fecc : FAILURE in 42m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dcb075b413e5421fadcedc847c2615ef : SUCCESS in 1h 26m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/474dff1940ba4b7eb26e86ff50abbe73 : SUCCESS in 45m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b7b48ae1ff5c4109bed6a5284e0e9cb2 : SUCCESS in 1h 54m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/379fd9196d4941f0a0226fbbcb12ca52 : SUCCESS in 39m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f2cb8859efff462790cc6ad39f6cf495 : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/66dc88fc2a2248c09eedc1650afbff26 : FAILURE in 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aeb032b8fa6db3224f235c25ca9e0705002a76b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-24 04:04:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8edba269224b915cdd87ee2ac0382bd9005298e5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 10:37:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f33ca9dd0696a4234a9680afcbc196446e573e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 08:56:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ccafc5e6ea971f6c07ee6905d0f16306619c67ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 11:32:45.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"ed9e7cebda30cbf1ac642120ded838fd54d0bbcc","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-05 11:50:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6e27a2e002365a42ea1aa62a7ac70c62c14fdbe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:37:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee7fe866c3a049cfabf05a0852438461 : SUCCESS in 8m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba7319cad32248cc910c12efbe81bbf1 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bbbe66266a4d7799fee30e95457f9f : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c380ab4d194946639f0cf1ca47469418 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/117859c6dde44dbfbe79a8fc3c694afa : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4aa5202a7912451da7447e07963102bc : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fb9170284d4db2aa0b961d269965f7 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f41a5f34b48438cac04010d7a768d31 : SUCCESS in 6m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/51208f62991640eeb4d40614d43c348a : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d4f0813c19dc400aa506f55d94b6f08b : SUCCESS in 8m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd6201997b814f68b281574bc70c6372 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98ad261ccddd47f6904124b20bbadf1e : SUCCESS in 6m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8ba564ec7b9a43fe8caf79b3563619a8 : SUCCESS in 1h 38m 32s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/080adf10e5e64ab282878e34bd877baf : SUCCESS in 1h 54m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95900d818911434a8ab46c80596ab365 : SUCCESS in 2h 05m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7a27cc161e7c44da8ea3c604101fb50b : SUCCESS in 2h 15m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/284e3dc7e7cc4797a6287e84421c0752 : FAILURE in 10m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7be17c3f944455c8fe049f1af3f5fab : FAILURE in 27m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/16503d1a86224ebdad8d12175e3ee13f : FAILURE in 50m 09s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fe0cf4d6eee443378fbe279c7a07613b : SUCCESS in 1h 29m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac8b56deaaae4c9aa7f2ced96063d1d5 : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3567d0c008684d7f994763877e621269 : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac49bb2e241a46c88b290689504c2020 : SUCCESS in 36m 18s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6bbdb1b2135b409b897f942faf557170 : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0ad1820a52d64817b32267c303b04113 : FAILURE in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20d106f0766ce092d3f4348d981a63b3b433a011","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-02 20:17:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad9a9e13f49e6d91cf75dcde73bda14e4f00e535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 23:22:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ad495970cbc44c6a32aebf048afce61 : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/168fde3b115e47c2a9e95b6ae05a6b7d : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562054d720554e3599efb01cae90e98e : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6c23b5638c546a881a95f322de48c8a : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75e5105aa49e4697b278a0d020d1b9c4 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6865ef1db19f4426a97b0d547c134e50 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1509d834d984f598eab91514a5296c6 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0ee61c5596548df8d8a9df072d9c7a1 : SUCCESS in 4m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8aecfddfd8984e68b152ab1d93397042 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ad07b5a2ac84af298e5b93a8fb5886a : SUCCESS in 7m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f2a5e224dcaf4757ad1f93b8cde575c2 : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec3d83fd79dd4381844b0c77698a6b7c : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbcd17915fff47efa9f9ae4db67f1236 : SUCCESS in 1h 22m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/6ca9ab1751fe418080788af6cae53540 : SUCCESS in 1h 41m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1ff9e18256c414f899ccfc26a30f30e : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a3ac627fc021473ca0a2dd329da68fc7 : FAILURE in 2h 14m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2799118e79b4453b8c0de7ce11ded8f4 : FAILURE in 11m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ede3b1d6bb34263b8330590ccd8ed72 : FAILURE in 28m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/885ae1a631694e17b18eff7dbd58b513 : FAILURE in 46m 12s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0eae23110414401dbb971060714c52dd : SUCCESS in 1h 33m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0e7d65ea3094879a3b2d207460f2713 : SUCCESS in 43m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/80580cacf7d14beea6538ededd0cd2f4 : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/402279c6f4034250a58f9a164bc57883 : SUCCESS in 38m 44s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1c3763dcc8a3477cbd2a7f4f03bcacb6 : SUCCESS in 43m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bc6f8d24abda40c094c193013787ce8b : FAILURE in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0762f35d98b7d6bd8bd9e387268e393a5ff7eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:29:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1993b43844f11c63bf0c7ac53af49575f17f2de9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 14:31:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd7241eeaae149ca8d0eef73890c21a4 : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f2f2fce207406b8200e46dfb6aa304 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19c305d3d4ef464597230ab8ce08ee23 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3adc30837d11465c9c49501751e13f16 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cac700db34a44d0ea914746dc417d536 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ed01857fc94af89e2d676d290bf34f : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0c22e73a9fe44b3b4ba9bab991b468f : SUCCESS in 5m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/30485ce1210c410f95e4d938fd5d9e90 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e761db2d23d9413684d0ae353f900546 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f67682cea01c40d48ba2feb4cfec7408 : CANCELED in 25m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d722815fed394dda85a2107c2c0c4dd9 : FAILURE in 21m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98543586e9204656881e881d5e81c37b : CANCELED in 27m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/416efa54173d4529b3d67011479369bd : CANCELED in 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3d9b37db2216a8670500c248eb55182aebf0dac0","author":{"_account_id":28258,"name":"Austin Russell","email":"kenrusse@cisco.com","username":"austinr"},"date":"2020-11-04 15:07:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b20ad86476307c75e4d5dbf52b8e106138370698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 17:52:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c945e7c8706042f3869431795e89b91f : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77c3ccd8aed14f2aa8963038ebacbcdc : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47bd917b2da240d5adad68a865c1b695 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d00a9670f90f4b40a017ed86e0dbb46c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8e1447eb76458d9f0247a13a732905 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecbf20b34a53407090e65c73a8d30966 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13dd595851ec4019b18810c673ea7cbf : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d160ffda8ecc42e0941d66a8edf14862 : SUCCESS in 5m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/76290087d13549c3bd66a0f21feadb13 : SUCCESS in 5m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c55233e2ff0f4e1cbbb2182872bdc6b2 : SUCCESS in 7m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d4b3700def9b4411841497ac47e94437 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7a33b649ea994e11bc94707fe015fe31 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0417e43e4e4e439aa921231e0faffbb3 : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/13c7d9380a564ffc94ef90ae007fc02d : FAILURE in 21m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6e7ebecdcfa74fc191d4cc14afe86411 : TIMED_OUT in 2h 36m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5e65885a9fd64ff0966f0bfb58353b33 : FAILURE in 55m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b87b711972b45859d40b544374ef506 : FAILURE in 11m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90b02b6e0e91494f8269ad7ae94f5169 : FAILURE in 55m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/85e719e2c07641e8b87439b613797fa9 : FAILURE in 53m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36bf79c05cdb4805a645aba2a1d992c8 : FAILURE in 1h 19m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8022677d5f0245b594abda896f2292b7 : SUCCESS in 43m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d32231847eea48ffbba24ae9d6ce58d7 : SUCCESS in 1h 57m 38s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/196c2ebd6a43421a86ec5ad6cd9e8621 : SUCCESS in 41m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/be8fc4cc304e4e14a1a32fdbbc77a2a4 : FAILURE in 17m 01s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a6c36425daa64b999923a82d763c21a8 : FAILURE in 13m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86d3e7e66faaf85496e90d62eb1998b1ca68aefd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-01 19:43:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8b908780b2deccd5a53ed13a09a7698de6405ff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 22:17:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed256be28fae443ca723b3b19c412e7d : SUCCESS in 11m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/380a50a2eca043b7949a1792deecc058 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c5a08504244c459e0fd450215e8709 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/972cb38fcadb4d588badfe6a05bb79db : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59834642b84e4807a31e9994b62322e6 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14ebe10b111e4885865a3d77356d6ea9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ad497980974665af4f492f1605f774 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f08a8b48578946f7b493edba1bb2489d : SUCCESS in 4m 46s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1eda09c784184adc992609e04ee23524 : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/93820a3f082c453c9fdeb28b6a66945b : SUCCESS in 8m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/96fef509dce84dae8e6398215986444b : SUCCESS in 9m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb08ba8b3ac644bd8350c870d9c12dc4 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d1eb73d40cc4aeb9ca32dceb6eb0499 : SUCCESS in 1h 17m 54s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4744e730ed5742528dbf44abfa5689ec : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06272e617a48481782ac37e079afccb9 : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3b2640768dff45fbb1109e7195b84fec : SUCCESS in 2h 26m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5978ed6d6b324867914b1c182bf6f1fb : FAILURE in 10m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1b5158ad286d4ba89fefc9901676bd8f : FAILURE in 27m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b783d1869406414b8d2d254caa4e3300 : FAILURE in 56m 27s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/721ba49ef5dd461e90162ec3bc1782e0 : SUCCESS in 1h 32m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eefe43782bbc4305ae1062ac578fe411 : SUCCESS in 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/098f39ae87c1401fb496fb7573d15b7e : SUCCESS in 1h 44m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4060372eed38475cbe2c115e13c0f2ff : SUCCESS in 44m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d8ab6b923cbb41738e394ce2726c8ad9 : SUCCESS in 38m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/89c4d15bb58c4e39b583f913c6836d41 : FAILURE in 9m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"57ac40836f169adef1898e731f901a79aaaf2ad9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:05:56.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eef8587276657b56fffee5a3f6f039c554a9ca0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:06:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4491e752223b00d6f491c987e1ed5106d2e52ac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 02:20:00.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72c2be7c05014d80b0272dd8c62008fb : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e1640499634189bf9046913578586c : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5784f479baa54e03a2a9dd9b67787f96 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7ef24bfc644ec0a32703747e251a84 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1282db0adc93442f81e57eb86422500f : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16a2457693044fa98ca8b7dd63a82a59 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/982a3317ba9948e5b57dfa3714063263 : SUCCESS in 4m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e6bbb978d3a243a1bb833e8af5b4063e : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f7ccb2ca7a547ea97fb598a917f1f40 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64862d4578a440f8b780676133ada559 : SUCCESS in 2h 06m 13s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3e92b51757524a84bd1a74e06c1a4f31 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20918d635f50481bb23556abdea4e6c6 : SUCCESS in 2h 05m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61e2d25fd5bc4a57b6579343c209f3c8 : FAILURE in 2h 04m 26s","accounts_in_message":[],"_revision_number":3},{"id":"60605e740524976f6856fc7efab6e84a9a80367b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-02 13:58:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4a0481c652c9e89fc9146621eb18907d8da1486e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 16:26:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d41f533bfe49479a8a2945fdee4ba6d6 : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eec9adb35a6f41d1a44eab9e37e9a114 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe73908336c40c4917ce68141d47414 : SUCCESS in 7m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9765d7f11664b5aac68c6eeacf7a504 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68ad16c7303449a2a1cca73e46d47d0c : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f82a86ff3e429a968599e93ce194b3 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4632af231f415ebb9d2a0d21976f4b : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e64f64991795448b843741591454dab3 : SUCCESS in 9m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7d8e6ad60ff54e61b125e59c707a3712 : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/af6e713501a44f749de719d9c923bf63 : SUCCESS in 7m 08s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a2971f0d18ce440f96f3bd432a937d2b : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b5267e53e37428eb4f995c9494b8398 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6aa7437eacc7489298575a8ec7d42b02 : SUCCESS in 1h 25m 16s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1b2d3f6ce0c449069a6aee21c6fe8c58 : SUCCESS in 1h 34m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2658107903c94d2797e4a480aa714b97 : FAILURE in 2h 05m 08s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7f188126bf1b4faf94b87cdf1a2ca879 : SUCCESS in 2h 07m 21s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/116a42d37a8f4cf38fe8ce4ab8b766bf : FAILURE in 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ff9a1e39fe84c7498aa9ce56245d633 : FAILURE in 56m 16s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3154aac142e4d74b424acedf08ecdc1 : FAILURE in 50m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1c98e781d494bc0b002021858a763c4 : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5bf8a14519b40c3b16f51daba448bb1 : SUCCESS in 44m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/58ac843182db46ad94d1b8d6ebe393a8 : SUCCESS in 1h 47m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0746bf009224decb950f4fe5cbabba6 : SUCCESS in 39m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4ee971fdde9e4071a1169a8681894b60 : SUCCESS in 52m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5c837a0d98864ed487dddc83ab5a0454 : FAILURE in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b6d3f0badbbf56fb975daed720dc4f915bc30767","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 07:55:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf2682b8ab9c5214837368ac236a97b3ac908edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 11:30:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52728c82a31b43448542d97af05df264 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9087618bb62407897942f81dbc79888 : SUCCESS in 10m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2821e7f57eb345a9bb21613eb99d937f : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d2cbd0e9cf4946c097c9210639bcc4e0 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b30e75b7588a403583f9b5b731da53f9 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa70bfe9e9764ee38c347ca718657750 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd868d4fea4c4876a2860778c5ef4ee5 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2ae12af30ae4c05a5650370bda1c3dc : SUCCESS in 4m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5b2b48c459f46429d6de9806b16e549 : SUCCESS in 5m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0fde954bd7824f299e60df542c1a834b : SUCCESS in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0cfe5b90d0c74943843b03ffd2d8e53b : SUCCESS in 8m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fa6fee99f9b41aaa27e991f323a8b7e : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c8b59c32716e409cbcbfc9f0f44d3a8c : SUCCESS in 1h 33m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1d1184f01224430ca40a598c2b05677b : SUCCESS in 1h 31m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b6028e9095294bbf99c1dcccb33c98e4 : SUCCESS in 1h 55m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/3468530fec0b41ebbb2131efc09ee5fa : SUCCESS in 2h 29m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/06f54c08552d42389c3ef811f1677ad1 : FAILURE in 9m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02045626508f41dc9f3784dcd8c271f2 : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f1e926b657db43a38512e4b7a8fd4408 : FAILURE in 1h 06m 39s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/30efd90362cf43778454fd9945980b37 : SUCCESS in 1h 26m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b90ed142c00746fd9d511f1bc3d848e1 : SUCCESS in 38m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/575fcf12a34743b2ba530e9b30eebc3b : FAILURE in 1h 47m 26s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eaea19e09765491a99768b927e645413 : SUCCESS in 38m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bd5955421df344c99696bfb133dbee6a : SUCCESS in 1h 00m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/14e57ff526fe430fb130eb8ab609de69 : FAILURE in 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de1566d0aa6bec41bc3db0219c73b196a3588b80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 11:30:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"49ede6777d73babb6cfb7e4cebce91d9e9484fe7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 13:36:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dcaf4c2c2044cf9a1b5fad8a61915cc : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eff5f748ae724741a7933e7776061bcf : SUCCESS in 6m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2543c1c1335f46739c6989910b402eec : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70bbf26ad80e497e8ae8cae0b69d6100 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/203a11988722422e86bd258f3e0ea5b3 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10fcc4a9eabd4da1a1ff8da63c546ba5 : SUCCESS in 14m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1241f4ebdef848219a55d870e9884d9b : SUCCESS in 4m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e0944c36221c4f1193fa2e0b6341df47 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/064844e7df4e4acf874d2b7b51bef8bd : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a52f79e6bd2e4c0c88bb6b5424a360cf : SUCCESS in 1h 15m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a8eca02060aa4d338c0ccd4f7283bc80 : SUCCESS in 1h 30m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c55c3419feb44e5b95853a3506a68339 : SUCCESS in 1h 48m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f775c1218c484b58af40cd23d29688af : SUCCESS in 1h 35m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5c474165baaf302e181becbdc2c0d976070792b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 13:36:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b9b3e7a00e03cb0b4ea097ff7b72c10c8523a377","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 13:39:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7c74eac68949c68f8ef60072e4eba3 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/604825a48ef84e0aa3e4696cb830ee84 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9c03dcf7275ab5d3962c96a7fc00f79f112238dc","revisions":{"084be59e49961eb689832d19923a673736ffb2be":{"kind":"REWORK","_number":1,"created":"2020-09-22 15:33:12.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/83/753383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/753383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/753383/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-22 15:33:12.000000000","tz":0},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n(cherry picked from commit 9453701fb46f0ad23dae00b3a3792bb379978d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/084be59e49961eb689832d19923a673736ffb2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/084be59e49961eb689832d19923a673736ffb2be"}]},"branch":"refs/heads/stable/train"},"52abbdde9661c2084c41f19399e0dcabe9bbe284":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-23 19:54:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/83/753383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/753383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/753383/2"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-23 19:54:22.000000000","tz":0},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nDepends-On: https://review.opendev.org/753838\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n(cherry picked from commit 9453701fb46f0ad23dae00b3a3792bb379978d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52abbdde9661c2084c41f19399e0dcabe9bbe284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52abbdde9661c2084c41f19399e0dcabe9bbe284"}]},"branch":"refs/heads/stable/train"},"9c03dcf7275ab5d3962c96a7fc00f79f112238dc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-01 19:43:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/83/753383/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/753383/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/753383/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/753383/3"}}},"commit":{"parents":[{"commit":"811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe","subject":"Enable octavia-tempest-plugin for grenade job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-12-01 19:43:57.000000000","tz":0},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nDepends-On: https://review.opendev.org/753838\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n(cherry picked from commit 9453701fb46f0ad23dae00b3a3792bb379978d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c03dcf7275ab5d3962c96a7fc00f79f112238dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c03dcf7275ab5d3962c96a7fc00f79f112238dc"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
