)]}'
{"id":"openstack%2Foctavia~753384","triplet_id":"openstack%2Foctavia~stable%2Fstein~I88e64fd5e8a4c709f735be85b85139dbb52e4acd","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I88e64fd5e8a4c709f735be85b85139dbb52e4acd","subject":"Fixes API list handling of unscoped tokens","status":"MERGED","created":"2020-09-22 15:33:19.000000000","updated":"2020-09-24 06:08:08.000000000","submitted":"2020-09-24 06:06:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753384-1600927567207-7e3b2230","meta_rev_id":"142593ef9ade1af8b4b68945bcac4a73125cb35e","_number":753384,"virtual_id_number":753384,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-24 06:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-09-24 04:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2020-09-22 15:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":2,"date":"2020-09-22 17:54:09.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-09-24 04:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 15:59:24.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-09-22 17:54:09.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-24 04:07:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-09-24 06:06:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4836f36d6910d189e46b72a64aa09798fb080779","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-22 15:33:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"438f3b371fefe604632480a7e4c239ca7c19da62","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-22 15:59:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0e4f391941c16b6e2a407b3a1a94d9acfffef4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 17:40:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/277bea557a4f4a57b729f7947700b239 : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21ee7652a22a4498957b182be68bfe5a : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f14fe7f6b3474938af1bd484b6e7df43 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/053fa6a29a96475e85be3067a365a76d : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4083ee6add944014bb660466207b08e3 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70801da8978e43479e719e92b4972054 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4c7859a22348acaa3e8b061366def6 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bc671d5e134c42af8ab7602f23873a : SUCCESS in 5m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b79e6ae5bf8b4ddf876aa310a6076324 : SUCCESS in 1h 06m 37s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8b592d6d12ff490db365e27c40382723 : SUCCESS in 1h 12m 23s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0327ec88db847969cc4b42dd4ea7365 : SUCCESS in 1h 07m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/24d6ec7ee8ca4177b65fe588baccbe76 : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d42536b2da334452a4e7af6b2b09d8d2 : SUCCESS in 7m 58s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/eab8de4ffef64a39878ccef559ad8c6c : SUCCESS in 9m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e532d0530bc04043a4601ae4dc1125c4 : SUCCESS in 7m 08s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5a66734e7e946d19bfa30137626bd62 : SUCCESS in 1h 02m 52s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f95941eb72dd474da2c6d684ae88929a : SUCCESS in 58m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b489f8de4b0b4acda555ca2c44c42245 : SUCCESS in 1h 18m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b34e98f4e55b45e6866068922c11e1e1 : SUCCESS in 1h 15m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae85e87baeda4b03b372d897d2f57679 : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/441cf79c268f4978a4bfdb5b73d003c0 : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9712d40da629422393c66ab728510e4f : FAILURE in 11m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e6a8edc538c49148d2c975af10c2c0d : FAILURE in 46m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4de82ed735bc42909de625230431fe6f : FAILURE in 45m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/061e884f059d469ea74f26528dcd9d08 : SUCCESS in 1h 17m 26s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd254e0de84d4c899d663efb266bb923 : SUCCESS in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"713f9cdcbcccf2699cac22fae1142500f21ac774","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 17:54:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71d556e1fcb9ff5fd14ccfe44ae3305de1c5573c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-24 04:07:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab9213c031283f4efd74e3535641533db7e95ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 04:08:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36e56d6aaea5ffb64d731157cf4bafc6c9556b0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:06:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee3e3c1ff5024e6a8c288c22b98b0393 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8be743774d8411ab9edf90bb7a56b5d : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49948c6bafe64f54aacce58b84fbede5 : SUCCESS in 5m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/19814f31b67e43a6b051b3b8ee2c8b56 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd6cdaae254a419e9de764f193dae7d3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b19d68ead845f8abe17c1c923ada78 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/308b45fc82644d3eab3b826b73a85733 : SUCCESS in 5m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d2f53698ce294077bcc8ae17d8af1bac : SUCCESS in 7m 50s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/98c16d6eac78429eb95c3a791fee3ad8 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d44c21ed3874ed69270e41609951fe7 : SUCCESS in 7m 43s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a178cb743b7849c7a65c3f28db10a9db : SUCCESS in 57m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/bdd92685b21349c0b9deb981c5383638 : SUCCESS in 50m 35s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/8356c7c05e3944e68c0aedf0eddcf518 : SUCCESS in 1h 16m 32s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/bf666cf5f46f4cccbde8c25c8fa33ce9 : SUCCESS in 1h 13m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f3229a21cab4ea1b7ac39ae843607f3 : SUCCESS in 1h 29m 57s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b91a047b22c24d8e8c07550790743662 : SUCCESS in 1h 20m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9aebdbb600b94ee48a3057e12d6f8306 : SUCCESS in 1h 28m 03s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/674964b64492488f9bf767cfa37ec37a : SUCCESS in 1h 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a049aeb418a1fb6279b0d58b49b0a476367bc004","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:06:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"142593ef9ade1af8b4b68945bcac4a73125cb35e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 06:08:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0ecc6176944e4b9e89f2773843daef : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baf515fda95b4bbd9411a915a6b70854 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a6011f7784303f7b1d2f55ca801f2e27d146419","revisions":{"0a6011f7784303f7b1d2f55ca801f2e27d146419":{"kind":"REWORK","_number":1,"created":"2020-09-22 15:33:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/84/753384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/753384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/753384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/753384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/753384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/753384/1"}}},"commit":{"parents":[{"commit":"76e754f34002e4e49ab265630401572d9bd730ab","subject":"Merge \"Fix accepting \u0027insert_headers\u0027 when unsupported\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/76e754f34002e4e49ab265630401572d9bd730ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-21 22:18:11.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-09-22 15:33:19.000000000","tz":0},"subject":"Fixes API list handling of unscoped tokens","message":"Fixes API list handling of unscoped tokens\n\nThe API list methods were not handling unscoped tokens correctly.\nIf the API is using the admin_or_owner-policy.yaml policy override file,\nand a user used an unscoped token, the API will list objects for all\nprojects. This patch corrects that issue.\nIf you are using the default policies, the API handles unscoped tokens\ncorrectly.\n\nChange-Id: I88e64fd5e8a4c709f735be85b85139dbb52e4acd\n(cherry picked from commit 9453701fb46f0ad23dae00b3a3792bb379978d0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6011f7784303f7b1d2f55ca801f2e27d146419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6011f7784303f7b1d2f55ca801f2e27d146419"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
