)]}'
{"id":"openstack%2Foctavia~753969","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I293587fa6d1f7a06aae9d1cca5b81342478c5df2","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I293587fa6d1f7a06aae9d1cca5b81342478c5df2","subject":"[Amphorav2] Healthmonitor operation minor fixes","status":"MERGED","created":"2020-09-24 08:30:29.000000000","updated":"2020-12-02 06:24:16.000000000","submitted":"2020-12-02 06:22:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753969","meta_rev_id":"e17d656d593deb7a1efc43d9b8e5fbed11104848","_number":753969,"virtual_id_number":753969,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 06:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-10-21 05:26:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-12-02 00:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-12-02 00:20:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 11:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-21 05:26:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-12-02 00:20:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"85694898b71300f73e141048ca6204fc97449f7e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-24 08:30:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd5387d8bbd73a8fa459d22c45eb1f218fe01b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 11:16:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2305ed0b5ae04539b1427205cbc76117 : SUCCESS in 8m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9ffb1630094f3e8a20678e2ab06e28 : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9da3447f23db42edad4bf633191dda9b : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e754a1299c45afa579c7474e173af9 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b84ff7590494a989ff804c723b51d88 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a6c497bc1ab40e69c89566f4539158b : SUCCESS in 11m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f28ad938056d4523b833c66fe8ab4290 : SUCCESS in 7m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f72290edbe0443d3aa7c0afbd980aef5 : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f040e6aa0f38424ca8e653b50dda72a0 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7851161b991d4888bccabe6949bf5c89 : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c07cd4528e649b6ac38b935ad03c45d : SUCCESS in 1h 55m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/984c95cd56d8431fbf647819ea7a3093 : SUCCESS in 33m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0106e283c69f4d36a993ee9053f2706e : SUCCESS in 33m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce8809e978da4d7da270eb5d56d2c73c : SUCCESS in 43m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3998a472314a4c75ae70b557f063d129 : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/266d7b83b1334471b784789d04a45546 : SUCCESS in 2h 18m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/039538411eb0437a931f4c43601f0fa6 : FAILURE in 49m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dab4cce0bb234a048a9a3434117d68ce : SUCCESS in 2h 00m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24d4d6c54b74163b562a48378d17acd05aaa02b3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 11:39:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8039baf4f01ad007aaad012f138f3c0a3b4a5eb0","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-08 08:33:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e1e3ba124aa9226126a29a7cf4f6ce590a0e63fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 11:17:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8f4f36a6ab04fc6bee8a12d2ac63af1 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d4993522d5f481b85ee4e3070bebed0 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0091771a679b4fcc8f890688e097d1d1 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca543cade33642f895575cc532a17ce5 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f41f847ae7e24e3ab3449c504724d947 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1159c58667694089b40179453a1fb935 : SUCCESS in 12m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cba7428be14a423ea9e0c73a349b34a5 : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/40a15ea48cde4e1db70b6aff1fb315ea : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e1d1f18febe418daeec44f4f91f414d : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/224716b3d8a44ab9a1493c768581370f : SUCCESS in 1h 18m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/472248a7e0db4f42a745e8c6a35800d3 : SUCCESS in 2h 24m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c33819e5d7cf4cb9ad769772dcae74d6 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/29df8a287db149f39c3ac54bc6b4fdf6 : SUCCESS in 41m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0969ce34dbbb49d68d2d9d503482f647 : SUCCESS in 44m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dea2dc20080b47d7a0c28609c37a3995 : SUCCESS in 1h 20m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9866e89e9fc744e68b1eb6fd0a36b21b : SUCCESS in 2h 09m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b46b44dcc1824b7e878eb534eb5afac7 : FAILURE in 49m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bfdf74585346460393c8ed6b54006bb2 : FAILURE in 1h 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7eee5f912b15e1b6b269864bb182a11fec28201b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-21 05:26:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c9cbb269d04e8eb86a20610c6a32120f18999014","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-02 00:20:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"c38fdef997038453af5cd34fea241564dc7e7269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 00:21:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65699e06448061692e3eb03115971230fe3cd02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:22:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acc32a953ff94001970cc6b604db57d6 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c58035222f74d5398ef129999f5e7b3 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f866d94d2c1c4c5bbe16c6d9c15b04a5 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61999a0440ab4163861ba725e36de8cf : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e72f53d8ac41338c27b891c7e9b0f4 : SUCCESS in 11m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0be34576f02f4f13b3ad4f2247e9ad7b : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fa2f7c9e406481197148696ba991f01 : SUCCESS in 1h 29m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53b39998ac6a432e8228c87bd889bd35 : SUCCESS in 1h 42m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee694141fb1547fbb54ee5f9c1cdb7d6 : SUCCESS in 42m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2688299e89a641dbad34cc4c506ed203 : SUCCESS in 44m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9502da5587ed4fc49d740077ebc4ca8e : SUCCESS in 38m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69ddf0c1fb6e4599b0906887f84e87b8 : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2fb039af2eb19135bd81225e23c75972394ff580","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:22:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e17d656d593deb7a1efc43d9b8e5fbed11104848","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 06:24:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb195a4dd13249ba9964de2a3918f327 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9952c06086457f766033467668cbf4b12a96287","revisions":{"16526601732a5b4d4f0581018f24e155dfff7f2a":{"kind":"REWORK","_number":1,"created":"2020-09-24 08:30:29.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/69/753969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/753969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/753969/1"}}},"commit":{"parents":[{"commit":"c7fce7a8a51c8ab42682ce9f37de05a5b7168468","subject":"Merge \"Set Grub timeout to 0 for fast boot times\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7fce7a8a51c8ab42682ce9f37de05a5b7168468"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-30 13:05:15.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-09-24 08:30:29.000000000","tz":0},"subject":"[Amphorav2] Healthmonitor operation minor fixes","message":"[Amphorav2] Healthmonitor operation minor fixes\n\n* Fix logging messages for healthmonitor operations\n* Add retry when healthmonitor is not in db yet\n* Fix setting ONLINE operating_status\n\nThis fix api test_healthmonitor_update test.\n\nChange-Id: I293587fa6d1f7a06aae9d1cca5b81342478c5df2\n(cherry picked from commit db180bf81869762f71423c4d99c50587d12f59bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16526601732a5b4d4f0581018f24e155dfff7f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16526601732a5b4d4f0581018f24e155dfff7f2a"}]},"branch":"refs/heads/stable/ussuri"},"f9952c06086457f766033467668cbf4b12a96287":{"kind":"REWORK","_number":2,"created":"2020-10-08 08:33:27.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/69/753969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/753969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/753969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/753969/2"}}},"commit":{"parents":[{"commit":"c7fce7a8a51c8ab42682ce9f37de05a5b7168468","subject":"Merge \"Set Grub timeout to 0 for fast boot times\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7fce7a8a51c8ab42682ce9f37de05a5b7168468"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-06-30 13:05:15.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-08 08:33:20.000000000","tz":0},"subject":"[Amphorav2] Healthmonitor operation minor fixes","message":"[Amphorav2] Healthmonitor operation minor fixes\n\n* Fix logging messages for healthmonitor operations\n* Add retry when healthmonitor is not in db yet\n* Fix setting ONLINE operating_status\n\nThis fix api test_healthmonitor_update test.\n\nChange-Id: I293587fa6d1f7a06aae9d1cca5b81342478c5df2\n(cherry picked from commit db180bf81869762f71423c4d99c50587d12f59bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9952c06086457f766033467668cbf4b12a96287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f9952c06086457f766033467668cbf4b12a96287"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
