)]}'
{"id":"openstack%2Foctavia~754056","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ie2c748ff69c8fd736e9c67d686068a33d803222e","project":"openstack/octavia","branch":"stable/ussuri","topic":"fix-house-keeping-stable/ussuri","hashtags":[],"change_id":"Ie2c748ff69c8fd736e9c67d686068a33d803222e","subject":"Ignore DELETED amphorae when performing certificate rotation","status":"MERGED","created":"2020-09-24 10:03:10.000000000","updated":"2020-09-25 16:59:26.000000000","submitted":"2020-09-25 16:55:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754056-1601052935571-b2ad87a5","meta_rev_id":"07e9aa38e2b69922d133d3326594dc7b2a6b1535","_number":754056,"virtual_id_number":754056,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 16:55:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-25 09:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-09-24 10:32:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-09-25 09:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-09-25 01:31:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 10:32:11.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-09-25 01:31:20.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-09-25 09:26:44.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-25 16:55:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fecfc66c6935fbd7cae7ac59f73d6af7330a845","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 10:03:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2a45f585b95b67c7147163640a0f53f6b24c57b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 10:32:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"816187f5447f3e449d0a7aaa7b67210f98f7b048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:56:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3c7c221ca78447bab4cce39a90a11f0 : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e496f74dc042049bff4ea9e87c7cee : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73216c8bc86a477ca205b18463fb5cb0 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1dfd6f81a084a8a9a6fefa6aea8a186 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c6b90eb084f40a9b1f4d804c0c310ed : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b092a7add648d695128805f9e40632 : SUCCESS in 10m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2ccd9f255bb245fcbda2718f1abafd75 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3073d1d3d77047539373f1c552a5cdf0 : SUCCESS in 9m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4bed4b4abdc843a39180ea5ce4eb98b7 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c84005319e5c475d8698c41b11add2c0 : SUCCESS in 1h 46m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/663c6d75fa294479ace8f3bbd9466313 : SUCCESS in 1h 58m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b50e76d4daa94a93a22415f7f429a192 : SUCCESS in 42m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8fa5885d4c13443c9864ce80c5e14d16 : SUCCESS in 39m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04aa4006d7814cb48a0cac16f8d81d42 : SUCCESS in 39m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f216a01a889438eb3a979b00d6825fa : SUCCESS in 1h 23m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b11ef3a83fd4387adc135b7092af8f8 : SUCCESS in 2h 10m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ad8443d369634582b67a63f596f22c0b : FAILURE in 1h 04m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/45c01b90d5c84136a09c8cf9e831a99f : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"764bb3f137808b4d60363ca26eb180789e99dae0","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-25 01:31:20.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nWe are using this on ussuri and works well","accounts_in_message":[],"_revision_number":1},{"id":"876e87dcf1b66b8b3ff82db39a69bbbc0e9a7d91","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:26:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e0f3aafebef99910d7d21612dd54f993940cef9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 09:26:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d3291f80feda22af5b9f3b7f0f0b4e51c3fe3e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:09:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5902008c71524caba90bdb156fa57389 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e309c2e0ba408eb7915a5a1b7b87a3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/358b3c8e50bd4b4c8ba02fa1d1225d3b : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ae4fcc8a28549feae2dce21b70c8ed3 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8571b73d96405f9b48106d3b69a48a : SUCCESS in 13m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac0885eaa8c042509b76320047ca2878 : SUCCESS in 11m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/797bea119e784165b356691cd2d9b777 : SUCCESS in 1h 43m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a121514e04354d06831668fef58a8257 : TIMED_OUT in 2h 34m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/955e80675844460c920fba0cc58ce4e8 : SUCCESS in 39m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/72d8d5ac18c4409792991e9acb7ec8d9 : SUCCESS in 47m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd9ab4dab3b24b42add538922564b5db : SUCCESS in 38m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6117a72fb8874f63b8384871329654b2 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"07fe31451a9d769e788bb05a4998dde79a1e281b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-25 12:14:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78cedcc7c8d9fa7bf6952d384f406b10affce484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:00:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d80717d2c0948d1a1529a687bb45e89 : SUCCESS in 9m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74dde5d7a0e7498094a7c8d1e6694d5d : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85ff6c58dcb459c9b8d7ee1bb976106 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb68899ebb8a42ef82273bdb86c23cdb : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5bff5d9a86b44f5ad673a211b584c1d : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a4a321950a4a99bc06413c7e353c76 : SUCCESS in 12m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/635ed027e9054900893a347a524163c3 : SUCCESS in 5m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61d9e44fd4414d218a50ec205244e6e3 : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1dc26d0c96404997a9fd0f9565a66fbe : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77a809f06bb44f7fa0ea217be10a1360 : SUCCESS in 1h 31m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83264f2c10604018b0a65ea7b2034cf3 : SUCCESS in 1h 55m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c18731d5b534f458855cb7dad1d2c93 : SUCCESS in 40m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2f21145601bc4993a7fe9b4fa51287b1 : SUCCESS in 37m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42c130be13d94f5298afdac98da9b1e6 : SUCCESS in 38m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/726785397a5b4307930728bc51f0e508 : SUCCESS in 1h 24m 03s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f90f51d6ec8c43bfb10766a0499fd317 : SUCCESS in 2h 08m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47572304b5d546f9857c9508cd1f513a : FAILURE in 44m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0a434c7097bd421082f40003ec4722c4 : TIMED_OUT in 2h 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"add113abe813dcea031dd940dff1d5b740cbe423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:04:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20065318790cddfde399bf633cbee020dbfed6a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:55:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30af766f14b9458f89c78c6b6bef0a9a : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758dc324110c40a59657598e5c8c7b69 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/185bda54b6214e73b895e650e60e5a6d : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b479058c815147b7a3e2421c7d05eb6c : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb472bb31894f0481dd92bc45133ad6 : SUCCESS in 11m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9681387aea714f03abbd1529f30d8441 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a79b5914ee64d67931dbe4139343770 : SUCCESS in 1h 45m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/541272236ab4418599c4d6da4cd984b0 : SUCCESS in 1h 46m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c3dcf6729a764d5e8ee42051d262108d : SUCCESS in 44m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0301d658a1a14d1580b73eab78c3176d : SUCCESS in 37m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f73f826eb4b2431a81875e931ef02cdc : SUCCESS in 39m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6e45b00bfdc04bcbb354c2698f7546fa : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7f0ac57ae7a6bc04b8255e2e8ddbe2583ed0d662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:55:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"845adec817abdd9eb0b78eb6cee1fbb26ac5e440","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 16:59:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33aa78454e4a4856bc79502c4d2d10f7 : SUCCESS in 2m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"879b08d320dc30bb1348e6c2e1e3e917685ee8f8","revisions":{"879b08d320dc30bb1348e6c2e1e3e917685ee8f8":{"kind":"REWORK","_number":1,"created":"2020-09-24 10:03:10.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/56/754056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/754056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/754056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/754056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/754056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/754056/1"}}},"commit":{"parents":[{"commit":"c7fce7a8a51c8ab42682ce9f37de05a5b7168468","subject":"Merge \"Set Grub timeout to 0 for fast boot times\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7fce7a8a51c8ab42682ce9f37de05a5b7168468"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-24 10:03:10.000000000","tz":0},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/879b08d320dc30bb1348e6c2e1e3e917685ee8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/879b08d320dc30bb1348e6c2e1e3e917685ee8f8"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
