)]}'
{"id":"openstack%2Foctavia~754057","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ie2c748ff69c8fd736e9c67d686068a33d803222e","project":"openstack/octavia","branch":"stable/train","topic":"fix-house-keeping-stable/train","hashtags":[],"change_id":"Ie2c748ff69c8fd736e9c67d686068a33d803222e","subject":"Ignore DELETED amphorae when performing certificate rotation","status":"MERGED","created":"2020-09-24 10:03:49.000000000","updated":"2020-12-04 00:06:20.000000000","submitted":"2020-12-04 00:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754057-fix-house-keeping-stable/train","meta_rev_id":"6ecac9596d2363f9b73291be58de223c0fe46817","_number":754057,"virtual_id_number":754057,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 00:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-12-03 14:04:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2020-12-03 14:04:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-12-04 00:06:20.000000000","post_submit":true,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-12-04 00:06:20.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 01:31:25.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-10-12 15:55:00.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-10-22 08:33:41.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-10-22 13:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 10:48:02.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"e18dd6ab138bbb2de8ce64c28b4884162d0dfed3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 10:03:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f756b4aa6c7032c344b4747c2db7ded7b7ffa367","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 10:32:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa1c2beb41e352780dc5f0ba25115411cb0f4984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:27:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eea45fd4d2045ed8713cb8db7c6812d : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68adbc2269e9421b968a422f7a47a948 : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ba6d0dda13449fb92abe27bf7c308a : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6a86f0db23446e6bfa4934e6d43c76f : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88e506dffd4d4e8aa061a05d05ac10b0 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fd32109dd554fdc8897b677195dc0ff : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1f12871aaf41db967915a78d914651 : SUCCESS in 10m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/61076369e4d64661b01c72346d943c97 : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6f218494df5449a08c659819fd820570 : SUCCESS in 8m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d371947d5cb8465e84b3771f93b6b817 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ac17d8879904a73b82425f3b36f752a : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80c3402cb65b4deb9ba16b42a1e62f98 : SUCCESS in 1h 42m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c98641ffed30428a8aadc3f7c2a6974a : SUCCESS in 1h 49m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b898a56cf5e6405a8b32072d164e5746 : SUCCESS in 2h 25m 15s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/730462b320be4ffa8f984132eb59660d : FAILURE in 1h 29m 21s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8484ddbf388946758e876253ff2ad7ce : FAILURE in 10m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9cc2db46b59e4ad087aea4a439f8a085 : FAILURE in 26m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5fda0a5cf11040d19b02dcb935ebe1d0 : FAILURE in 41m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/44aa782cdd2140189d00fc6ddadf304e : SUCCESS in 1h 33m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3370b39bf974407fb9895b1b76a23c28 : SUCCESS in 44m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c60861b9877d4083bbd00d6d308b9379 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f98e9a9ede249479f51fd0d20ea3edd : SUCCESS in 34m 22s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/27e3248fad064cd6acc25bb06590a31a : SUCCESS in 43m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2a579bae90b54b1285cf399e496bc409 : FAILURE in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"362b3e5e521d4c2c4344d93560576ccedff17405","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 19:40:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4f66d63e9ed3e4a6566fb09f2af2023683654162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 22:02:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a797019b7111405cb29c6c7ba60b50bf : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b9f2a3beca4a56aead174acada6b01 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17eeeab7c7c24763967ef3b702d97b45 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/877294a7b2c241df8b4689504babc2f3 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042170dad921492d952f45d576e82503 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8790453c674c48efbea2171ac14e46a6 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96539115f90d4774841d1c9f9fdf9d67 : SUCCESS in 12m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f599686e5084a98a9f7fce42ea40240 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a98cb55d459c491bb9e3052983d6df4f : SUCCESS in 7m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6ca7f03cdee40b692a340982c911a20 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1620ec2d6e94479e95b9ac2a678a4005 : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/23d8f61724d245a3971c724726aa55a1 : SUCCESS in 1h 42m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d4aada31bde04903b1965926958b8b61 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6859cef76e184b3a813a2a07f741c23b : SUCCESS in 2h 13m 20s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b1b20ba3bd6542d9ad6b0e908f9360ac : FAILURE in 1h 41m 32s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5ab49510cb1349e99ea40ec21b52c34b : FAILURE in 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c21e854ffbe74d7983c514c20d402f20 : FAILURE in 30m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/cdd36c84aeff4ccaa265dbb511fc6d40 : FAILURE in 43m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f9f27718585412b9d0e36e72e7bc4b3 : SUCCESS in 1h 40m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aefaa65b61924d74a52ea343b3bcc338 : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1c3965822de24b33a8f6f725bc24a3c1 : SUCCESS in 2h 09m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43071109a27445a8ab7c11f3739ea0c1 : SUCCESS in 40m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/97018bc844fb41f0ab7567b3b1b8c902 : SUCCESS in 40m 30s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8fa076e4edbb4b2aa151847dcb7dccba : FAILURE in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c612f1566d4041e11e3edb2be217e7958ada6cd5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 22:56:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9e4620e824df9afb3e11b9d28bd51da32d69a88","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-25 01:31:25.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"52c213cf968dfd416590c1bb1bc72a0b98cd69dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 01:49:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c92cb53eee94ea885bbda3bc3e84428 : SUCCESS in 10m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1049099054a94a07860ac57e55be443c : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18c45f00e1144ae884c768da4d5af47 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79b78def046a468ea4b71f0bd25e5084 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82540edd1ddc440e86986a5b2c687c31 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22b451d8dba549c3903c0ce90a118746 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0739f607c3fe499ba876ea133a184041 : SUCCESS in 10m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/715a7e8eaf2c4599bdb0f9d7d2a43130 : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/100dee11029841708de47a26c6064cdb : SUCCESS in 7m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d04d872125794383a0e0888c9f138edd : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/073f8d8083fb48b388d3cd6056955e8d : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f48310fa42164f17b05a9bf8583102a7 : SUCCESS in 1h 49m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f22f55a5cd574475955df44de7142954 : SUCCESS in 1h 58m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dc29151af0d4e0da24e2854abdf95d6 : SUCCESS in 1h 50m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/234570516ea9452185ecd73da0cfa869 : FAILURE in 2h 48m 15s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5eeb981102c1436684e4cc3ee93d1e38 : FAILURE in 11m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd69d6f646ad4a11a36ed98decefaf44 : FAILURE in 28m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/69e527a602f14bccbb20ee54f63192dd : FAILURE in 48m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7079a4e8db644d7a4cfd096419b8868 : SUCCESS in 1h 28m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31389d8d5d6247f2bd43934b18eda723 : SUCCESS in 39m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9afae7fd6caf416d8aa3248ae519787b : SUCCESS in 2h 05m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2c8cce2e1194a6783002ddee7c2d5cc : SUCCESS in 40m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4a66c5f1876f42eaa4dfca592e9f368a : SUCCESS in 43m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/ae34c824e1b841368fa376459da86e7e : FAILURE in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51b5d0a6f13b0347f614129f765947e157906911","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 02:17:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"622ac145834f1dc119c35dbc0a556d465a3f1f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 04:57:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f9310adf194ddc9095f7ea2424e0a9 : SUCCESS in 15m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0678b835fe904b238722961a02e0f67f : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8fd68dcd426469ba87176e989fdb2ef : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54415231598d44afb9caeb25542db0d9 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67ed1c3e94eb44fb918b634d3f55c5d8 : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a802e248ba624db1ba8a9bc7f72ed8d3 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3f5af1a1f804b5ebca268a5d7e07aca : SUCCESS in 12m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/56229eb5eb3a484a839ce9f2bf689210 : SUCCESS in 6m 32s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fcb9b10fa0b74462b46be70544bdeceb : SUCCESS in 12m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2555fdf33651468d831b1ad05c75714c : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cad2aa08f50d4ea4863f3e4cc522ba7d : SUCCESS in 9m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efa0ff4cbf0049008e908d981e4c8427 : SUCCESS in 1h 38m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/174a183ed8d942bca0be0cea72c9bfe7 : SUCCESS in 1h 53m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5243774d9b494b508f5862564e8ddd49 : SUCCESS in 1h 56m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/581a625396924f8ba29cef4c40bb6ff3 : FAILURE in 1h 50m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3b821d0e6636469f8f153f6b8bddf11c : FAILURE in 13m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c99f2d48caaa4d2796f30323f8520f2b : FAILURE in 28m 46s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/9717882766924a95b57069030044c2fb : FAILURE in 46m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f2c687f61804c54ad5241e4283e1215 : SUCCESS in 1h 41m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5fb7eedeb7ea4787aa0ef48c47387d9a : SUCCESS in 49m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a8f411d7d19144a78c3f96c6b789bd3b : SUCCESS in 2h 29m 35s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc4b3e9f43134fafa2cf12d6b145e5f0 : SUCCESS in 48m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/eccc32b5872b411da5ee2ba63ab8f436 : SUCCESS in 41m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a757c2384fc24224924f39af6820c897 : FAILURE in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4496fe89b10a82cf56280ef20fad0c945cf0da66","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 05:49:42.000000000","message":"Patch Set 1:\n\nERROR:\n\nMaxRetriesExceeded: Exceeded maximum number of retries. Exhausted all hosts available for retrying build failures for instance 30a997da-c4a4-4115-b839-44eb8b543e9f","accounts_in_message":[],"_revision_number":1},{"id":"2537fae53cd5092e0b6c44a6f68c6c8139c540e0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 05:49:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5dddb6b1174f0c236ec446163806804c27fdb635","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 08:10:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc56b9f1f33f417fb7eea21b27acde5d : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45014c7385db47a7a4bb25cf9202efef : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/610d03cd42bb477bb14f0aa12d5d21c1 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4faf15dfb19d446993546c9d920f24a2 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64e02101cac44b9cb3665a02939a77ad : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e42e8e480c0a44b7bcd805fca53cfc96 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f89402b50e048dc81e9f19b509928aa : SUCCESS in 11m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/01c4b64c62e14938878af3907f624c87 : SUCCESS in 7m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ccd55251440644b8abadb30fb3cb9dd1 : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e1895d7fe3664882ac062b000656b539 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9cce6b46f5f047a99e1d4db15bff8f3c : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efab7a37c7b04e2494a914eaa3b7050b : SUCCESS in 1h 46m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e4f219a64df7405ca85b1c0936658f2d : TIMED_OUT in 2h 13m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc51117ded1b4076a3decea88a3075f8 : SUCCESS in 1h 50m 26s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/77e8c60a25ac40058c1f3670634dad11 : FAILURE in 1h 34m 55s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0618fab30d8c45bcb4a2dc6fa71a4340 : FAILURE in 10m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/848cde2f71154a91ad0e02a5251ddfd8 : FAILURE in 31m 24s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d339478ef9f4ca49a63da1299d73e76 : FAILURE in 52m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c427aaf4ebb48748773b9d2e5de6840 : SUCCESS in 1h 29m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6243da78ac5f45e9b0edaa379ac3807a : SUCCESS in 46m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a42fbec28a224d89941549c22bae1dbb : FAILURE in 2h 06m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/51fa62d2ce1f452f851ed9828d5f803c : SUCCESS in 46m 21s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5cf36f628f7a414183fc8cca75a90853 : SUCCESS in 33m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4f91a3c6c4f74a79b9b7547e3ddee8f4 : FAILURE in 11m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab5ca8ec9bd31043633605c894bc7e3eb6370e01","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:27:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c3e37093355d742d34b6324370aecda257ad457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 11:54:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/956a0b3099664e4983d3dbcdcc6a7920 : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e487f8bbf1864092aefffde110b608f5 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63ca763b00174812922986507265efe1 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9349131c8e41434b99c235db33c7e2f4 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5682d87190574844a9627b5b881f8a1f : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56aab25de8a44f4aaf41e8ad48def2f7 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efed930e34343f38d152b752763689f : SUCCESS in 10m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/69462d049b594b2083e3c0bac2df54c1 : SUCCESS in 6m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49e350a2f5924a27b4fecd3916965cad : SUCCESS in 7m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5d3d718cb55042639ab92ece386a5c7d : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f11615aeb6c94034a4344c34f7131b27 : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a858ab9089e34a8cbc4ec0b737bd4e6d : SUCCESS in 1h 34m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/33747b3f1ea14543b2ec3613a7a3f59e : TIMED_OUT in 2h 13m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d22f2dd656a24b9291061afb27898722 : SUCCESS in 1h 52m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8da80d25ba824762a2326b0763bc9135 : FAILURE in 1h 34m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1ccbc8b066164ae497426815b1634b36 : FAILURE in 13m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e8f2f55385146f4b28322ed4e4ac8d3 : FAILURE in 29m 54s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/688a548f554e48248dce8904b07926af : FAILURE in 48m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2aeb306079ae44cb98345b05c7e79451 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/13c83571238d48959b73bf8fc21c2308 : SUCCESS in 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51af9e21e9fa4749bdbf2380169da5ba : SUCCESS in 2h 22m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5acaaa4a7631429189b7761cdf36347b : SUCCESS in 49m 46s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d48d608c125c486b950afac94ea226de : SUCCESS in 39m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8ec4c4ce6b6f49aa8fcbfb36906f514e : FAILURE in 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab13c2f579de9f13b8f925bd7d587a53a1dc09df","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 12:31:22.000000000","message":"Patch Set 1:\n\nthat\u0027s annoying\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bca45818572e8bd758f32fefd9026bf558c0c433","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 12:31:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c091f48780546156b87fbb6c1899467654ed03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:23:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f72dea0bc947d5be57a98bee2ea8cf : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d86215190e2c4671a0d4b04eab5d278f : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2fd401b9a3448196ba695445c079f7 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4d2d97e7a074c5aa9d07fd716002e2a : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a8c32bef0cd41e89f256d9e91340708 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f34a0cee6de849b6a863a57f191bbc75 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd150e8e009d49009e56ce551f049247 : SUCCESS in 10m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/007620c3f8de4e4d86c921bc4f7b7b82 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/73c62b36272149909b307f9a541dd326 : SUCCESS in 8m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a35a953e88fa49559aa5be4ac33ce1c7 : SUCCESS in 7m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c8efaaced3343808f5ca52a47d54c0a : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ab668ebcdeb4ce19727084f421fab55 : SUCCESS in 2h 02m 04s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/4c63c6a2c52c444faa77e8dee83e0439 : SUCCESS in 1h 40m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4474c5eeafc4352bb6f21bbe861db51 : TIMED_OUT in 2h 33m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2c7ef7eedf8541b9b746e0c47f3894d2 : FAILURE in 2h 16m 03s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8fd145abfc0341e9a4e04a96a86c663e : FAILURE in 13m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f12bd778d38e4616b03ac0cd1a942c12 : FAILURE in 29m 26s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/73f22c787a294a7484283196c6c1805f : FAILURE in 45m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c9d14d6c24547a187c680d285e4e59b : SUCCESS in 1h 41m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2c8674bb723484194cbf17def294f03 : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f6a25a15eac49dea9008b02fbf20b8f : SUCCESS in 1h 57m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c436bdd693d64d329aa71a9bad8c15ea : SUCCESS in 38m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2b2a1d4b45334f42adf0616f1195eacc : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/436b988748134b6aad404afd5efde058 : FAILURE in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"022c2b25bef39bac053526bce8b673adc3bbfce7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-25 19:51:17.000000000","message":"Patch Set 1:\n\nneed https://review.opendev.org/#/c/753838/","accounts_in_message":[],"_revision_number":1},{"id":"38321109768e450386d8a7298d6842c171276b4c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-29 10:59:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fc5a9b0dfd3ec2cc0780157cfd0b2a1d5b8c19a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-29 13:38:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83b97521531e4cb18e14e4d719249c9e : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8098f6c348c240c1b6a559b3682089fb : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf6538c5d164019aa41e863015daca5 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d1e8c7f6a5040ebb0fe09ce67fbec55 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb1259829da4a97b325fe51b213cc8d : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ddeaefca6864bafa411cef3dae0d249 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169c1e0ecccb4c3499cff24c2c2d36bb : SUCCESS in 10m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5fa9f25c624f4c269134bb90796aad3e : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6af3db50afbc4aec88e2402a2e59577d : SUCCESS in 7m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0b6aae4d73d144099fb70cb260efd8c3 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8d70fd3ebdf474e9a539b85e587ea71 : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8d6f6ad7f48f4aaeac1c37470ab5c8e5 : SUCCESS in 1h 36m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ce9695fd56ed4c818d291ae4ebb1b9db : SUCCESS in 1h 39m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1f526925a0524ed7b2311cf1b9507ecd : TIMED_OUT in 2h 33m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/57a6b7946e674932a04d3233eb69a0d2 : TIMED_OUT in 2h 35m 08s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/622bc0b0b0c643a58b95f63cda7e64d6 : FAILURE in 13m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc2dbc40d24e476a8885339ec25e5b2b : FAILURE in 28m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b2765d3394244c4aa88bbce9a19794e : FAILURE in 45m 42s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b2f0994e96d147168712b89fe8db3001 : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fc21d0212214f43937468a7fc5bfed6 : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df0899c6b0364b8692204dad9652e42c : FAILURE in 2h 15m 39s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ebb7f822c474a8fa7c7dafc4116127d : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0d896aa6292c41c8ac1351c57b71c221 : SUCCESS in 40m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2216b038f3b24fae83ee458fec87809a : FAILURE in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80c172d3f006dddf0a9e3aa7ce29c8deb9d37ee5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-30 10:53:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82ff40743d2bc1910711fb9d919de3542c8fd500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 13:33:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e27385dde8642f08942be296587713f : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bc13fdbe92146b6b0da24eee3ea4ab2 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ab91bda8f24397a3ccffe0c567a723 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aaf9aeaccb74e6d85c50ac26e202eaf : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a44501639247bfb19b96c21b88c9e3 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/946ba432375545ff90424994a56ed44c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0cba86054349d5b886a618bf877f22 : SUCCESS in 11m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/64fa266e0fe441c3b53d6f82d8be551d : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1f23b9f5eba549aba45ec1a29b87fd7e : SUCCESS in 7m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8dd44472b5284535986d222254551432 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86f4cc058d8d48548695e704fdf53036 : SUCCESS in 6m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f25c99effcfb461c81bfa3afdc9e37d4 : SUCCESS in 1h 52m 48s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/68314f95592445efab86df7a6e97b01c : SUCCESS in 1h 47m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/846200b9075f4ff49da64053c8533d5e : TIMED_OUT in 2h 33m 56s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/30b96ff6683047a7a208632e00857060 : SUCCESS in 2h 05m 13s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3e5e57ea8b654693a17e732cd1a46b30 : FAILURE in 15m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c285d4ec0a8a4e52a4b431b6dd12e788 : FAILURE in 29m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2aec5cb902f749d1a438146c69ea2903 : FAILURE in 47m 59s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2b18f97592fd4a2ea8b0dd916fa0130b : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9acc7864f32a40b5a3a8cf0cb6d4bd00 : SUCCESS in 49m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1fbe6952dc5b46849c65809ba24ef4dd : TIMED_OUT in 2h 34m 59s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/557edb9680a24008950572dcdcf422a3 : SUCCESS in 47m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/95054eac185e4bed86b0e466e01731aa : SUCCESS in 45m 00s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4349146aa00943ff8a79680fd821f650 : FAILURE in 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2513d225fb83e93011db34b34d7121c7f4fe979","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-06 01:13:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ce905bf4b32a20935480056c9f8de9425db51d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 04:20:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/389952ef07614d53a8d5e6306c864dc1 : SUCCESS in 9m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0627d6c0f5624e5592827698a26c969a : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d1c3888f0e94947afefbf80cb7650d2 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa4703f072e84e9f983eb0f09930c64e : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6c02f0b77b341aebd6fb163f476d840 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35eef5ec3c3b4fd4920990cc7f959a0d : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65aa65d1dd884f98a836c9bb6e682e2b : SUCCESS in 10m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f42e8c83a8a84bb095c7fcea6ac074af : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9450a447bb514863bbeed880a92459f1 : SUCCESS in 7m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3168685b9e87420d9c6d7a7acea02441 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcb8a4c0bd8344d8b782281735b92aa6 : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13d97cffe83644f0bc80571b686d0f18 : SUCCESS in 1h 38m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/d26570ecfed4409ca0f40ff070390cda : SUCCESS in 1h 51m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1fab6e91b7724afba2cb5bf5ed9c5f72 : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a9890c235c0649ae8f737f93e25beba2 : TIMED_OUT in 2h 33m 18s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/730c0c0113974557b96cca796b65d203 : FAILURE in 15m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0829d55c442642d593aa6f5835a0369e : FAILURE in 28m 02s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0fd54e657cb24cb998e0a1abc3031c3d : FAILURE in 44m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7fdae0234e3441a7b927655b71108d95 : SUCCESS in 1h 19m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a8041a4333945a898b36b098889ba8a : SUCCESS in 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ec54817d0c124c48bba1b77c6ae8f806 : SUCCESS in 1h 54m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4047cb6dac2c4de0a18ad384472fff74 : SUCCESS in 38m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/beb43c83fd5a4e88bb2b6d37ae1508d7 : SUCCESS in 49m 40s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/69ba2245450c48109b8da06a46cbc475 : FAILURE in 13m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe2e86c455b2be429f497a615c2e6d362a0aa32a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-09 09:31:36.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3254334b11dd8a78b5e2aa7316924a6ada31a775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:33:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93797607aa2e4d0685affd46afb21387 : SUCCESS in 8m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/103ec50ce1794cf7aca9777c59ac625c : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d1199b619349f5b972bbd580c25954 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ee0973f26b4413c91e157c411a9b677 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c852144bd09404eb42ea48c76c58728 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0b37888f9fd44d38d7042df7db1ae57 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e18f0912924661b5485d10b2bddc39 : SUCCESS in 10m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df0d4e44a9d7483f9379d56e4c31ed82 : SUCCESS in 4m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b3e39ada9ce34b1b84f40e16018c95c8 : SUCCESS in 8m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e929a161901a4f3eb2f680fbe9fbd66e : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0767d9fa721f47eaac72fd4f975b2776 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80fd27a2ebce48c3aeaae32b2bbafbfa : SUCCESS in 1h 54m 06s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a20c9778846346e688e2daa8d342f254 : SUCCESS in 1h 41m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d1d4187bcd8b4c6db0179b841f4873b2 : SUCCESS in 1h 56m 02s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/beac244e70b04f2c90325e117cdfe4ee : SUCCESS in 2h 28m 51s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b7701d6d466c49b0bb7a7bc69d7a3282 : FAILURE in 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0906a4154e734fada4508ea36cdceedb : FAILURE in 28m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c40f7888c7146499e42b5c7c258c9bc : FAILURE in 50m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b529e8c4ef4347908cd31f5476c5e62d : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/125a01120f0848ba8e0004d9c47b7525 : SUCCESS in 43m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fe503deb597f4b5c8c69e3f74c73cc95 : SUCCESS in 1h 38m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eba4de9aedba40a4b8d355031a727b18 : SUCCESS in 44m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1f97b1f35b2d4a9a9a185e39860651c0 : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/23424dde4e7f4717b2b2cd598c4f1eb3 : FAILURE in 14m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f28a109cea691817d4ff71db34d6e6f05b8cd15e","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-10-12 15:55:00.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7a11281ffd4bc9bc8a447d784dcc33ba3cc04ff7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-16 08:04:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"18dea71d054f24557eece926c25de7ad00776586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 10:46:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a021d550945141dd8d24dbab98a04cde : SUCCESS in 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b50c9e2ea81404caac71efb20ff11fd : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca179f567e6496d9c57b1b350af266f : SUCCESS in 8m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fea87dd806b416ea026dee1e586bb2c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1a0d8fbb5374ec7add2cf2dbaef8737 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8858deb17f8646e0b7519fad66ff846d : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f39b44c8f840c29681fa84223ffd1f : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c80a6911803a4df8bbb8b294b2f3c386 : SUCCESS in 6m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/810301e5fcfb4adab40a31d016f85c43 : SUCCESS in 11m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/840d27c26b25483796f7e37e596c7e55 : SUCCESS in 8m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/190965171307427cbb5cfaee87bd9831 : SUCCESS in 8m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02ca8368ccc94d12a826a87e2a402603 : SUCCESS in 1h 36m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/43ced89c3dd64c29b9c037b1c5ba2934 : SUCCESS in 1h 34m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0be686d0fed647b3a29d4fbdf16843d6 : FAILURE in 1h 53m 50s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f85e7536ce4c43e4aa68074ce11f9cd3 : TIMED_OUT in 2h 34m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/786f039111e04c44bce303ad0ecfa6ef : FAILURE in 17m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1fdf8bba003b44fdaffc37165a646ac1 : FAILURE in 31m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7266033ff0bc4ad39b0e7c2c9917df35 : FAILURE in 49m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/469877a322624653bd7cfe8476a0bb2f : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1db2991211634321b732b674aa27bb12 : SUCCESS in 43m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4a1b0f13e4c24bff9125b6c7c0531cc6 : SUCCESS in 2h 22m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/756190bef2f74a3fbc7fb2bbfb1bed85 : SUCCESS in 40m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/8685a2d434db4e81bc5c91a59081c7f8 : FAILURE in 1h 27m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a3c609a81b654ba29b95616bc9d51e0e : FAILURE in 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e7fb644ae11c12b4a20a466905bcab1ca96fc91","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-21 12:23:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9b508577966cb18ef68f5f908d4277a1b1cca4bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 15:11:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d71223900a8846928d43b7edeb2b0dbc : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48fef3f17b494172b0b6f8dfa04c016e : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0de037c14744813b98840385c71aadb : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e069093e93f24e62a2e01622c283e773 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1df8409c15f54ee2a3ae8a94456ae1e7 : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4706cd5f6054852891b03c9dd7d1a76 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f300a5e4d9b42fea6336844e2e7206c : SUCCESS in 11m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/91f9054f1835477b9c2f50b26cd25079 : SUCCESS in 7m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bb1c8e0ffcd74d9fb3fffae89d37a715 : SUCCESS in 8m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/95b25af66ad94d118a7e5ea73b609839 : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25c003d7d6364edba85a12abc52af344 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dca190874f6347fc8727b6dd4b5c833e : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/00ded87f431945469aed51c14df091d1 : SUCCESS in 1h 51m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79f5c5e376fb4be4a5948c42824c3de4 : FAILURE in 2h 23m 29s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fc6a066dcae94106bc3b37185925952e : TIMED_OUT in 2h 36m 11s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7a911ab75e114c909ecceeb6ebe23cf5 : FAILURE in 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6b221bc50a3424eaf63ad071ff9fc93 : FAILURE in 1h 09m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/23758a7d1df1453fb8916c2614f49486 : FAILURE in 57m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8b23577837f64039bcd43e4dcd92afd6 : SUCCESS in 1h 26m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fdbc7e2313445c5924f5fd6812291f2 : SUCCESS in 38m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c62db081ffc4df2906da87a7a9e2e2b : SUCCESS in 2h 22m 31s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71bd6d09484f48deab570e0a372be640 : SUCCESS in 46m 55s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/bbe01e24a02c4ad39b335db36b050ead : SUCCESS in 1h 41m 45s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/405e338bfe2a4bf7b00a52f571c32c8a : FAILURE in 14m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dffdc6e7cdfb6196ec98e304eae913d549059774","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:33:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e160700dfe383b8e2ba396d4381e0b476e51b0ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 13:00:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abad0b47fa7a4b95a4eaef8ed80c3bbd : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d36fac5e55b545e6b54ce3a81b4db7f7 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07675586533488e91b8cd64c3d95824 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c25c70c50e4b4eb5ab9d7d1782cada15 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e66a70451924f0eb31c990d6ae5dc63 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a90383f4ee2b40e59b174b279c7d3c08 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8018f07543724bf2931d85e9e66ac2dd : SUCCESS in 12m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4cbde4f15e684f7b99b1b1eac1df9bb8 : SUCCESS in 7m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28e83b39d06547caa0b58a85003cb11a : SUCCESS in 9m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/12a84a9a2e5142549fc361eb43a43382 : SUCCESS in 9m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31556ce55c3940e2ae5ede6e004e33b3 : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c23560a8d9bc42e8bc3e8348e3096971 : SUCCESS in 1h 43m 01s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b9381d36420143778610b2c7df95f895 : SUCCESS in 1h 45m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f869f48ecd8c4c6ea3f3cee9e16662e3 : SUCCESS in 2h 05m 17s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4890708868104a139fd83e12ee84f1ec : SUCCESS in 1h 28m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b8f8824dd4bb4a8b8fadfcc93fc494c9 : FAILURE in 46m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/847a35f8adb847d99623f4b71fc116b7 : FAILURE in 27m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/05300082f2ca46359a8c408720f04eb5 : FAILURE in 50m 18s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/062231eef1584cde88d71d9c5d1b240c : SUCCESS in 1h 25m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd37fdb6ba2f4701ae8a1e904a0bf6cd : SUCCESS in 48m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4550ff8f440c4481bf9902d8636e3916 : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dd5ff719c1b54998b475f88c9155cdb6 : SUCCESS in 46m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5b78a8b74dd04f56baa1c8d825e63bb3 : SUCCESS in 45m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/b96eaecf02ab4fe484175a354e630b8e : FAILURE in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"babd5f82a724d77d9d80d77d3d94c5c9eb3ebf09","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-23 10:48:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"685eeb9de3a3b86c3fb83515cc9295a2c77d8f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-26 00:09:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cfe5e5ccb7e018f88acff02c29e7c62fe0423a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 02:10:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2ac4fbd317040b782279097c6770d6b : SUCCESS in 9m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6340491679b4c69bc715d30d68edbd3 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d780f7120fd14bf387007c2437629074 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/718df246e21e41668f875d73237f0fa2 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8a7e43e8e84102a8f9fe10547c9424 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e978ed1b87a488587a02776ee4fa584 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2d83b27e66421fb6a0c4df9d5fc3ab : SUCCESS in 10m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6b40cf24d109473191ae45d973326cfc : SUCCESS in 11m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c464abec3be43fa8ece597a24dc9479 : SUCCESS in 13m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f63a845f33df47c5a634ce8b8516f5fa : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bc3cb729bae94d8daf813e8ebaf8468f : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9ae31655aa4f42d6b838afde5389b688 : SUCCESS in 1h 26m 53s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f12d65de4401485cb22b59952dddc11d : SUCCESS in 1h 30m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/67c53fdc337c4c03b7e5dc8fcad5b5fd : SUCCESS in 1h 44m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/84bf73f95cd24d5fa692e5a038a09dfd : SUCCESS in 1h 52m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9fc00ca5cee748cba1aeb58191a66afc : FAILURE in 13m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5e197c83a2a843239de490c825020894 : FAILURE in 27m 17s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c09df35beae441bea68e4d4820d997d8 : FAILURE in 48m 35s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7986bb55cba14851b10ce71c8d6fb4b4 : FAILURE in 1h 26m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17d7f32124b443ebb005761164077093 : SUCCESS in 41m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/02443c8fe4cf40febc28ea994e01809d : SUCCESS in 1h 57m 43s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8191cf2e1384d9b88a4561e0ace6085 : SUCCESS in 1h 14m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/65ee0bb431d5459bbe654be228896d08 : SUCCESS in 38m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/434de835d1bd4fa2a1dcfff0b5b24b4a : FAILURE in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9222c74e3b4e480897657c860932b17c2c7071cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 14:04:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"13e232e1a2e42db8bea7c04a6e539feabec1c55c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:32:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a742a613b57247b9a3a59b50ba5a648f : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/161b4bb054e64507bbd5249b74f60ce7 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05113fc9c4b24db3ab260d596bb6b1d6 : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ac52e5a95b240c09d66f05a33cccb84 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9572f6358250401ba2b67031015bb634 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60a5db79520a46fabb493897d3fd58d8 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b433a4127946dc8c52a5ac230b3ef6 : SUCCESS in 11m 08s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c5a8a17057804a508827ebb287fd69be : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9a549c54fc354f81bd9b52a252e1523c : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/17eb48fb29bc432e82fd3dcb90bc4766 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f66bb5976e8b43b0bf2d149d1a1e46f9 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6adea3d260b941aeb2db4955391d8c5c : SUCCESS in 1h 23m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/458f0f6eac264c32ab9d8a43950e92ba : SUCCESS in 1h 39m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/696dec26e85c493b8733cd8f5c794c0e : SUCCESS in 1h 46m 34s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/39cca88804e3485994392dc24cf529eb : SUCCESS in 1h 59m 56s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c0d8829eeb334d42b2211439b58b9844 : FAILURE in 13m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2f251143e1e8466bbb4ff18ed348ebeb : FAILURE in 38m 12s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5485ce09ab343e79592326fc924ec2e : FAILURE in 49m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/78511142ddf3467b84f2f74f6ddf3df1 : SUCCESS in 1h 26m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2f158fe987b4fc4bd1c1920993f3ee7 : SUCCESS in 45m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a870db45d69b4812835174606ee24b26 : FAILURE in 1h 57m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e552e6cd5b6d4f349061018e045c7192 : FAILURE in 37m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/13f1dd4c76a74f499d4f411b6ad9fd9a : SUCCESS in 44m 06s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c3083127ffa74881849fdcc5508c6d67 : FAILURE in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"321e84cb891e648b733e6e3cf40529021a5acce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 17:32:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b79fb738a0080cc3bd5ad2c7cc77d70c90cce8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 00:04:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b427cd94b13b46deba136bc4c44e15e7 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73915de64b074464aee39b289f6f319f : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5eb31bf0bba04a5c88eb35656153c8c3 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e1a8ecf5d744b36ab274e2743ff56cf : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/184800dd25c4430ca6997a3e2e6b65ad : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8e1636fbb74d1c812781220bb5d003 : SUCCESS in 11m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7f2c3587d8314e558404cc63a1bb0819 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04794b3ee0ac446bbd24814b826c0f20 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3cd9982e13304a15ac7cb798f84735eb : SUCCESS in 2h 10m 46s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a79fb0867ae94d8297da994cacd99cf6 : SUCCESS in 2h 09m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f8bf3d620c64b52986f31369da01a55 : SUCCESS in 2h 23m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/753cd41d1cf74e8eb5c1b5cd5591b1d4 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":3},{"id":"5184a83c802d1fc273e256332e136cd1d34b69d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 00:04:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"61eb146831addeb006bc9c4c3cd40667c9ecdf86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 00:06:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75a21365cd140539949107f8ffe501d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3ce6d3b39d64a5601c803a9a328d7e51594ce423","revisions":{"e85d35a650d391b028a1faefcc5a2b270c1ddb01":{"kind":"REWORK","_number":1,"created":"2020-09-24 10:03:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/57/754057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/754057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/754057/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-24 10:03:49.000000000","tz":0},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e85d35a650d391b028a1faefcc5a2b270c1ddb01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e85d35a650d391b028a1faefcc5a2b270c1ddb01"}]},"branch":"refs/heads/stable/train"},"9b6e53f3b928e46fb5518a0312993dfe41cdc414":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-29 10:59:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/57/754057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/754057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/754057/2"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-29 10:59:33.000000000","tz":0},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nDepends-On: https://review.opendev.org/#/c/753838\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b6e53f3b928e46fb5518a0312993dfe41cdc414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b6e53f3b928e46fb5518a0312993dfe41cdc414"}]},"branch":"refs/heads/stable/train"},"3ce6d3b39d64a5601c803a9a328d7e51594ce423":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-11-26 00:09:01.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/57/754057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/57/754057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/57/754057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/57/754057/3"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-26 00:08:54.000000000","tz":0},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ce6d3b39d64a5601c803a9a328d7e51594ce423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ce6d3b39d64a5601c803a9a328d7e51594ce423"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
