)]}'
{"id":"openstack%2Foctavia~754062","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ie2c748ff69c8fd736e9c67d686068a33d803222e","project":"openstack/octavia","branch":"stable/stein","topic":"fix-house-keeping-stable/stein","hashtags":[],"change_id":"Ie2c748ff69c8fd736e9c67d686068a33d803222e","subject":"Ignore DELETED amphorae when performing certificate rotation","status":"MERGED","created":"2020-09-24 10:33:44.000000000","updated":"2020-10-22 12:54:00.000000000","submitted":"2020-10-22 12:52:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754062-1603371129378-e8c13dd9","meta_rev_id":"bed84edcc43e4ce308f58fad320ef9d48fb2a0bb","_number":754062,"virtual_id_number":754062,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 12:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-25 09:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-29 16:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-29 16:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-09-25 01:31:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"date":"2020-09-24 22:56:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 22:56:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-09-25 01:31:38.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-09-25 09:26:56.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-09-29 16:16:32.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-22 12:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"022c885140c0e49a2e7930059e331ae03998110c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 10:33:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c0434e2511c4f10e34b9402e97231d2019f95bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 13:22:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed2f68a2e1854ea0a52c8421ef718138 : SUCCESS in 8m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd00fc2d606c4e62b11b8a6d22564e6b : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f57e322b50f48599c668f9fd91d51e0 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2380caa84b37425293a29f0e678afc2e : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ed046a06d09d481c9ba59029a16cef9b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a34f0cc4f6d432f9daa20dfe7c3a344 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e92f97ac86b40e7b08dc68bab3f3f6b : SUCCESS in 7m 24s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6af6db68cc3146b8bec2f0433692c0d7 : SUCCESS in 1h 21m 12s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/aef6cd1ecc0d46618e8fa200bd58b5dc : SUCCESS in 1h 17m 48s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b935053430c48a2b1c1b219648550fa : SUCCESS in 1h 19m 37s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f99c68e255aa4476b272c99b7cad9e09 : SUCCESS in 1h 18m 42s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/54603e701be9436e89b9d0093ddaf189 : SUCCESS in 6m 44s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cd1250f532f5443d9e3866913a002652 : SUCCESS in 8m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/30cd9a7c0637477dad08466771302593 : SUCCESS in 6m 30s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/273877d0588c4dd1bbbb780f083107db : SUCCESS in 58m 48s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/5069a1ee1ae8480a9c5c89e38eb10998 : SUCCESS in 1h 04m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad0842adc8674ea59ac456fd130d2bb0 : TIMED_OUT in 2h 13m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/64b09f669d48450f819347ff48542f06 : SUCCESS in 1h 22m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/613dea5b12304d3a9a00a70b917699fa : SUCCESS in 1h 21m 38s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/51e2ae1254534caf8d86e5d7b1885140 : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f4d57dfe3b534efcaf16087b54c350c5 : FAILURE in 11m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b0e30d125c244f4dbb2f2cefd3ba0e3a : FAILURE in 46m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/fae6e982df4a4dccacf4e5199ff7301b : FAILURE in 45m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e7cc8a58fef49a8a95bcb3e86838f8d : SUCCESS in 1h 07m 49s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f413d86dd545456fb32474a76621250c : SUCCESS in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"307017711bfba561afd00a0594bb0bb34b1bdec2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 13:29:39.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e1204aa6df8d4d13f55d7b5948a12ffe0c9a1f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:35:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe75332d75e14caea8c55270590b1f43 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d9e44081ebb4a219c60376cb3fc954b : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6deae3553047af9a1dbd24110e9907 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1cb947452e104d0dba9c5a9376647ddf : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dab5c18932914412ba182b55469e237b : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46ff49306803436ab502606f3e89de83 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dfb6ac9f6104f629298acebd70fa152 : SUCCESS in 6m 36s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/66b000f87f354989bc3be383fd290bec : SUCCESS in 1h 18m 33s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b04881a606664b78a8ea8228ab8dda67 : SUCCESS in 1h 17m 40s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f56643cd99db4ebfbfac25db8cea604b : SUCCESS in 1h 21m 08s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/7ea2f995d54a44058239910d75bb255e : SUCCESS in 47m 11s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f1333b5362d1470b8477fa37036894a1 : SUCCESS in 6m 27s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f6107669cbe743b58ace4e0fe06d99e9 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/874cc0587bc24b24a21ff484fd84f28d : SUCCESS in 8m 05s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ddfb974e4544025bd9564a34351f7cf : SUCCESS in 1h 02m 39s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/ad319975643649c4b9ada3aafa07c6ad : SUCCESS in 55m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7bcbc7a7906c483bb8dd79c8eb878dde : SUCCESS in 1h 20m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/89affb4d8b7e425c9c1946dcc4bcfbf1 : SUCCESS in 1h 44m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8b0ef88b4e24b70bc890e5c7b6baa8e : SUCCESS in 1h 13m 33s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/a956942a47c345999a71ea65effc080b : FAILURE in 1h 15m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/115c3dce2bcc4f7198e204a7303325ea : FAILURE in 11m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cb634bf61ab14505ac7bfba0966ba664 : FAILURE in 1h 00m 52s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/11a3d089595044a6b61ce326ff08d880 : FAILURE in 42m 32s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0dd2fb1324a9430db3f8504e67d909c0 : FAILURE in 33m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26e8816f880c4f79b09137cd94271078 : SUCCESS in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56f7eae9ed6d7aefc89f0dc9f9fe771fb1073fe5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-09-24 22:56:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da993d3ade6d9be93f2342a522c7e375ba9adce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:39:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29955502c8174540a1b5e9c544285345 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec6852cecdfb42c39617509a03656f99 : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ae816f2ad04d19bfa746467895ef35 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84dbe2ecd45e4b5da049a74d76aa200d : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15681e7ebd2044e888195fcb478b1e47 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/132f2ca2568f4f19bbbfcf3fc5eec614 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b614049187b4f38aeb1a9d3a28e4f1d : SUCCESS in 5m 31s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/9fc97da4920e42ba9a971e7b5156cb4e : SUCCESS in 1h 22m 54s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ea8867003349402d8681429a1fcadbbc : SUCCESS in 1h 29m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c01f0bb0162e434bb98f120ed6a645c8 : SUCCESS in 1h 22m 39s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/75d429a551fa4339bc8ec03d521db985 : SUCCESS in 1h 20m 19s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7260a06ae9a44ee0831efadea43b767c : SUCCESS in 9m 04s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ee5c6c11c4584f58b27e6099610923b9 : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5bee630efb74bbb8389b688350bf061 : SUCCESS in 7m 27s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e46647d2cab46a2824c03699a5262f8 : SUCCESS in 1h 01m 31s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/9891b6b152ab41eab751069a8e34dbc1 : SUCCESS in 1h 06m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0d9b1b61147420e8e79aa0edb8e0811 : SUCCESS in 1h 25m 03s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/3fd480503d984907a8ed7cb4881bef05 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95cf1d0d5d0f4bb480dcc378e9b8c274 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/94b09ebadce84c9cac746d6132045c59 : SUCCESS in 1h 27m 46s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c6384d93513b49bf9748b1d208aa730d : FAILURE in 11m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8af5b2649a57459f87a1e863738b3512 : FAILURE in 43m 42s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/216726eb4c23448d9addac30264e039c : FAILURE in 59m 50s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0161334199424686a242c8a835d8384b : SUCCESS in 1h 16m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c5ac2f8fb9174e5493bbbcc745af7447 : SUCCESS in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8513aa475ae35d1f4067f7bbb7c9c262de00656","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-09-25 01:31:38.000000000","message":"Patch Set 2: Backport-Candidate+1\n\nSimple and important fix","accounts_in_message":[],"_revision_number":2},{"id":"09c65c58f3a9ff6eeff12a6a1110ae298a32b9b2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:26:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"57a8ad7e8afbc84896c6041e4ceb4dc29a76de67","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-09-29 16:16:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"78d4a62f6c96efc1365738d8400b6536c0a12faa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:20:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26f35f7ad6e765cca3d71b1f91467ef9efea7f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95358c62684c4dbd923612c780c3dfe5 : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ec83895e5e4951aa205891f75e0b9c : SUCCESS in 9m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d823c428d93b4fbc92d741d9fea70392 : SUCCESS in 7m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/11985a0d5dc747c7999ccfb68b8bcc34 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f7da8577df04543bc25811253be0e25 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b19adcce52924c6289478db95f7a8032 : SUCCESS in 9m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3e769ff5ab184270a8a0fb075e2f55fe : SUCCESS in 10m 19s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c77ebb3a5dda4140bea53b746a9640e8 : SUCCESS in 10m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2946b967cc9544369b0939d9f18db467 : SUCCESS in 10m 11s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e320d45a58245a2bbec4bf53c524ecb : SUCCESS in 1h 03m 11s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/01bb941762564ada88b00f9bd6b47c96 : SUCCESS in 57m 40s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/7795635a7df74f4299d5e4e593182868 : RETRY_LIMIT in 15m 41s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/8b491d64c8b6454d9b1b9611e3944b72 : SUCCESS in 1h 20m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/74d0f346c5924c1792da0216f3083c63 : SUCCESS in 1h 26m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/072b6e80ef6c4708a9bec22f7eae40ee : SUCCESS in 1h 41m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeb8e96c2ee24bb3976d98473e42c957 : SUCCESS in 1h 40m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6fa7d65cf08b35a012b2f2721d49b7f9b7b3ebf7","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-22 08:23:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2dac24ed6ddf185206cab4d0468648246cd4fcac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:35:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d4db08c74e14bcdb43be7d87b5206b6 : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2aaf5993d4174813acebf8f98377d206 : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54ab9717bdc4c46a81f1fe22be9f6a7 : SUCCESS in 7m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3fb849ae739465bab3664b603e3338e : SUCCESS in 7m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/da256db3c7274431b61eec26cad60ae1 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e114b25003944a248a675eca1ffdc2ca : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7683826c8ae24de0b6effe1e0203d607 : SUCCESS in 8m 27s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/5742224f78024e97915f385f3c065777 : SUCCESS in 1h 19m 25s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a671f8560f5a4295896e1ee559cbb3e6 : SUCCESS in 1h 22m 44s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0843cd71b0d4184b09aeaec82dac1b7 : RETRY_LIMIT in 17m 57s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d92fb153505f400eb26709cd70c44500 : SUCCESS in 1h 19m 16s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fc08e90ee8a64a6e8fc367daec50d1df : SUCCESS in 7m 56s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cafcccf4ac884b87af1d74899d09972e : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2762dce5a9942a5b5490cd19f82df15 : SUCCESS in 8m 16s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e5124a6afdf4446e85cc8e9789edd2e2 : SUCCESS in 1h 08m 16s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/981508085e474a019c3b574ef7eeae7b : SUCCESS in 53m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2907a2b907ca4e05a015100c00508867 : SUCCESS in 1h 26m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8460bdb2f0794295863b95567c3d78db : SUCCESS in 1h 32m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91bbd5ec8d41466e99a84a5ed723fc51 : SUCCESS in 1h 23m 59s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/16d56901b9ee49f19b333e30eedeb56e : SUCCESS in 1h 31m 37s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9ea1bbe7094048ac9b82cc821a308bee : FAILURE in 11m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/30669539daf04a60bcdfb01fa04a4800 : FAILURE in 47m 25s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/3ed9718c205745db9d5ee4b6d484fcfe : FAILURE in 48m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c3812560ab849918bc134b5b0d45eb8 : SUCCESS in 1h 34m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1f0aac4635624bb9b4d7c6636bd39cbf : SUCCESS in 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80be4ee038253069b9f8d24f6bb98dcb38a976ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:36:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a47f9a4b016276d35dcccc181c1b64dac0af54b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:52:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da4270ed7c263b450b5988f76c37c5edb5a9c27d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:52:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b8a5814862942d1b34f97979a0224ea : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2609bd60144c7f834b07c61fde110b : SUCCESS in 8m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1348c4f9bd3f4b8cae7ddd4b1d8cd0f9 : SUCCESS in 7m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c64f24c2b6348c5a54a4dd5a31cacca : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d9da4048e72477589552b1f324f028d : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36eca979f5a4e12999730ee1a4a3cc6 : SUCCESS in 9m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4b00795e37304749a0e1016790f87e70 : SUCCESS in 10m 36s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/2732dcb9a11c43eba698a24dfd17cc07 : SUCCESS in 12m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bfabd9c7569420d98a8b204413dee80 : SUCCESS in 9m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ae19ab1e4ca942ada0ab0caf52508947 : SUCCESS in 59m 58s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/dd2b5061cd124c1aab5c2a435a33a775 : SUCCESS in 1h 07m 04s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/d93a6b1167224ed09c1df65e2a72a713 : SUCCESS in 1h 21m 08s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b6a6940c388949aabfa458d7ce6898db : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bf5df9d8e664e34b0e6c2317e7c5efd : SUCCESS in 1h 22m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/f17866285c614783bc5bfcf2604f46d6 : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/65759316936647b4ac6f1b45d6322999 : SUCCESS in 1h 49m 24s","accounts_in_message":[],"_revision_number":2},{"id":"aa6e6e7b889682a3b9acfcaa955643ffd97cd57c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 12:54:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bc5b8ff27044e1895240ecfde90b04 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d061dca5353012afb66ac0aaaf7b92b7f7ecd58b","revisions":{"ead66231f90cacd0b0f947bf170e4331a6ccbd78":{"kind":"REWORK","_number":1,"created":"2020-09-24 10:33:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/754062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/754062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/754062/1"}}},"commit":{"parents":[{"commit":"0a6011f7784303f7b1d2f55ca801f2e27d146419","subject":"Fixes API list handling of unscoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6011f7784303f7b1d2f55ca801f2e27d146419"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 10:33:44.000000000","tz":0},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ead66231f90cacd0b0f947bf170e4331a6ccbd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ead66231f90cacd0b0f947bf170e4331a6ccbd78"}]},"branch":"refs/heads/stable/stein"},"d061dca5353012afb66ac0aaaf7b92b7f7ecd58b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-24 13:29:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/62/754062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/754062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/754062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/754062/2"}}},"commit":{"parents":[{"commit":"ea507d296c2165141e18e5fb25d44e1437291d0a","subject":"Fix the amphora no-op driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea507d296c2165141e18e5fb25d44e1437291d0a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-09-23 12:24:49.000000000","tz":720},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 13:29:19.000000000","tz":120},"subject":"Ignore DELETED amphorae when performing certificate rotation","message":"Ignore DELETED amphorae when performing certificate rotation\n\nDeleted amphorae should not be included when housekeeping service is\nperforming certificate rotation. Otherwise, deleted amphorae will\nbecome ERROR.\n\nStory: #2007635\nTask: #39681\n\nChange-Id: Ie2c748ff69c8fd736e9c67d686068a33d803222e\n(cherry picked from commit f3eff29ce17e5d27867b6e6f4049ae6d51d2751c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d061dca5353012afb66ac0aaaf7b92b7f7ecd58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d061dca5353012afb66ac0aaaf7b92b7f7ecd58b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
