)]}'
{"id":"openstack%2Foctavia~754094","triplet_id":"openstack%2Foctavia~stable%2Fstein~Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2","subject":"Fix the amphora no-op driver","status":"MERGED","created":"2020-09-24 13:29:39.000000000","updated":"2020-10-21 20:55:41.000000000","submitted":"2020-10-21 20:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754094-1603313549630-b6344793","meta_rev_id":"f6010d525c7d97d535e20434c20155454e41ce42","_number":754094,"virtual_id_number":754094,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-21 20:52:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-09-25 09:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-10-21 18:20:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-10-21 18:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 09:29:11.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-21 18:20:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-21 20:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0151bdb83cef9ccdd871b7b53af482361d669692","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 13:29:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e83293c20d1e11a455ac465c5d355ad67446771b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 15:21:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0efccafe9509441d861b31aa40f7c7a6 : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c278b9b3ec21401ca87e3d626a8761d1 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a9b365dc9e47cb99a085d7ed409f8a : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1b53340a0c5c4ea2af90db907f013581 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1fa2a7c189494539948ae78550beeb27 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb47d60cbea4bb8b28111dc3f2c392d : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78ca083fafc41df895089ca1b7fc516 : SUCCESS in 7m 16s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/551acacd9b374ea9a62bfdb73ff1d215 : SUCCESS in 1h 23m 09s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/0222f03fcc214cec9321bba00c0502cb : SUCCESS in 1h 16m 58s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dca0bcae29b64a4eae67ae32eddec29e : SUCCESS in 1h 10m 04s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/78966d782d7740a384066f0a547164f0 : SUCCESS in 1h 08m 54s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e0a8d607503f4736add4c31163230afe : SUCCESS in 8m 06s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cc0f32403bf847d98dc361c99b769474 : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/60f0385eff004047ae7b1101a0595e5b : SUCCESS in 6m 58s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/029c74e315454bc5a59d50b796e84da2 : SUCCESS in 49m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/1331a5a59f774f16af52ae91e03e4708 : SUCCESS in 51m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed09afe0499d4809b2024a594c1199bc : SUCCESS in 1h 13m 59s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c75eccd2d97c499a93f3397cda152c71 : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e43dac02b36544c9b77d76a7e84c3333 : SUCCESS in 1h 31m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/694d6c36859646818e09218afbd511d7 : SUCCESS in 1h 21m 17s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/fc7aa9dbebff477daec7b4b451a7bb20 : FAILURE in 11m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/975bb2dad38b4dc0a9132fc95442fc91 : FAILURE in 44m 07s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b4a3e1c32fa4c09845229cac812a47d : FAILURE in 53m 54s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/73835a8dd072448d84b925076b20758f : SUCCESS in 1h 10m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1dc8f1e7d81e46c886d58b5519805414 : SUCCESS in 1h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83dd54d38348efb92f2472a36084cc8fbe841b31","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:29:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b53a9708366724c66ca6c1d3de1138c914568f84","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:20:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"7118772c824579bdb00f116c421b7026d997972f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 18:20:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf90bb5959c1a0d5ee651b4fc133c0fa5cf85849","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f65ec5d9f72a1a6f9cffb1a0ad95a46b5aa63011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:52:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7f40e8a8337406bb52dbe0d3c9920bd : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78bce09ceaf4426b82f533cd3bac8bfb : SUCCESS in 8m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9494c8af1da4e99ba17c1cd3f9714e4 : SUCCESS in 7m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f95d2ee6665a4d14a2ac1bc78e6c1db6 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a57c6e5879654c60a8dc8c1c0085f952 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9213c7234e434dd2a8fd1724f25853c8 : SUCCESS in 9m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a8cedfd136b446378f386d9f8072c208 : SUCCESS in 10m 47s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/b2e4ed9620be471696517562ce4f0575 : SUCCESS in 11m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/353f88f35fef484ca1872ef1e0392e68 : SUCCESS in 9m 33s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30e4afbe7ff04fe4924e30c1353beac9 : SUCCESS in 1h 07m 50s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/8cf89212dece49cabd58fbd4933bea3f : SUCCESS in 1h 08m 19s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/c7286e808c8641b78bd0e7a9974db5af : SUCCESS in 1h 22m 52s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/a68a3d985c3c49739e427e3a57d89721 : SUCCESS in 1h 24m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfd11e62105645a3b174f89a49b5f6cb : SUCCESS in 1h 28m 38s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/468bc30c163e4d05b98161d803c2e521 : SUCCESS in 1h 21m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71c3a7c5b2854ac88dbd54397640eadb : SUCCESS in 1h 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f6010d525c7d97d535e20434c20155454e41ce42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 20:55:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbcf2866371c421c8211d05ddddf8306 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea507d296c2165141e18e5fb25d44e1437291d0a","revisions":{"ea507d296c2165141e18e5fb25d44e1437291d0a":{"kind":"REWORK","_number":1,"created":"2020-09-24 13:29:39.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/94/754094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/94/754094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/754094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/754094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/94/754094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/94/754094/1"}}},"commit":{"parents":[{"commit":"0a6011f7784303f7b1d2f55ca801f2e27d146419","subject":"Fixes API list handling of unscoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6011f7784303f7b1d2f55ca801f2e27d146419"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-09-16 17:05:11.000000000","tz":-420},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 13:28:53.000000000","tz":120},"subject":"Fix the amphora no-op driver","message":"Fix the amphora no-op driver\n\nThe amphora no-op driver had the wrong method signature for the\nupdate_amphora_agent_config method.\n\nThis patch corrects that issue.\n\nChange-Id: Ib1b0df3b7227d8a8dd68276e279cae1c4974ded2\n(cherry picked from commit 5defc1e8a4adc6c8a8420b2005691845d0121939)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea507d296c2165141e18e5fb25d44e1437291d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ea507d296c2165141e18e5fb25d44e1437291d0a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
