)]}'
{"id":"openstack%2Foctavia~754107","triplet_id":"openstack%2Foctavia~stable%2Fussuri~If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0","subject":"Fix AttributeError on TLS-enabled pool provisioning","status":"MERGED","created":"2020-09-24 14:11:25.000000000","updated":"2021-03-01 14:56:30.000000000","submitted":"2021-03-01 14:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754107","meta_rev_id":"0c0cae554190d320de56858a3aadce37fca4df8b","_number":754107,"virtual_id_number":754107,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 14:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-15 08:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-01-14 21:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-01-27 17:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-01-27 17:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 16:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 09:28:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-14 09:21:34.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2021-01-14 14:18:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-14 21:52:50.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-27 17:54:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"545dc0c94aa04f3e69c5c80d53111a81df85f5ea","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 14:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f369f8270c10c9ad7b8d98d30ed8580b408bbd7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:48:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef7d6c0610e449bc8b3b570bfdb4a771 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dba588279ee4920a857286413944fe1 : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0dd61f81a7d42a4a41b2d5712a63309 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08af030ea09428f87389ca36d411ddd : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f026b867529348bf8021a9af0e24dff4 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf6b3447cc544c0bf131fa4dbf9d7c4 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34cf2d674bb45f58d7857740c243446 : SUCCESS in 5m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8c1cb043a6d74195aba958ff866c4ed8 : SUCCESS in 5m 44s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/501b6f124bff4466b19e7774b9330c54 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1a26885b90e4759a6d13b25fa3ca51f : SUCCESS in 8m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0f4a4ba5ba8f4eff86cfc16485e6bd1a : SUCCESS in 1h 33m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/003533a3f4a5411ca3560c2105a3eed3 : SUCCESS in 1h 59m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a32615d858464498a8a026a650d9d7cd : SUCCESS in 42m 59s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/39da2398c7f84c51b09f3be7404a1ef9 : SUCCESS in 40m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c9c6338aa7a248eb98c7677f101c3bf4 : SUCCESS in 42m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0ac2d21233647d6bd2c9173a5cfea93 : SUCCESS in 1h 18m 02s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/799df127a9bb46bf8258c646bb185edd : SUCCESS in 2h 03m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e972cb74c9f241e099d9c900cb49adee : FAILURE in 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3c8bd8937c12434892782cf4bf92f18b : SUCCESS in 2h 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62e10115d212c99823a5beaa9e72bf491b53e00a","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:28:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1074853939405e53b5aa560bb65bee852a2aabb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-23 08:01:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"433cc2f2628b5df59326bbd2a9745073810ac0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:13:01.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a3bd08ad4c48f3b877b9a12a710589 : SUCCESS in 14m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b01fa9766e2455fb4bdd8491591f458 : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eaf300dff3e4f098a2691f2db4676fc : FAILURE in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23887ab3aeeb458489d28e746f8d2d7f : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c2aff21150b4e3bb256ca2ce7d57fa7 : SUCCESS in 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9e8f28557a4e5c85d4dcfb4e28dc83 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72386ae13f24e74bcfbd3f1fa479693 : SUCCESS in 15m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/80293418e51d44c4b7d74ad3be4ad5fc : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0cdf1e5c630f4f66a37b5caae7aa6e9e : SUCCESS in 16m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45ece4c0264d4ed5832039eb1890c0b7 : SUCCESS in 6m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/94f3881de73a4cd2a24af6ca76188647 : SUCCESS in 2h 03m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c51bf81733543cfb945dce2f6ad1fa0 : SUCCESS in 1h 55m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2e0214009b9847b4b69799b78ef3aaf7 : SUCCESS in 48m 48s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b65107fbc0a64ef29e1a50feb03ede06 : SUCCESS in 1h 06m 13s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e94a9b1eb480400ca0ed583e2bef2381 : SUCCESS in 38m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f049da86bf145ac9532eb584360b96f : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/84e6af0156ee4e69a29a6279809ec382 : SUCCESS in 2h 02m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ee6eb6fba7347429c893a2132181158 : FAILURE in 50m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/79326e61807c40db92ecc57f1cfd90fc : SUCCESS in 1h 49m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e68f0b81566749ed056c42d629e996348678a80","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 09:21:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c909b889dbda0a80f8a8c67add39e1180101c4c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 14:18:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bab3cc089e66a7542be86834918a646d733bb584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 19:46:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1f40fcf505c41c8be129e680577c47e : SUCCESS in 9m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96c5a174917e4535a7a04bfa65e1ab74 : SUCCESS in 10m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3c522647854a079390a480fcc19613 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f56f5171d45141c3a5e77bfc676bdcce : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dba2f2ed08f64bfe8ec0ca52868c683b : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29744ac599e34eb6841d54b723cc5b3b : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb649c03ac3d44f1ad5fb3ac2d599ac4 : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5f53a95539034a8095c18e597799187d : SUCCESS in 6m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3c33b12bc40b430a95c33da4b8af0df4 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df6f536fd053478288b95204c906633f : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dbb95bd5f51940718f109df9966e511b : SUCCESS in 2h 01m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a401c7c6613d4df8acfd708372ff8293 : SUCCESS in 2h 02m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/179b3cebf907484495d9ff4b2a0b9405 : SUCCESS in 41m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8d4643c9a0c14dce8d787ecd3188b122 : SUCCESS in 39m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cda5eeb3ab4a4e5abfd84dc932194213 : SUCCESS in 39m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c219d0c9abcb4d09a1e4ac2c59daec4d : FAILURE in 1h 01m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43f89b8eb4eb43e79c2a0086c629b571 : SUCCESS in 1h 45m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38470d8fc34742d78a28dd83c7ad2b43 : FAILURE in 49m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62777c007861485ea336a2b73d72668a : SUCCESS in 2h 15m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"66627a585b3e9580e7f0e9bccee94871e77f81f4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:52:50.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"139093bd97a08edde06a045b058c54448c69ec9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:07:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6be0c6a51bc4ca58978bc1ab4c64e72 : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17bcb8bf456b49c09d68b379dc64bb81 : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e967ee90e18c440286bfd92da006da86 : SUCCESS in 14m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb42ad70405b492da243d72aa01a2312 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/775bda63f2b84149833dbd9c9c0ca38b : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a75b14cf62c4a9eb0acd3a263dde772 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ea25615775c45f2becca39a3f806825 : SUCCESS in 24m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/74b7d4c55609438aa4f148bc4632561d : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e6750811a8254165bbc16e3faa936906 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1afc0bdafec04d68b516cc9d9363aa42 : SUCCESS in 10m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd7fb7e1d982410d880f37bcb57ac7f2 : SUCCESS in 59m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60d94bb6d8864a15bac4e03784e67e17 : SUCCESS in 1h 50m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf23a37b21a147b1a885505e4cde2008 : SUCCESS in 44m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5a9b03e679f64789adeb02a7586094b8 : SUCCESS in 42m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69f744f62dbc46bebc4b22e62b131e38 : SUCCESS in 37m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/557c2967327e4d7e99fc479b3cf64abc : SUCCESS in 1h 15m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4fd9879650f148129fc5495ef18814fc : SUCCESS in 1h 54m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc36e7c0d44541de8508f59a0b1f3e4a : FAILURE in 1h 24m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5042eab5e1394d039b3a25050f2fe006 : SUCCESS in 1h 57m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ffaa9a2413f6725eec39f8bd04a6cfdf0ca891e","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-15 08:48:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"815b820924a5b78803d4d40549380fae5f6d9329","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-27 17:54:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"b5842a68134f52b14995eb80a592da2195838fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 17:54:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cbba572f643011f205ddb4628f5fa8b610f9a34b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 18:33:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07fefd40f54a476f8b59f621cf8e10ac : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d909ac24ebbd494abf0317e90e3b588a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0af3558154943e5a5d662fb2fa13274 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9534964395544accb146ee18f275f7a0 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6bf0e83fc7e47ec89809758cf10b5a0 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/847c6ba44c1f4194affb53a5bc32164d : SUCCESS in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d3d8d6828984d08b79f45860f36e7cd : SUCCESS in 8m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/725c2174df994ae4aefd90fa93bdc069 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0b018fe0e024978a8035bb37aadaec2 : FAILURE in 36m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f971bb31d3094bd8b47fc9ea65d9804e : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/75c321fe716244cb843c1e41ce6f8fac : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3dd589a6989b443baa05ed30b575963a : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/356a961cc0c847b5a94946bc5a6eec66 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"1ce2ce231fee8576748b39ab5f0fa0d8daf36189","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-28 08:34:30.000000000","message":"Patch Set 3:\n\nDepends on https://review.opendev.org/c/openstack/diskimage-builder/+/772254","accounts_in_message":[],"_revision_number":3},{"id":"529010522699def024ec724a46f11e6bcb2554f8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-03 07:44:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"19366a79864a59f59c7b54c54e45129abe832f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 08:42:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/181821e2edb84875939158edca176f52 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57cf9badc0184ef99698ec4d020957b4 : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f10ec0bb7c7d42a5ac27b0e5d3b3fa80 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4072331bbe704a61bc19dc4d42e6fdd6 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25bb2cc04cac4132819e3f5ee2a0e36b : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce91b79b5ad4e2db99a65c35e0097ef : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8fb4259ae74419da44c4aab4cd4653c : SUCCESS in 4m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/82c38b5734ec45878ec77790be7c4d8a : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8e20011f2b14309910067462fe71c9a : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e27f36cbdcde4812a9b4eeddb4242ef5 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bca61d065ade4b3bb25d34586e1fdf86 : SUCCESS in 56m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d4a897c012e4f7a8cd13d16fdb114f8 : FAILURE in 34m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cdce4a59d1c14d94bc4ccce05f506b08 : FAILURE in 39m 21s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d4b964ea9d4a4526b34bc7df6d2e984e : FAILURE in 42m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/be5158d912d6421895f895936d5dc16d : FAILURE in 26m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e9895db038840fdb2805c8cb9f8acbc : FAILURE in 21m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d33557e0c9d4686a293758e886706ea : FAILURE in 28m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56aef5e283ae4f96be01072c8cae55da : FAILURE in 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/74e187d74a8248e382ae78d401485d2a : FAILURE in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7cf26f2a67905462761daffbb29695579784ce4f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-03 08:50:59.000000000","message":"Patch Set 3:\n\nhttps://releases.openstack.org is super slow (HTTP 443), CI jobs timed out","accounts_in_message":[],"_revision_number":3},{"id":"c00b5c4a7234ddea75fb98a5fdc10d1a3559506c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-09 13:03:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a6f9f2ffb1b6ad2e5cab945ad53685b3aadf1f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:16:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf4ffd80fbc443ca9387226277b2b545 : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/700f3d371f58480d8de489fe3a351352 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e7b4f217664be398e07d61a14aa481 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0ab56a8f86641c3b748e1afcf145411 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7367267ded91436187ef5bf0f2208590 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73bcedac6b9743e8b45cb4251610dca3 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/daaaa71a969e4f2bbcd4559a3b15eaac : SUCCESS in 5m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/115d2ffaa29748a2bd75bc99ae64c372 : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c12494b24acf4a52a7683fbd219253a0 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a176ef45a1141d297951c0d47098182 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6886bb3a1034d1f8a09c894424ed917 : SUCCESS in 1h 14m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e03b38498e5b4a78aa90c48fcce02677 : SUCCESS in 2h 21m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82d2eece6a7d433280cd9f6b68e93811 : SUCCESS in 40m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8e6c18fea8d7477c8bb494dfd5cf16d1 : SUCCESS in 45m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/24de25ca8f7940a79c91eaf82139d09c : SUCCESS in 41m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bb036eb3a5042d7bea0249746e060b4 : FAILURE in 35m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3873f0ae21fe4e559a7f7f2539605d1d : SUCCESS in 1h 58m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/44535a282d0d4fd09c622a9a897bca6f : FAILURE in 40m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c736074929274f59b39e6da81ee900d8 : SUCCESS in 1h 40m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31a3bc18f0b826942fc3949b48c34ba0563d76fc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-22 10:10:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8c896e498771ed02da61453f88bba498c73b351c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:02:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/245241b5d47042dda3d74fbe44d0fb67 : SUCCESS in 10m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39fa61c8c8604567bb8e2086e6e113da : FAILURE in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce98ea901f9b4d86bca90fd4e0b3ed64 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86c11e9ca43144d9aedf5b2cd99ab994 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10bb12216ef04adfab8b44dfa5d9d17e : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e05d565665f4683adc848cf172283c5 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b7aa080b8f484f9d18c1cfbe6ac0fe : SUCCESS in 5m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1e19e0946d47408298d6b23276193285 : SUCCESS in 6m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/552cc2a6847b4310a00a135af692efd2 : SUCCESS in 9m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd9addba4d6f48f6938e3fb5e99c9002 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb5cafe088064e5a8bb3943c0d046176 : SUCCESS in 1h 22m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3340b0c77d634eb48905c5fbe806c2ba : FAILURE in 2h 21m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/52a4932317204628a0b2043e13bd361b : SUCCESS in 59m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/04519504f19c485fa230b62289a18bc7 : SUCCESS in 39m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b86c9e07aa7a4a78b1b8481d292bcca6 : SUCCESS in 57m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87ce31f49e004786984b95e3dab87ff7 : FAILURE in 56m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3004c2bae4244c3db08027acf3ace0ed : SUCCESS in 1h 44m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/99aa5b6dbc574b0ab1667568f9bb6be6 : FAILURE in 51m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9db5c637ebb449cda41ce7ff16cb58a2 : SUCCESS in 1h 46m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37bc3872fe20538b7918187a63e1bab56f78d5bf","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-03-01 09:51:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5ae5150afe831ce20dc569c2ae4753c9afd8d5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:33:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7330f1ceb42448aa6a0483531cd5880 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28a48322e90147948a96b6d3c1a93bf4 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bfaa9987554b03bf1ff78ac0a959c2 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dea46807e8b4a41801bf9f7688cf079 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13a5f0488e294f4095b469059e027b34 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e557995f67174797a56d387507fd0d54 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be9a4b99ba7f4bed96e9ce39aeb9d8c6 : SUCCESS in 5m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7f2e476ce6cc49d1b6c23a95b1e5e85f : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/41877015b16e4864b81e09e6e6e07b38 : SUCCESS in 8m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db1ff29d01fd44e9a0b68a34d74021d4 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d0636f41812e446d9e448bce8a2216d6 : SUCCESS in 1h 32m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/664b40258ca24b90b87e5febd8625ce4 : SUCCESS in 1h 28m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/759000af53974d208452f68509f95b30 : SUCCESS in 42m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/76690ad376eb4f40bdbb0a2357e24177 : SUCCESS in 43m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88011ae6824e411e993399aa4a4d1e97 : SUCCESS in 41m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4b784ae9522409da178d887ca686b06 : SUCCESS in 53m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0bc206f776294a58ba1b4466915157f4 : SUCCESS in 2h 12m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6bf5baa9b91e4d7c8aaf53b643a3f154 : FAILURE in 45m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67f63d6cf4d64804b7346935520f0108 : SUCCESS in 1h 51m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"99e61b7899954d6ba553c3a38737656e68018b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 12:33:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"89c2bfd3c1ff13344704173347152b6875ea438e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:54:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7e315f7bcec48df9b5519bf0ce86a4d : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3716e7ae7c4625b924eb9aac30f61f : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79e1f4482eb84d179cdf00cf8a75f3f4 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/712999c4875b46f9a45463f2a2b60853 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70e6096e65754cd0bbf24114a512c497 : SUCCESS in 17m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f33c5e1c49f404b8fd698c2691f103a : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d48ce9c85d94da9866b9603903e7feb : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/955530690d1443db88dbe08e6287a548 : SUCCESS in 1h 29m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc2b7328340344ce95251695995c2d02 : SUCCESS in 2h 17m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bff6debfdd8942d6a9f6f51d3b092a00 : SUCCESS in 38m 18s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/acdce9e1005943549254da3b0d8e0329 : SUCCESS in 38m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18f159b148fc4b759d61faf0f6219d5b : SUCCESS in 37m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a84296013e574248b215616037ac4d3d : SUCCESS in 1h 07m 54s","accounts_in_message":[],"_revision_number":3},{"id":"28f19af2fc1889cfd838f49f1645f4fd0fb29337","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:54:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0c0cae554190d320de56858a3aadce37fca4df8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:56:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1bdce699ca446f955124d857b4d8e3 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1defb4c15b14fef9cb217c396de2cd8 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a053d7c3d9519d71b38492ddd09343efc03c6b1c","revisions":{"fa639326f9896819725a715653b3ccd279c1c64b":{"kind":"REWORK","_number":1,"created":"2020-09-24 14:11:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/07/754107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/754107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/754107/1"}}},"commit":{"parents":[{"commit":"cbc7598f4d058c00e9f55711067aab12dc08a2cb","subject":"Merge \"Fix memory consumption issues with default connection_limit\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cbc7598f4d058c00e9f55711067aab12dc08a2cb"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-16 12:21:18.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 14:10:35.000000000","tz":120},"subject":"Fix AttributeError on TLS-enabled pool provisioning","message":"Fix AttributeError on TLS-enabled pool provisioning\n\nThis patch fixes an hidden AttributeError exception in the HAProxy\ndriver when adding TLS-enabled pools to listeners.\n\nTask: 40895\nStory: 2008150\n\nChange-Id: If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0\n(cherry picked from commit a6a4001f5739e124a0ea9abb0f064a89e4ba5aae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa639326f9896819725a715653b3ccd279c1c64b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa639326f9896819725a715653b3ccd279c1c64b"}]},"branch":"refs/heads/stable/ussuri"},"fa7772e3b9958951bf20980927d37e91b17dcfd8":{"kind":"REWORK","_number":2,"created":"2020-11-23 08:01:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/07/754107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/754107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/754107/2"}}},"commit":{"parents":[{"commit":"214c386277254131b6174f5822437c0374be1d1f","subject":"Merge \"Remove unnecessary joinedload\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/214c386277254131b6174f5822437c0374be1d1f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-16 12:21:18.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-11-23 08:00:53.000000000","tz":60},"subject":"Fix AttributeError on TLS-enabled pool provisioning","message":"Fix AttributeError on TLS-enabled pool provisioning\n\nThis patch fixes an hidden AttributeError exception in the HAProxy\ndriver when adding TLS-enabled pools to listeners.\n\nTask: 40895\nStory: 2008150\n\nChange-Id: If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0\n(cherry picked from commit a6a4001f5739e124a0ea9abb0f064a89e4ba5aae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7772e3b9958951bf20980927d37e91b17dcfd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7772e3b9958951bf20980927d37e91b17dcfd8"}]},"branch":"refs/heads/stable/ussuri"},"a053d7c3d9519d71b38492ddd09343efc03c6b1c":{"kind":"REWORK","_number":3,"created":"2021-01-14 14:18:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/754107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/754107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/754107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/754107/3"}}},"commit":{"parents":[{"commit":"214c386277254131b6174f5822437c0374be1d1f","subject":"Merge \"Remove unnecessary joinedload\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/214c386277254131b6174f5822437c0374be1d1f"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-16 12:21:18.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-14 14:17:17.000000000","tz":60},"subject":"Fix AttributeError on TLS-enabled pool provisioning","message":"Fix AttributeError on TLS-enabled pool provisioning\n\nThis patch fixes an hidden AttributeError exception in the HAProxy\ndriver when adding TLS-enabled pools to listeners.\n\nTask: 40895\nStory: 2008150\n\nChange-Id: If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0\n(cherry picked from commit a6a4001f5739e124a0ea9abb0f064a89e4ba5aae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a053d7c3d9519d71b38492ddd09343efc03c6b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a053d7c3d9519d71b38492ddd09343efc03c6b1c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
