)]}'
{"id":"openstack%2Foctavia~754108","triplet_id":"openstack%2Foctavia~stable%2Fussuri~I7e2b12f19ee7b172464e809df6342ade1e87fb2e","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"I7e2b12f19ee7b172464e809df6342ade1e87fb2e","subject":"Fix backend certificate file paths","status":"MERGED","created":"2020-09-24 14:11:25.000000000","updated":"2021-03-04 02:52:13.000000000","submitted":"2021-03-04 02:49:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754108","meta_rev_id":"83c497ebbd14569a668e4172b884cbd92dd891d4","_number":754108,"virtual_id_number":754108,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 02:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-03-04 02:52:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-03-04 02:52:13.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-03-04 00:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 02:52:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-03-04 00:45:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-24 16:49:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 09:27:58.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-26 15:01:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2021-01-14 21:51:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-04 00:45:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"eaf9c9552a2c2746a00c7a1210615970f2609d85","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-24 14:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d6fbbbba013a827e1ad9cbb9862188a146ce368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 16:49:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/546f72e7afec453fb697a1495eab851a : SUCCESS in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b71a74de46344a5a3be8061ad9b5d43 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479883ac68ae4c6e8920dff4ca7198ef : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c88c4c06626b4034babbfdbc106874d1 : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ecb27eeb7ff4953836a96fdc06ed415 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62d06ab23044a9387d2d77f2cb43cca : SUCCESS in 17m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05c7bb51986a4ac0ac59b059a9f7f617 : SUCCESS in 6m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3d1ea6ab966b429096db53993a34e906 : SUCCESS in 7m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f8a137159a114d929bfb2a2d9741c87d : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef725422a46b46bf8ccdc272d8218754 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e559f2e55902455a9ef029bf09594f9b : SUCCESS in 1h 37m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d293a95c7b84e7d927c242cf077bcb7 : SUCCESS in 2h 18m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4135577ac7c4a989deab8c65ec669c3 : SUCCESS in 41m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f5fef4c30c614478aaf9f9b7662dbc4b : SUCCESS in 38m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/072083faca5247019df98e199b7cf001 : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/164dfd17f3b44a8696b45ccbb5b88ba0 : SUCCESS in 1h 24m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/396fdb57e89b4f9b9c4f705e05e4e849 : SUCCESS in 2h 15m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d46c3e52c36c416fa7a08fd8891cefb2 : FAILURE in 44m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d63ed9834a744d2fb7529caad26534f9 : SUCCESS in 1h 57m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"108d8d5c6ab0e054880820d55966d97387cf9475","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-09-25 09:27:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cb876128a9460ac5f2bb86cbda33174e4b967c3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-26 15:01:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e47b8af563ce42dfb0caf92fb118c559bfb7087c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-23 08:01:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f06e97c899a4d789701c54676cc0e9f8efb52636","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 11:44:04.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eee732b82474933852512f002afe5ea : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/287c4b0f74f24a158882c0e7b9876a7e : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be699e65dd14dd4b01de584312cd939 : FAILURE in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40341c1fdcc42c2a9c82b87cc96911d : SUCCESS in 9m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92db0d3821ac40e08ce75aad65926a6e : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ffeb8271e44883ad657cbb920c657f : SUCCESS in 26m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94183f89ca5c4716961e0efd7ab1836b : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8da39aca8b8e4dcda6d5222f804f9ae3 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/818d84d3a6984a6997f7ba8d55acd655 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44969617754f44289c756904c8973a99 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/434115444c3e4abf88283f4673593f20 : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0557eefdd18e4723be89e111e9d7e458 : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/693fc6823ab245449cc468d0de50ef68 : SUCCESS in 55m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b82aa1aed3be428ab3bf9beaa235b770 : SUCCESS in 35m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0db8c679657349ac9d8435fd5ae4acb4 : SUCCESS in 46m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68f3a1f70e0b4230ae080fc11cd06b82 : FAILURE in 1h 11m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c5e3ad4e051441c88569e5456a67992 : SUCCESS in 2h 00m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8849f9d61a4b410d9840f558482b31f1 : FAILURE in 52m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f14a213ffb4942bf87ad6539730fb596 : SUCCESS in 2h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2dcdc838d4e35443eefff3221548c968a89e18a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 11:44:43.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6eee732b82474933852512f002afe5ea : SUCCESS in 9m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/287c4b0f74f24a158882c0e7b9876a7e : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be699e65dd14dd4b01de584312cd939 : FAILURE in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40341c1fdcc42c2a9c82b87cc96911d : SUCCESS in 9m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92db0d3821ac40e08ce75aad65926a6e : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ffeb8271e44883ad657cbb920c657f : SUCCESS in 26m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94183f89ca5c4716961e0efd7ab1836b : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8da39aca8b8e4dcda6d5222f804f9ae3 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/818d84d3a6984a6997f7ba8d55acd655 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44969617754f44289c756904c8973a99 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/434115444c3e4abf88283f4673593f20 : SUCCESS in 1h 23m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0557eefdd18e4723be89e111e9d7e458 : SUCCESS in 1h 50m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/693fc6823ab245449cc468d0de50ef68 : SUCCESS in 55m 53s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b82aa1aed3be428ab3bf9beaa235b770 : SUCCESS in 35m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0db8c679657349ac9d8435fd5ae4acb4 : SUCCESS in 46m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68f3a1f70e0b4230ae080fc11cd06b82 : FAILURE in 1h 11m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c5e3ad4e051441c88569e5456a67992 : SUCCESS in 2h 00m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8849f9d61a4b410d9840f558482b31f1 : FAILURE in 52m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f14a213ffb4942bf87ad6539730fb596 : SUCCESS in 2h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6d869fdbf803d7b13d05758d26e0f88537727b0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 09:21:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"16dea0f8e93bad5922d9407e3df510ed3e29bc94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-14 15:01:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a2b47126f4d663699ea0c1b04abe4b9632d6f40d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 20:07:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41ad889dbacb44e9951722ea05b65f04 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d926248f2d495d9d035b50e2a96c64 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dbaa9c9eb14aef9f3b0e78ba31c0be : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f48755a22664440699e6688e6c2c5e34 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d0a3469cfee4d1a8d20332e268922c8 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0b83ebd45c486ba5cec060ef7507cb : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54a9ff8a86cf4ae99e7434e3099dbf58 : SUCCESS in 4m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71f9e62e65b14d8fbb15aeb5d476aede : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1d063437751543c3857ddc98f7d6cf59 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aca5f65490c84593bd23df5724b2c092 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2dce209b9434b598d4826f9bffb43ec : SUCCESS in 1h 14m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8847b382101f41c8957759d64c88c577 : TIMED_OUT in 2h 34m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c3f0c169e124fb0974befc8955b2bb0 : SUCCESS in 43m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8355b9f46a88478eaf99d663c18b7348 : SUCCESS in 1h 15m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33f64dd24af046f08fa842c5f212d56d : SUCCESS in 42m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3af7f36bb984d02a9b09fea2c1aa03e : SUCCESS in 1h 15m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cecc2893bd8c4b47a8bc47100316efeb : FAILURE in 2h 13m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b21002165167457097b7327371784b4e : FAILURE in 48m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3fa63abc5b8468a8275848d68a4bfab : SUCCESS in 1h 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0af0d9a4d048829724717c0f68f5d7ee769b02e5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 21:51:54.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d35a052775bd28a5272ff739251694133c82e466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 00:37:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e89a7ba980a64f43864db06ac173e9e1 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4faec075099e4252a6c9362b89f35bda : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14a8751ad92461b9238367bee21586b : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc09ec94776462d8d4062df18783705 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fda8e3c2522b4ceead67eb8fae120821 : SUCCESS in 21m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3d9288b2624546b4c9617df4015115 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1795c41dac51423fa1824d6c11eec20e : SUCCESS in 5m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d2dbf833fa8147e2811873bfe77b2f11 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/387b5b724a2a4b18993e372c204f1885 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c09aec3dadd42829864897c0fd53bc2 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fecfb0a852b4afb982e68d24da59148 : SUCCESS in 1h 21m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/805ffd429b344c7babcbec97d3a0ddb6 : TIMED_OUT in 2h 33m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a18454ab0ecd42bbb87614703a0ca859 : SUCCESS in 42m 57s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d38b9a038ae644668a743edd1ed5f1e5 : SUCCESS in 41m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/20cc809caf424be4ba1cfa4dc76de4e9 : SUCCESS in 45m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f59ca660a67f49b0a250212e6342c1c3 : SUCCESS in 1h 19m 50s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2232c752b0984fb985b6b168da9167f6 : SUCCESS in 1h 55m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f344b505a5f64611a10ffb8d07e8218d : FAILURE in 47m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0b6e8b5d9aab4d4c9cb2e178aceb60e6 : SUCCESS in 2h 06m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"73dbff342fa8c23403c2a0279be9dd0f31c2619d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-15 01:45:36.000000000","message":"Patch Set 3:\n\nrecheck TIMED_OUT","accounts_in_message":[],"_revision_number":3},{"id":"d0533bbc0e99db63441516fe995af35597cab7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 03:46:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb30902c8bb34436956a8ef98ffcd12d : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c03d1390924a6aa31df2c969a70a85 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a76c898552481c8533e985ddd85b2f : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee24e9854ed54dae8154eb6362a794ea : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bde50661bbb84150ac9eae5f63f92add : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19af7daf338e49cda7fe60febcbb2d16 : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1c12c40aa6f488686e9e807c8fe51dd : SUCCESS in 4m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/438b13d9471e412d82b9f620dad4fe2a : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/42c91ce3b1e846278f4b25fa9a2cbeb5 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68f7dc019de943cb9c9680367de4dd7f : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7479d142f1db4f7dbd6fac15b0a246a0 : SUCCESS in 1h 17m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3983caf1209442a6abbb1d4b6ddb6f00 : SUCCESS in 1h 53m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac864f80c3fd4a219e88713f4bbe8689 : SUCCESS in 41m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f69e08f62df74b72889b7251b5f9d935 : SUCCESS in 35m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c8790f1b1d14adf96b2c28eb5d6d1c2 : SUCCESS in 53m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6a2dd494c1948c7a9c9f3e97a9a02c2 : SUCCESS in 54m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d199ac49e024c7b89c756d5432620c8 : FAILURE in 9m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7df848f66cf43619af5b05f85882031 : FAILURE in 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1bfb4a496df246d88ccce8c25bd442ef : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3dd99a9467cc7633941003637d1ffc4101e025b2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-01-15 08:48:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d2024e37799411c2e0df041c675d50120fc9e638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 17:54:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ec405a76957faa4ae651b486db4df57750cb9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 18:33:58.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"733412c44e5db5542b29d3a7ef02388c43e8ea40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-03 16:48:40.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7c0598171c502b9e4720e40bfefa4c905e523206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 19:15:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abbefdff14464459b0625134c9279586 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09765373af054a40b162e9c8a6329027 : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cad702986b84afea25349d6fe8d6603 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/007e920ba48546ae8d4986159af460a3 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e13872daced40f8a8ef3c6ace1a976a : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d57fcc4b6674f42b6746e01430bdc68 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9c61da8f0164af3bf9780edf4abeac6 : SUCCESS in 7m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/04dffe7a9c1c4278a9360906034b0134 : SUCCESS in 6m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6ee91aab533f412b8ad7141ffefe9e37 : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/212e0e7a8d8f4351ab178ebaf9090611 : SUCCESS in 9m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/898558057907467b99a7ee18a4e186d0 : SUCCESS in 1h 17m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e9607da2b0b4615a981d70b65708e9b : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/239e983d4cf34ace91946c5d49ce0df5 : SUCCESS in 41m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0e60bae33a6a42d497088626797795fc : SUCCESS in 46m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/532b40026eda4ad9920efe649e49a1c2 : SUCCESS in 47m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5f70c416c67e464ca40041fd23af85bd : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/234b9c3d4f17448194909b944e53f35f : SUCCESS in 1h 49m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/804657c634214fa8abbd97ad29f52460 : FAILURE in 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/90669ee32ca14a159a67e991ac23bfe1 : SUCCESS in 1h 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"16c5cdd5d39bd77380869c87ce748ffb6add78be","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-04 00:45:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"0269422895cd8d7083f312978dcc82e2000504e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5fe87fb8ea573d8f35d4c6df8ad50e64d7f21cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:49:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ba8fa51c2ee4c238e2c9a4e1fb46141 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30194f12e75e4330a08729c4527a8455 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/024196fe394642bf9d081176c47dd937 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/269e3765579942fd8471fa1bd1382491 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c5e4f7ddb74087beafd37ba5f16b84 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dabfa01f61184987b6ae45eaf308005f : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcecf5c6e52645ada4edff9e0ef90443 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/09176de18849412ca6a73e073e7a839b : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/02fbef6ea17b4cfea741f1a4d9d94e87 : SUCCESS in 1h 44m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/808be54a94154019b30b2c252c659d64 : SUCCESS in 33m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9994bdd45dee4d479d6b53647bf2dd0b : SUCCESS in 36m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8139f4b113e14148a94158f798acfc1c : SUCCESS in 32m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea67688652bb43198187311ec92540a3 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":4},{"id":"a866c1bf7ef5d92ba582a78b71c01de526e3132c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:49:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c351f8cfa99d9dc5379850be6efdeb16da87a120","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 02:52:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102f688d226a4955a13fe933550a3f1e : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65d5ae84759b4b43a3e6828aac54a6b8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"19fbb64df88e5edbc36e0492569dcc353b0ad3fe","revisions":{"3c86f1f680dc009c839897eb35b660af56c9664d":{"kind":"REWORK","_number":1,"created":"2020-09-24 14:11:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/08/754108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/754108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/754108/1"}}},"commit":{"parents":[{"commit":"fa639326f9896819725a715653b3ccd279c1c64b","subject":"Fix AttributeError on TLS-enabled pool provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa639326f9896819725a715653b3ccd279c1c64b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-17 09:35:50.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-24 14:10:50.000000000","tz":120},"subject":"Fix backend certificate file paths","message":"Fix backend certificate file paths\n\nIn amphora-agent API version 1.0, backend certificate files are uploaded\nto /var/lib/octavia/certs/\u003clb_id\u003e/. However, the \"server\" line in the\nrendered HAProxy configuration sets the certificate paths to\n/var/lib/octavia/certs/\u003clistener_id\u003e/ (amphora-agent version 0.5).\n\nThis renders an invalid HAProxy configuration file and as result HAProxy\nfails to reload with the new configuration, and members marked in ERROR\nprovisioning status.\n\nTask: 40903\nStory: 2008157\n\nChange-Id: I7e2b12f19ee7b172464e809df6342ade1e87fb2e\n(cherry picked from commit 89d5e9fcf6d456c11853e73978cb6c5530f28904)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c86f1f680dc009c839897eb35b660af56c9664d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c86f1f680dc009c839897eb35b660af56c9664d"}]},"branch":"refs/heads/stable/ussuri"},"afd1f22a498be65dc036528f474c4208af55b222":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-23 08:01:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/08/754108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/754108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/754108/2"}}},"commit":{"parents":[{"commit":"fa7772e3b9958951bf20980927d37e91b17dcfd8","subject":"Fix AttributeError on TLS-enabled pool provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7772e3b9958951bf20980927d37e91b17dcfd8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-17 09:35:50.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-11-23 08:01:15.000000000","tz":60},"subject":"Fix backend certificate file paths","message":"Fix backend certificate file paths\n\nIn amphora-agent API version 1.0, backend certificate files are uploaded\nto /var/lib/octavia/certs/\u003clb_id\u003e/. However, the \"server\" line in the\nrendered HAProxy configuration sets the certificate paths to\n/var/lib/octavia/certs/\u003clistener_id\u003e/ (amphora-agent version 0.5).\n\nThis renders an invalid HAProxy configuration file and as result HAProxy\nfails to reload with the new configuration, and members marked in ERROR\nprovisioning status.\n\nTask: 40903\nStory: 2008157\n\nChange-Id: I7e2b12f19ee7b172464e809df6342ade1e87fb2e\n(cherry picked from commit 89d5e9fcf6d456c11853e73978cb6c5530f28904)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afd1f22a498be65dc036528f474c4208af55b222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/afd1f22a498be65dc036528f474c4208af55b222"}]},"branch":"refs/heads/stable/ussuri"},"8d8c7ad527f53540a08fff9be59bd93c6507c6d2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-14 15:01:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/08/754108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/754108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/754108/3"}}},"commit":{"parents":[{"commit":"a053d7c3d9519d71b38492ddd09343efc03c6b1c","subject":"Fix AttributeError on TLS-enabled pool provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a053d7c3d9519d71b38492ddd09343efc03c6b1c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-17 09:35:50.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-14 15:01:49.000000000","tz":0},"subject":"Fix backend certificate file paths","message":"Fix backend certificate file paths\n\nIn amphora-agent API version 1.0, backend certificate files are uploaded\nto /var/lib/octavia/certs/\u003clb_id\u003e/. However, the \"server\" line in the\nrendered HAProxy configuration sets the certificate paths to\n/var/lib/octavia/certs/\u003clistener_id\u003e/ (amphora-agent version 0.5).\n\nThis renders an invalid HAProxy configuration file and as result HAProxy\nfails to reload with the new configuration, and members marked in ERROR\nprovisioning status.\n\nTask: 40903\nStory: 2008157\n\nChange-Id: I7e2b12f19ee7b172464e809df6342ade1e87fb2e\n(cherry picked from commit 89d5e9fcf6d456c11853e73978cb6c5530f28904)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8c7ad527f53540a08fff9be59bd93c6507c6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d8c7ad527f53540a08fff9be59bd93c6507c6d2"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"19fbb64df88e5edbc36e0492569dcc353b0ad3fe":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-03 16:48:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/754108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/08/754108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/08/754108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/08/754108/4"}}},"commit":{"parents":[{"commit":"4b06bd1c4b947e8b63d1d45a6de20661cec32542","subject":"Merge \"Fix AttributeError on TLS-enabled pool provisioning\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b06bd1c4b947e8b63d1d45a6de20661cec32542"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-17 09:35:50.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-03 16:48:40.000000000","tz":0},"subject":"Fix backend certificate file paths","message":"Fix backend certificate file paths\n\nIn amphora-agent API version 1.0, backend certificate files are uploaded\nto /var/lib/octavia/certs/\u003clb_id\u003e/. However, the \"server\" line in the\nrendered HAProxy configuration sets the certificate paths to\n/var/lib/octavia/certs/\u003clistener_id\u003e/ (amphora-agent version 0.5).\n\nThis renders an invalid HAProxy configuration file and as result HAProxy\nfails to reload with the new configuration, and members marked in ERROR\nprovisioning status.\n\nTask: 40903\nStory: 2008157\n\nChange-Id: I7e2b12f19ee7b172464e809df6342ade1e87fb2e\n(cherry picked from commit 89d5e9fcf6d456c11853e73978cb6c5530f28904)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19fbb64df88e5edbc36e0492569dcc353b0ad3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/19fbb64df88e5edbc36e0492569dcc353b0ad3fe"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
