)]}'
{"id":"openstack%2Foctavia~754343","triplet_id":"openstack%2Foctavia~stable%2Fstein~If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0","subject":"Fix AttributeError on TLS-enabled pool provisioning","status":"MERGED","created":"2020-09-25 09:39:30.000000000","updated":"2020-10-27 18:01:07.000000000","submitted":"2020-10-27 17:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754343-1603821572785-ac21ae59","meta_rev_id":"795f0482704cf0cda06ae971ffcad82ab3eba84a","_number":754343,"virtual_id_number":754343,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-27 17:59:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-27 08:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-10-21 18:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-10-27 08:50:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 18:23:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-27 08:50:42.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-27 17:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b7cc59b8e1fb8fe10a3ec0eaafbdd192ac46c73","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-25 09:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f322fa88b81709d85bcea94428624799d775ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 11:32:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b23703823fb145b8b493a04e09b508b0 : SUCCESS in 15m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfeda707b8be4f978b8e55c039aa8015 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74da6de007e7459690327a12ddfb6a99 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84509f230c3b48f2aed00815bce38aa6 : SUCCESS in 6m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d4f5ab34a76c4301b02f2142d607d335 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd1ae69bb974a20829e29cf0218de85 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f69f762507842c48af88a0bc73bf4be : SUCCESS in 24m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c59f132e2ad744b19cfc0db8aab35d2f : SUCCESS in 4m 47s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/26cdb7f906fd4489852a27eedde54ca1 : SUCCESS in 1h 17m 15s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/9d9f11bb7c0847429713e250547b63a8 : SUCCESS in 1h 26m 08s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d783949ccd034351b8c36a280e508229 : SUCCESS in 1h 34m 36s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/d1613c4ffe6c41718c28e054189c8cd2 : SUCCESS in 1h 15m 41s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6a8f7acfc1534b29bd76364bf6c596ad : SUCCESS in 6m 26s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0d2c2950ee2d4dbab552162713982264 : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e809edd6f5546a58941286dd047eb12 : SUCCESS in 9m 15s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad0392c7133b4b4bb3ec3f0225a4397c : SUCCESS in 1h 00m 09s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f8f23a64e2324b76bd54fb88283c6b33 : SUCCESS in 1h 01m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/34d11c7a31dc41d187645ad4e1e80881 : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/360d3e95efb04333b1750cff597628ef : SUCCESS in 1h 39m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5fa635ebc7404824898a53bdc0918cf7 : SUCCESS in 1h 20m 36s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/197f0793760c4c57b68f9f7de293e4f4 : SUCCESS in 1h 34m 25s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/291c855dbd62412a9bba8fcfa6c5d4b1 : FAILURE in 13m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fdc274de78b44d898c0c617eeff0273d : FAILURE in 47m 03s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5528c552c74f452cb67da3df16a6e875 : FAILURE in 50m 00s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3dc3624aa05e43a6965a7198460622f9 : SUCCESS in 1h 15m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/58aff9efc14749ae94528f93a2ff4a9e : SUCCESS in 42m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7c760b491ab9029f238e26e568647e9e6956c88","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:23:56.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b62a1e70e3254fb7f0021713221c80e367d7abee","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-27 08:50:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ab54bb42ba6e1117a480e0ffb4fe4ed8db5c1c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 08:50:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"801b2d380a685325b4ed6d011094029fa07ab093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 17:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"150c14b0dd0559f57e0c1dccf846fccaf9ded407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 17:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bb9bcc3a0624283b534492474b05117 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60bfc4a75734ca6bc13708fd392271b : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5752b24d96b42e5ab48f93bba83afb8 : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4dbc888676f8406ab052f2771c8e81c1 : SUCCESS in 16m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/246b9eb2b75246408455f1a14cdb7544 : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f2c1737a7941d996ae1357d861df8b : SUCCESS in 19m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ad10fed40cd444e8e428176f106f415 : SUCCESS in 28m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/66e5eb2d6752417795adbe0b303e8b49 : SUCCESS in 18m 00s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/5afd1203df6a46ef971ba0433e47ce3a : SUCCESS in 18m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bae72adffe4d419d9298347aa228a0da : SUCCESS in 6m 37s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a58a91d311b04e35bcffbe38847365a3 : SUCCESS in 2h 00m 29s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/e65d48fc887a4e7eb6a5547f52f6a626 : SUCCESS in 1h 16m 51s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/fd4fb21620034a2694adbc2b43e49d6d : SUCCESS in 1h 20m 46s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/56bea5026b4f4c6b8685430427bcdd95 : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd117957019c441c92a99c6c75240a7b : SUCCESS in 1h 20m 26s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9d21346e8a084e409baf0be3774168f6 : SUCCESS in 1h 32m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/465ca6eb90204590804b2324322ee3af : SUCCESS in 1h 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"795f0482704cf0cda06ae971ffcad82ab3eba84a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 18:01:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4778cd88b84deeb05302d1be96501f : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6594af07387b49388f48c4c563ec88f3 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39ae303c19c46c0ed604f418e4659e1441b270f8","revisions":{"39ae303c19c46c0ed604f418e4659e1441b270f8":{"kind":"REWORK","_number":1,"created":"2020-09-25 09:39:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/43/754343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/43/754343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/754343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/754343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/43/754343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/43/754343/1"}}},"commit":{"parents":[{"commit":"0a6011f7784303f7b1d2f55ca801f2e27d146419","subject":"Fixes API list handling of unscoped tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0a6011f7784303f7b1d2f55ca801f2e27d146419"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-16 12:21:18.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-25 09:38:49.000000000","tz":120},"subject":"Fix AttributeError on TLS-enabled pool provisioning","message":"Fix AttributeError on TLS-enabled pool provisioning\n\nThis patch fixes an hidden AttributeError exception in the HAProxy\ndriver when adding TLS-enabled pools to listeners.\n\nTask: 40895\nStory: 2008150\n\nChange-Id: If165e995a8b61d8a8ed6bc0c4dd036af0c55c6e0\n(cherry picked from commit a6a4001f5739e124a0ea9abb0f064a89e4ba5aae)\n(cherry picked from commit fa639326f9896819725a715653b3ccd279c1c64b)\n(cherry picked from commit 06fb4412ce6151dff49682d54f893897304566ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39ae303c19c46c0ed604f418e4659e1441b270f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39ae303c19c46c0ed604f418e4659e1441b270f8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
