)]}'
{"id":"openstack%2Foctavia~754344","triplet_id":"openstack%2Foctavia~stable%2Fstein~I7e2b12f19ee7b172464e809df6342ade1e87fb2e","project":"openstack/octavia","branch":"stable/stein","hashtags":[],"change_id":"I7e2b12f19ee7b172464e809df6342ade1e87fb2e","subject":"Fix backend certificate file paths","status":"MERGED","created":"2020-09-25 09:39:30.000000000","updated":"2020-10-29 12:51:58.000000000","submitted":"2020-10-29 12:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754344-1603975766106-fdde6bbe","meta_rev_id":"5121fd7c8ebb7aa9889dd22f33d141a2e231817a","_number":754344,"virtual_id_number":754344,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 12:49:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-10-21 18:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2020-10-26 15:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-10-27 08:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 18:22:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-10-26 15:01:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-27 08:50:33.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-29 12:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"421a763621dd509270e7215e8b4bce8cc66c2983","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-25 09:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a022a3406504690ccee06fe2206b8df7a2a7d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 11:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4bd61fe778549d999596966d5232353 : SUCCESS in 9m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/176603281336420996f8f042ee87e7a9 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fdc13717f345eb8ffff47280a53a5a : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0bc287ae9a2d45879c6ec25ce9a23636 : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/403d7ad1d3a64b62a25701c76c8a8845 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0121f732fd88433090c437637a377885 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4479ae57ed3a482391633437a08fbde1 : SUCCESS in 33m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b175bf2fea04dd8846f4db711dd8289 : SUCCESS in 5m 52s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b682a4b82aab4501823dc7e97ed4facd : SUCCESS in 1h 09m 34s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/55e70af5f84948b89244e67cdff2085e : SUCCESS in 1h 21m 34s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c5f80ebed9943a5890b395ab57148bc : SUCCESS in 1h 30m 52s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/644538f17d2c44f28ab52655e7566edc : SUCCESS in 1h 08m 31s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dfcdaa13201c41c4b83d1c8b066e8abd : SUCCESS in 7m 43s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e0c542447b8e4198822a27ceb31792d1 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cd5885724c643758ddf8f669280693a : SUCCESS in 7m 55s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f009547e34d1472a84fa9d394e07fdfc : SUCCESS in 57m 30s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/3f405f1322174dfda23e217907563ab0 : SUCCESS in 58m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d3e9fe0a7ccd4537ad193583bf8f6653 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0f3af4a224864c6c8cfdc4bcc2c49bc8 : SUCCESS in 1h 25m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3725b36f2e2446c3a8a6c5042364d32a : SUCCESS in 1h 38m 49s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/30c3e3bffaab4884b65cb6eb2a78080d : SUCCESS in 1h 43m 27s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d2cd7e3868424a428bbdebb6a4271e2f : FAILURE in 12m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5639f0e229d49c7a8e42b166b5b4f17 : FAILURE in 48m 19s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/6982575957614fb4b312714fd5ee199e : FAILURE in 43m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/68c70743664149d3a60fb5f0bd6c4bba : SUCCESS in 1h 15m 45s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82ccf0742a004c3e8a4ca57d8bed5f0b : SUCCESS in 40m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a539b1e17c5e808f0b149b7fa5a830613900bf7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-21 18:22:43.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"df71d9ae0c869eacfb28759f1a82ea9c9b6688fd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-26 15:01:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a704564c10967b216127859badbcefe2ae40160","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-27 08:50:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a22ed8cb50caee6396fe4a5a217a70a25eb63b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 08:51:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79b89d05196cd5beba0e3e0e287eadfa51b45c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 17:59:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6505fbf0d5664039afe3914fe0471b5d : SUCCESS in 9m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995898e01004447d9a57b5b30f8b408f : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/529876b1a0b64f55a30dd098a6cf206d : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f18e8c50a2948af8363bc02706d70d9 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13763b00ce244e1a88c341984169c92 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2873b2707f314f5a8869bfce2c862b30 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/647c6a8307dc4cb78a6402bc76dea1da : SUCCESS in 4m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1a4b88a3fb7b425da3e510bef7da6dcf : SUCCESS in 5m 42s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/3ef46e6d62af4c7abe2da63fdfa8cde4 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d856edc84a5248219b28b5001f526cd3 : SUCCESS in 7m 59s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc2f946065794de29616ff4fcf1ef0cd : SUCCESS in 49m 25s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/0c8abc8e948248f3b8fbd35095b60759 : SUCCESS in 1h 01m 28s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/b12e4c4d240a45a78708bab319742491 : SUCCESS in 1h 23m 48s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/d2f44b2fa8b34f04bf51be9d3a7d1708 : SUCCESS in 1h 18m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6035a66b1d944d7ea99c869d75eb26c3 : SUCCESS in 1h 25m 45s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/69318410a46243f09c5f34b6cdc8ac2a : SUCCESS in 1h 14m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba06d7422e36435bb82215ee295b6a96 : FAILURE in 1h 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5470ffadc2f8ff94bcdb2946c084abd535c9e8e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-27 18:04:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e0dac5f65c1429a058e1fa61335213790a7981c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 20:27:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad52dfb848c43a18fc6615b6d943e41 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd8ad2d518994d4d883fbea43e540fed : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a00c46df7f48adb910ffc03e9d3280 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2b28fbdce354150b959779f1b990f05 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fac3be81eb714cc3aa460596d32561e5 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba9c49ef2d84d2f875ef702b1b79162 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b32be3f49443f5ab1c38eb6e14839c : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec782a479eec48798f53f3c49c0b24eb : SUCCESS in 4m 43s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/6933ea9d83734c949e24a6ca18fb43f1 : SUCCESS in 1h 16m 40s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/ec38eb0315374953820e6991782b4659 : SUCCESS in 1h 12m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/080dec62ccf4439db61cd58aff98afd1 : FAILURE in 1h 10m 25s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/f379fd71aa474869b1f9c5acfac4276d : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/828b3110d3a4446eb26875e5861a2eb5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8b8e3862f8994768a692a1292e534830 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19ad7f6abc6b4b35ae3882e80bdceb56 : SUCCESS in 7m 29s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/609ee851e2b84f03ab7a98e4051e74e3 : SUCCESS in 48m 41s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/be17dc96167f410cbd8d7c5b01594802 : SUCCESS in 1h 04m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/088f256e4b334cb5a609def9e26def72 : SUCCESS in 1h 26m 18s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/a435b5ee4c134ba4bc1ad3937c8dfb74 : SUCCESS in 1h 27m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b562d07d8384afcb6740f242a9f105e : SUCCESS in 1h 29m 45s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6e8190fd4b294a029e0619b6007e5917 : FAILURE in 1h 40m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d3217ba1d09646719d81158f0d86f5c1 : FAILURE in 9m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9627c0a2c2434498b92558b7268a4ae2 : FAILURE in 52m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/47917a51bdda469d96de6acfb9113bdf : FAILURE in 46m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eaaba2829e1b42abad5e743f6800b87f : FAILURE in 2h 07m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/207f1aba23014337bd80be57d8fc3f70 : SUCCESS in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48a24ffbdf6493d2c3509ebf84ead9ed7a308014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 20:27:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2955bf5dd2bc1155af59251885d524237758a51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 22:38:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b93451818214ed5954dda2db69662d1 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b288e9cc07cb49b98ec9e2b28a22e9f2 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d465a1784cb2458382f2e15067b56619 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/488a1e1abd774126a07c0d31b0f98a85 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adf8b4b932b54aafb015fd193ecb21e3 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f647838ae640569a99aa1d661ac456 : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/172c6d4322b4454a97057c99c258678b : SUCCESS in 4m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c13da46720847bdaa55562469110509 : SUCCESS in 7m 46s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/6500c23142b14d3f9feb0b6c248edd93 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f93048103c9e475b8f2b81366de4c1af : SUCCESS in 13m 49s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/56784aef5bef499e9dbf1be44cf5574a : FAILURE in 2h 03m 13s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/193312fa545c4bc1b98f1b3ffb3a810c : SUCCESS in 59m 50s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/774e3a9904ab4da9ba52154d429629df : SUCCESS in 1h 13m 14s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/c6ccb484795f43f2bb204a553c000c6c : SUCCESS in 1h 24m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4ffa49087b945208382633203a80664 : SUCCESS in 1h 18m 40s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8648902fabd94837af021d17826bf628 : SUCCESS in 1h 42m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/edc55781088643968ed6e4c925e32256 : SUCCESS in 1h 29m 03s","accounts_in_message":[],"_revision_number":1},{"id":"95a31dd9708e9e3f3c0233125cb4477170d70b57","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-29 08:53:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b65a4c7c8af1cd2af18b57b72946eddd63d39aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 11:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb02dbc559ce4af48645f772a7f562b5 : SUCCESS in 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d66fc5a097b64546abf64b44ac54342e : SUCCESS in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e3496ee89704b10a63bd4113f7d2a72 : SUCCESS in 5m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7987c878a32248b2b7696b4d70c74822 : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee516d8babd04e999c3cb55b256f2494 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7eb7efd1ad284509bebfe1a595b25269 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317221ff2a6f46998993f5f3d97d66ee : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf951945f874b75a29469f28e0090cc : SUCCESS in 5m 12s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/2870b77fb5654358b8e9d3752caa99e5 : SUCCESS in 1h 17m 24s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/5eec1fd99784455ebd2e2cb4bbec90d9 : SUCCESS in 1h 14m 26s\n- neutron-lbaasv2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5269a7e5ae444bd48618c520de42daaf : SUCCESS in 1h 08m 11s (non-voting)\n- neutron-lbaasv2-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/900df668d1ee470bb81222680f949f97 : SUCCESS in 1h 39m 46s (non-voting)\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c85543399813435cb5352b92b055301e : SUCCESS in 7m 07s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/87acb10e39e04e25a93a37f2e9e11d66 : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b09acf6952b6487ea85a6767fc4d2c62 : SUCCESS in 7m 17s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fdf821c0d0a54ad2b84e02eaad7109b0 : SUCCESS in 1h 00m 33s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/b939e464a933438baa4fda11ca775808 : SUCCESS in 1h 16m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2bc114b42ec042a4ae06debf44f0bd07 : SUCCESS in 1h 02m 08s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/ae747d9d4ece425690922376283b6101 : SUCCESS in 1h 19m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2657ed2ab830440d92a624dcfeb6e866 : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/08debef635e249768297440364fe56df : SUCCESS in 1h 31m 27s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/230e418d9ffb43a2bf4678851e4c39c1 : FAILURE in 13m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b92bd608c05d44f390ea45ab31ad449c : FAILURE in 46m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/ed9720669aea40caae18d85ff3309451 : FAILURE in 50m 58s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/20ae00738840410584a5ffd4644ac37f : SUCCESS in 1h 22m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af22637efd614b03b2ed0357495d36ba : SUCCESS in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84cbb275e828ad9b75aa95e3c73a554ca43f1ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 11:03:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3462af3fe483dc30ce7d3801c969fdad661ebb17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:49:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e904bcc3dbb4e9ba9cb2a6245400b6f : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835c47fa2f7c4f20b97bf7c1e3ac2e29 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b739bd00d95490fa5dffabc90183fe3 : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f8bf259a3a94210a3d8e18ee693ed03 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9d149a5b3604879bac29c9e625a3509 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5adcf779c02249459a47db34a832e9ab : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6c2004454ef4abbb3297b87b35d91c1 : SUCCESS in 4m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e3ea41b3ad2e45b9af2385990a836258 : SUCCESS in 7m 14s\n- openstack-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/653cd5abe7eb4fb6836b200605d06834 : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d9e96c8f2d34cc6b47702492767479a : SUCCESS in 5m 22s\n- octavia-v1-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7da98e11e21a4d118910d97f8c7c94e2 : SUCCESS in 49m 47s\n- octavia-v1-dsvm-py3x-scenario https://zuul.opendev.org/t/openstack/build/fa29dba4363f4ffa840d66f9e71dd753 : SUCCESS in 51m 17s\n- neutron-lbaasv2-dsvm-api https://zuul.opendev.org/t/openstack/build/01d8b4429eda4e0387c61d5d93fb254a : SUCCESS in 1h 15m 04s\n- neutron-lbaasv2-dsvm-py3x-api https://zuul.opendev.org/t/openstack/build/b2155af51c564719bd02990557d705bb : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/693e0c66e317467c9ca53502325c7cb7 : SUCCESS in 1h 38m 12s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/aa0d2b9cc7db4cd9951992a3f8ff94e3 : SUCCESS in 1h 19m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/244ee59c7ed84d89994c4de45b243b23 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0f7d36d639ded163c1ee2b2369694e47300ff40a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:49:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5121fd7c8ebb7aa9889dd22f33d141a2e231817a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:51:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa07b2860175499fb5f38378e04b796a : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d77b930144c14009ad51a1b64334462b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4703077d626366f2d018b162cb993fcac170afaa","revisions":{"4703077d626366f2d018b162cb993fcac170afaa":{"kind":"REWORK","_number":1,"created":"2020-09-25 09:39:30.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/44/754344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/754344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/754344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/754344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/754344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/754344/1"}}},"commit":{"parents":[{"commit":"39ae303c19c46c0ed604f418e4659e1441b270f8","subject":"Fix AttributeError on TLS-enabled pool provisioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/39ae303c19c46c0ed604f418e4659e1441b270f8"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-17 09:35:50.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-25 09:39:04.000000000","tz":120},"subject":"Fix backend certificate file paths","message":"Fix backend certificate file paths\n\nIn amphora-agent API version 1.0, backend certificate files are uploaded\nto /var/lib/octavia/certs/\u003clb_id\u003e/. However, the \"server\" line in the\nrendered HAProxy configuration sets the certificate paths to\n/var/lib/octavia/certs/\u003clistener_id\u003e/ (amphora-agent version 0.5).\n\nThis renders an invalid HAProxy configuration file and as result HAProxy\nfails to reload with the new configuration, and members marked in ERROR\nprovisioning status.\n\nTask: 40903\nStory: 2008157\n\nChange-Id: I7e2b12f19ee7b172464e809df6342ade1e87fb2e\n(cherry picked from commit 89d5e9fcf6d456c11853e73978cb6c5530f28904)\n(cherry picked from commit 3c86f1f680dc009c839897eb35b660af56c9664d)\n(cherry picked from commit 539d14c577ba34de7b7bad92f549d91ce623b0b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4703077d626366f2d018b162cb993fcac170afaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4703077d626366f2d018b162cb993fcac170afaa"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
