)]}'
{"id":"openstack%2Foctavia~754605","triplet_id":"openstack%2Foctavia~master~I6c237619cfe641bfe3cfbe4bfe54dc53f3d8d32b","project":"openstack/octavia","branch":"master","topic":"story/2008192","hashtags":[],"change_id":"I6c237619cfe641bfe3cfbe4bfe54dc53f3d8d32b","subject":"Modify named method of NIC in the amphora namespace","status":"ABANDONED","created":"2020-09-27 10:44:45.000000000","updated":"2021-08-23 07:45:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2f570b6ccdc117c3bd82404ce9912ac392475a27","_number":754605,"virtual_id_number":754605,"owner":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-27 13:20:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-27 13:20:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 15:19:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-29 22:21:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"}],"messages":[{"id":"920c9d87e0ee6dbcad44678f08ea52365032a0dc","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2020-09-27 10:44:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c105ef2e6e5a279f4cb9a02c58c7db87471712b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-27 13:20:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/317a9ab9dde44f4bbc358faec39ec7ce : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e38703970594472aad135dd94e71ebc : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ae781e6a4e4f4590b97070e992884c : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f462d023fb314876b201e4cf419fb7df : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea91d56d5da4f0f8ac3e90d902f682f : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8343c6b68eab4af48e4a86f91c06fa1d : SUCCESS in 10m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2f72ce6bbdb14ec6b4dc9dbb0c2a7e6b : SUCCESS in 4m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8a9dafd0219c42659a8974d776214dcd : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9add724438b54cd6b13146baf87bd1ef : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c42d36b9e804c52a237df54b6564361 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e0fc65b09884865ba78d9a16e8e40c2 : SUCCESS in 1h 49m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91daca270a4145a4aeeba4769f19afd2 : SUCCESS in 2h 18m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93195a1bf86743ad8e82abc845e7ce22 : SUCCESS in 49m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c32de95e8a7e49a18060300a0ffe570f : SUCCESS in 35m 14s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de75c0e729784822865f256e824f68f4 : SUCCESS in 46m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/575ba4c9afc742a68d940c5127bca00a : SUCCESS in 56m 20s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a4e401c39cf54446aa8a1572244309b2 : SUCCESS in 2h 04m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/03d4a1718fba449b9fb9e23a0c953f52 : SUCCESS in 2h 26m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3be5dc61659d4a18b0b11534b218a8ff : SUCCESS in 2h 03m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c110fb8d2b8b4dbdb2f04132220a6940 : SUCCESS in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5fe5477b47d1488e9297d5b0c5ef0b35 : SUCCESS in 1h 53m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c3e7a56179c44bd9b2e920566c1bece3 : SUCCESS in 2h 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75cba6227337cb7087956c044ef7d8be8d490971","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:21:27.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":1},{"id":"2f570b6ccdc117c3bd82404ce9912ac392475a27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"date":"2021-08-23 07:45:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c49970f72137a6cbfc4f4c4161f3a57662a606f","revisions":{"3c49970f72137a6cbfc4f4c4161f3a57662a606f":{"kind":"REWORK","_number":1,"created":"2020-09-27 10:44:45.000000000","uploader":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"ref":"refs/changes/05/754605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/754605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/754605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/754605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/754605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/754605/1"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 10:29:34.000000000","tz":480},"committer":{"name":"MengChi","email":"chisimon98k@gmail.com","date":"2020-09-27 10:29:34.000000000","tz":480},"subject":"Modify named method of NIC in the amphora namespace","message":"Modify named method of NIC in the amphora namespace\n\nThe original Octavia codes will name the NIC in the amphora\nnamespace according to the number of current NICs. In some\nscenarios, it will result in conflict.\nThe patch modify the named method of NIC to avoid the issue.\n\nChange-Id: I6c237619cfe641bfe3cfbe4bfe54dc53f3d8d32b\nStory: #2008192\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c49970f72137a6cbfc4f4c4161f3a57662a606f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3c49970f72137a6cbfc4f4c4161f3a57662a606f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
