)]}'
{"id":"openstack%2Foctavia~755371","triplet_id":"openstack%2Foctavia~master~I4ab126f90deef287309ff72715bace77373677b8","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I4ab126f90deef287309ff72715bace77373677b8","subject":"Add resource auditing tool","status":"NEW","created":"2020-09-30 18:57:24.000000000","updated":"2020-10-05 21:40:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"102970ed3c5d2b1b4c139836b5fb8e55dd9cff60","_number":755371,"virtual_id_number":755371,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-30 22:25:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-10-05 21:40:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 22:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-05 21:40:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"7f09c34d27c1ed5d029708e8ce5f38f1b1e937a1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-30 18:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"950dc3e9462bebf322da63b54be9f5adf3ee6cd3","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-09-30 19:57:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76eacb3fedf80a4750c5dd6054b0c0f82f14d0c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 22:25:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d76c4eb74d9d4dc7bc1a6757ac1c4edc : SUCCESS in 12m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/506d48f383674f278192571adc62c7ac : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f7844fa509249eaa0ecce2818a228fe : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/788782dc08fb4382bcc76d66223d89c5 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dcaea8516444398afc7e960189675b5 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35bf1687592450aa46e3fc9e7e16728 : SUCCESS in 11m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1f86614cfe1742dd9fb2ac4e6aac7793 : SUCCESS in 4m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a9a2030824664e5da807a2873a643e87 : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/252c4677d33c4d00bc31c3ada9392294 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f01f460bcdb949b7aa1b47557cd327be : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38154766f3634201822143b892f404ad : SUCCESS in 1h 34m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ad1fd1576fc43dca6e7780202a7e716 : SUCCESS in 1h 39m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d91aa5effccb4183bb9ec349ebc059c1 : SUCCESS in 1h 12m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64ef3aae08c849cd9ba04bb6360ad872 : SUCCESS in 40m 55s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da106e7b51db47298018c5910878e196 : SUCCESS in 42m 33s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4223833bb94d49f68857785becb09537 : SUCCESS in 2h 14m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3f9f27fe82d6462e8bb6d9767a975e4d : SUCCESS in 2h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3e39660106254323b965fba3d0a3b121 : SUCCESS in 1h 31m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3609765cdc5e4f0892f9b836b7148e22 : SUCCESS in 2h 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/631cfe98236e4423b44983b255c3d9ff : SUCCESS in 1h 59m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5fdf6360263b49e1a61164978d3ddb0e : SUCCESS in 2h 11m 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30db436079cfbaf5c2fee2823bb567577db4d569","revisions":{"3ee548580f854d4b1b98cf05676d5da6495d80f5":{"kind":"REWORK","_number":1,"created":"2020-09-30 18:57:24.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/71/755371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/755371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/755371/1"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-30 18:55:33.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-30 18:55:33.000000000","tz":-420},"subject":"Add resource auditing tool","message":"Add resource auditing tool\n\nThis new tool will perform a basic audit of octavia-created resources\nand make recommendations about resources to delete or loadbalancers to\nfailover based on its findings. It can be instructed to automatically\nperform the recommended actions using optional command line flags.\n\nChange-Id: I4ab126f90deef287309ff72715bace77373677b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee548580f854d4b1b98cf05676d5da6495d80f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ee548580f854d4b1b98cf05676d5da6495d80f5"}]},"branch":"refs/heads/master"},"30db436079cfbaf5c2fee2823bb567577db4d569":{"kind":"REWORK","_number":2,"created":"2020-09-30 19:57:44.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/71/755371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/71/755371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/71/755371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/71/755371/2"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-30 18:55:33.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2020-09-30 19:57:33.000000000","tz":-420},"subject":"Add resource auditing tool","message":"Add resource auditing tool\n\nThis new tool will perform a basic audit of octavia-created resources\nand make recommendations about resources to delete or loadbalancers to\nfailover based on its findings. It can be instructed to automatically\nperform the recommended actions using optional command line flags.\n\nChange-Id: I4ab126f90deef287309ff72715bace77373677b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30db436079cfbaf5c2fee2823bb567577db4d569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30db436079cfbaf5c2fee2823bb567577db4d569"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
