)]}'
{"id":"openstack%2Foctavia~755432","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I0e7119d5a379f8662c4913d7232ef28535a29b32","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I0e7119d5a379f8662c4913d7232ef28535a29b32","subject":"Fix python2 AttributeError with strptime","status":"MERGED","created":"2020-10-01 06:45:21.000000000","updated":"2020-10-22 13:37:13.000000000","submitted":"2020-10-22 13:35:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755432-1603373721760-0fb9faad","meta_rev_id":"f9b53258c62b05a81bea6f52a497921c961feac2","_number":755432,"virtual_id_number":755432,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"date":"2020-10-05 14:41:00.000000000","_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 13:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-05 11:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-10-01 06:47:27.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-10-09 11:30:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-10-12 13:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2020-10-01 06:47:31.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-10-01 15:07:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 06:47:27.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-10-01 15:07:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-09 11:30:21.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-12 13:09:46.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2020-10-22 13:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06005024eacc8b7f04b27a8907c795d498604f9f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-01 06:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11f9a0e89bb1cf314855977610e753f10ef6db79","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-01 06:47:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"210e9f94f0b7ac2f53663e8e2faaca752240989d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-01 06:47:31.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"562b42c6bf2b450e008d2fa3f69af118f4da2cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 09:05:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b96288f728444567ad98ab3873d6fa9e : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf315ab883314c5e86e088b1c02f3e2b : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d1f7f7ac454df7a29b8502cfab9acd : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77bbed7f306945d7ab89eb62e19c840b : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e289672194446e7b7e54a5cfe29ce81 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f238fb9fca134d7a9522367f1e18eaef : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644fc33c141e43fe8b14986964aa4624 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6302aa7adf54ab3b6c29de5e7a1b1b8 : SUCCESS in 7m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/431ddc304a2649b8b00c67e82171ebf1 : SUCCESS in 7m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c08c4e849fd4155871783b2aab0efd2 : SUCCESS in 8m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ebb304c66ce447a9b20b8434dc24140e : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1354e21844824abdae428f41a3e0d371 : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fba7d22b7b78488b824d6ff860739235 : FAILURE in 9m 33s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/0c57c538cf84424ab5efbec2c4abe9cb : SUCCESS in 1h 47m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7de487af505400fafdd74dba15b8f2d : FAILURE in 11m 35s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d59edf4c7dd41119440bc1b20fcab04 : FAILURE in 2h 17m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/7f2bbe5126094da3b322f291ff0b192f : FAILURE in 12m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f024143cfbc74e8eac81a6eed8102a1e : FAILURE in 26m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7059fdd366c14898ab4211fa69dd65e8 : FAILURE in 56m 25s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a3c10459f1984860a7b0ffc97a5a98bb : SUCCESS in 1h 27m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a75754e63ad94294826d65ea0cea4a40 : FAILURE in 9m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dc19e02092f34ec799941ccf589cd30e : FAILURE in 10m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e6ebecb51db40b0b766623a6602de69 : FAILURE in 10m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/2cb4520526d54c959e327cbe17fde3a1 : FAILURE in 8m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/3ad1d92a67da40abaf65f7e766e3623a : FAILURE in 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a133ac4b0ea1bb1521e7cd1e9f0dd06b5cc7d34","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-05 11:47:39.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b001e3a58a89a5eb13d69b49a0597b8ece99a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 14:38:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be8cad1d713c47d7adb288cf4ecd7e36 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2629df3f9a684dfabfda767138722689 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391ea2e7eb684c87b59363dda9fb3b71 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb971fddff8e434ea5b18e82eff7b1ba : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3db741431abb4f49bf292329236e3d83 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04206a4b8d7a4bdeb258ddde37095dce : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1907e676a0b34c65bf509d698aaecc22 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19fd086a23b3446d928a6a73ba1d9095 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/86d618dacc3c4f53b35ce94875588ba6 : SUCCESS in 5m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a300d2fea7b441da9c537ba8ed3daff5 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f80f8fde23b148f28c55506f2185596c : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3ba2e1b8c6e4fe59e6d377e39987e57 : SUCCESS in 6m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cab5a2199a41446998c9c229ab90dd55 : SUCCESS in 2h 06m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1520ce7d280f43659bba5a1c2ee7d9e0 : SUCCESS in 1h 40m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dba25689220f45a7bf2df0d0e4a8ba0c : SUCCESS in 1h 50m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/74a62330cb6a4556901b0f16bebcc319 : FAILURE in 2h 33m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/693b3b6030bd414e8bc6c8bfc02c4e49 : FAILURE in 12m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0c2c3dedd11743a6a4595342fe76c186 : FAILURE in 28m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/eea35f3256884863814114655fed1a0a : FAILURE in 46m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9b87659688e42b380ac51f5ac6c6611 : SUCCESS in 1h 27m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/110c3410e43248f983c32e13f7ad5584 : SUCCESS in 42m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6f5669ce0bcc49f59ab7d8bc7d5f7dd8 : SUCCESS in 2h 30m 32s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94e400812a47421889d0c02a82edcaaa : SUCCESS in 57m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/cade05d891bc46ceb6c915d891324f9a : SUCCESS in 57m 53s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/46eecd461c084ab3a7e9129e4828dc68 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c7004cd804cb4cb7388fb32a3515a7d3eb940d1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-05 14:41:00.000000000","message":"Patch Set 1:\n\nrecheck\n\nexceptions.ComputeBuildException","accounts_in_message":[],"_revision_number":1},{"id":"f4e83b31c51a86451cc64019bff44c1c8021c0a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-05 16:59:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4320b201fe04a36ab007da5587ab2b0 : SUCCESS in 12m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8f6616425143d593813e08d4a51a46 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf6d7af84764de4b936404a90c8319b : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ca6af5403764c7db636d82ce3848d8b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f67208994594fee893d600c82f114ad : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bde8dfc9f53448e899ef550d4f6fbaf : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4afb3f380a114c4f9af48ceebf568c1d : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/425ec7fce7c34118804240c2c05aa4e0 : SUCCESS in 5m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/54a4fdaca2ef468e8f82a5b2cd05537c : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6fffc477156b4cda934cc1d9373b696b : SUCCESS in 8m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/39a60c1ff27a4f77a18ec57c38affd75 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce1f6ecd82b145fd94719a0f6c7b31d5 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f778a34c711b42d4b8f81cc3b7d1eaeb : SUCCESS in 1h 36m 52s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/127c646f1e204059b9aa28f7112a35e6 : SUCCESS in 1h 52m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e43fa54f2e6845c5944f2d6b09153591 : SUCCESS in 1h 58m 13s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/00ce6533e69441e49ee317ff91a9058a : FAILURE in 1h 19m 53s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/076aeac5cd1448458abf0790bbcde4a9 : FAILURE in 12m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd4a8019ff7f4c9d8342264a4accaa79 : FAILURE in 33m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2b34bbf4ca3e436b9d82f56f8e73182c : FAILURE in 52m 26s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/92937afb66564aa4aabd2b39e7e57913 : SUCCESS in 1h 21m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c3c8a3246444491b83f90f948ad8223 : SUCCESS in 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bd2e974a79004a689f51d61ef8c5765e : SUCCESS in 1h 54m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed03f7715f044b60b2c30329a7d1a6af : SUCCESS in 43m 13s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0d798a92806540f2a1714216a9c7c7f6 : SUCCESS in 48m 10s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/bb725e100e7c422aa9b93b56faa7ef22 : FAILURE in 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"711e4d696eb3591a02b458f956ed6b9847c23d1f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-09 11:30:21.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4fbe5fc11aa9d3813bf722f9c8986aa6a9225d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 14:34:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71a05191b5344fdeb6cc9673d8a154d6 : SUCCESS in 7m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b272b25690794639bc9d7c87407383c7 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac53f82d38946068653e067f4640441 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6721a0da4eaf4cb1bac77b5176213c70 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a0fd2279cdc4c7bb48a7c99eb66b32a : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/148ccab730774c4d99ea996108ce7b87 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b02db5a643454788516dae5989c50d : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a8d8863ea6458fb0e4af34be5f1492 : SUCCESS in 5m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/58d6a429cc65494c9458a153cf2bfa0c : SUCCESS in 5m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3ac898d8f06746a19d8125f8472a87d9 : SUCCESS in 7m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3233ccba44bd4f16be437e97f0c95524 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c82da329e20d40b98120a2d0064ba65a : SUCCESS in 6m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b9b23ba97d34715beb8b76c5af5712e : SUCCESS in 1h 45m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/99a8c7af10644fcea0e14299f06285a7 : SUCCESS in 1h 59m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2568bdfa4c074a9c910b1e61610dc484 : SUCCESS in 2h 19m 19s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/912819f370cd4788adb50b27060b4f25 : FAILURE in 1h 50m 09s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f9008a017b804f8f9c791f4066fbe2e0 : FAILURE in 13m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7d60a80534ea470c9de3d81cd08d294b : FAILURE in 26m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/d426d9ed6b7e4d1e966d91eaa0244400 : FAILURE in 45m 48s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cff852ba1de4423c92c3f072e688d2bd : FAILURE in 45m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8150b5bdb0714f84865d739465e1bc54 : SUCCESS in 1h 14m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1d908ee473b2466aa7667f0b1fe9ffa2 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/742ab866a27f454998718ac79ac9225f : SUCCESS in 42m 24s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c96ee3e4edac4ce98c3df33994947837 : SUCCESS in 34m 25s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9764ce0ff66a4bc0b9d296c42e8463fa : FAILURE in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26d5832ff876af9f5094ffc3a5b732c1aeb0a701","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-12 08:30:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"655a5c94e4f172d53ebcc0dbf4c9400e2e208214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:05:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9ced97134a74aac9bf965f35ef8a170 : SUCCESS in 8m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee5cbdc7a35046c995b4cd0dd5dab89c : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e1cbc413b46410781a30e749b284ac1 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3eb504cb119c47e1afda01d50e16580a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/967abc742df24a648acad6ee82a32d2f : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ecc37d148410447cb42736210b2ac092 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4436514fcfe84ada910b8591f8edcd59 : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4107d2ccaa9409499c78e70b85c0c18 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a71a042593db46adac225cdc4413e188 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b6916c57dd43430382b6c4c127c8ad24 : SUCCESS in 8m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aac4ba52307e48eaab66ee81859d8e8b : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3d7aebce7f9d4578a01df3edccd4eff1 : SUCCESS in 7m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/329cd16bf7754bfab27c3c99e59395ff : SUCCESS in 1h 44m 24s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/42496349767d4384896b0805b0c05fae : SUCCESS in 1h 46m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcaf867f06bb4217b6f2aa87cdd92e51 : SUCCESS in 2h 13m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/779a7d2c8ab040359b677be96c6a9aab : FAILURE in 1h 59m 24s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0e7e0e71ccbd46c596b9033529b8b550 : FAILURE in 20m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fa944bd7fd334ca1b85ac370336e0725 : FAILURE in 43m 45s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c45df67d93cf448f967d9c987e000976 : FAILURE in 50m 06s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9517fdd0a384c80b787d74902e14219 : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb0c6716b79a469996f3abad86de6fca : SUCCESS in 44m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/429f972388544094b3ddb978975cbb9f : SUCCESS in 2h 06m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc80caa5fb0847a7adec8cbc358f56f5 : SUCCESS in 41m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/fff27f493b0f4577893691655c48c2ef : SUCCESS in 53m 49s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/254a7ee6092c4888b4863ff98d915430 : FAILURE in 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90f17a2ae26b6992824874f7c43bb6701a5886c7","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-12 13:09:46.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"893a2d4f2085a88607946e571d4fa006cafc9e7b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-13 07:26:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be7a4cb352260191e9b888a6f2929f31f10c09ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 10:17:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7af0a13082784003af706c7841bbc089 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87eb9da3000c401caf8d023be101f691 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2c7bbfd08d42a9bc3708009ed99d09 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eefe1e65894a433ebee80b4f29d1f859 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d53f6eb863440fb7d2325f0f1ed6e5 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2daeaeefdc47598ba4b405ee034c38 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f18bd1b16a43f2ab50cbc853cfe796 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fc7cf6d33454a59b05570e63a8165d9 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/da4ce11d23dd46ffa4bd0ca09f6b5585 : SUCCESS in 7m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5fca45b60d4e49f398898ce696e7fa01 : SUCCESS in 8m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4142974c0a0b4145852d82605f11f8ab : SUCCESS in 10m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f52518cc301e48b28a1c4e7b95a5e9cc : SUCCESS in 8m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d379769793a14d04b1f5b1afc5e4559a : SUCCESS in 1h 33m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/15447b687710436aa0eed524f565ff4c : SUCCESS in 1h 40m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/35bc7c155c684aa1827cfcf22b19ac7f : TIMED_OUT in 2h 33m 44s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6d69dc88ff3a42779a6ef9737c3e29d6 : SUCCESS in 1h 59m 41s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/374c6476bd2c4ebc8bd12f75042e8db8 : FAILURE in 12m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b2ad604e411a4a85b8fcbb53f363bf7b : FAILURE in 54m 28s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/682fa787ee2a4e839f67e443799eae57 : FAILURE in 25m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/138e59a526d24714abcff4fa2fd8b971 : SUCCESS in 1h 23m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe5dda65a7c24f1c8e2396ade67e94fa : FAILURE in 24m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdb9b899a01147c09fcab2adc4a6a3de : FAILURE in 23m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1b0ec8f6794e4980952a7574c5ca2b2c : FAILURE in 24m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9b0b8a7fef0d408d92d8f0aa7581c1a7 : FAILURE in 25m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0aa0fd8007a1407e9b085f271abe5d6b : FAILURE in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5cc1029df60ed243bdc071a7da70a2ed0db95796","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 08:32:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e33be1081a817fbb9bc81724a3066c07fccf2783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/805a0eb5839e43fbb3c6caf961a39d3d : SUCCESS in 11m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23756c22833c40dd994e8326459102be : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e79568e253704ec8ab5f3df2166bc33a : SUCCESS in 8m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e302c35e8b83414ab2d6a1702a7109dc : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54c4a438c80947fabff5ff576e531050 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe44419e1b154cf8a9cb9dd59659ebbb : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07643fb214a14a26bc4ca3b15d6cd308 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/760fb12b6765409a9c12b68540366540 : SUCCESS in 6m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3df029165136402a8137a4d0809fad6b : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d48e3b1f14754b0fbb1b2c1e6a238944 : SUCCESS in 9m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e36489503ca44735aaaa551d25cbc16c : SUCCESS in 8m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15416c6f05ce41a59417abe3f0b9de0a : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/20e87ed052e74650abd166f85e650f90 : SUCCESS in 1h 41m 00s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e998a357a0b9420586d26510d6ba171e : SUCCESS in 1h 49m 07s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d18e5c96d8614a5fa16858864c4a6f41 : SUCCESS in 1h 44m 40s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4fad515552494a1c9613b9f5cced6a91 : FAILURE in 1h 27m 39s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9d4ae82b13aa4e35911496d2320ce706 : FAILURE in 12m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6bda7438f4ac4c1da8bdc33dc8302e37 : FAILURE in 30m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f350767dd5214349bb97235ceadcfba2 : FAILURE in 50m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8809297d32c84cd596c525b2fc378129 : SUCCESS in 1h 40m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f98db7e9f70a4637ba9ce7b36558b07f : SUCCESS in 48m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e81d2154c280429db056ee1c3d8aa18c : TIMED_OUT in 2h 34m 56s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8268f746041c43c79f738e49d3670813 : SUCCESS in 40m 28s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/9e037090332842d9b7f7cc7963d96bad : SUCCESS in 47m 04s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/2d5c42c616984f9097b49e5f0a78914a : FAILURE in 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65d5946c15e3e8faf6361ba2971f4d8607903c06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:26:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"05ce02ccb551b3e5919e4ce959d2971cc3fcb5b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 13:35:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bf724e1da704ec9a6ef88cf30c7bcf0 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae7adf53adc4ff9953917ddea7d71bb : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dea39ef35fbc40f0b249fae9b7feb175 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8110f1848424e0ea756090a1b2b7590 : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9c6535e8c8548d288544befe0099284 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14d4b1f503e4b1089fe5cb124323818 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14398f23700a4ec09389d5e6d1faec14 : SUCCESS in 8m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a142e2cae5614043b9a9b1f64eac384f : SUCCESS in 9m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e2f4490dc71488394ad6997d7b2e0d7 : SUCCESS in 10m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1e127d8a75f4fb9b643ab6c4eeea588 : SUCCESS in 1h 40m 56s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/9d489a421b7c4ba2b6f52a3739529f74 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a2842a768fe41458e98b22aa5615355 : SUCCESS in 2h 06m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d714b845dc449fbaa52ff999f3c59a2 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fd72d68da5644c5946be86922ddc12939e7d2662","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 13:35:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9b53258c62b05a81bea6f52a497921c961feac2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 13:37:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea5efb496d8469e84d299d2d416c26c : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8606575aa6f4174993c427383d0fb1d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a69da392f8951fa3716c81f81f1ecaf5dfe3480","revisions":{"5a69da392f8951fa3716c81f81f1ecaf5dfe3480":{"kind":"REWORK","_number":1,"created":"2020-10-01 06:45:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/32/755432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/755432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/755432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/755432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/755432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/755432/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-09-30 19:26:34.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-01 06:17:04.000000000","tz":120},"subject":"Fix python2 AttributeError with strptime","message":"Fix python2 AttributeError with strptime\n\nFix a potential python2 error when calling strptime, the first call to\nstrptime is not thread safe and might trigger an exception.\nA workaround was added by calling strptime before creating a thread but\nthe caller is itself within a thread, so it may fail.\n\nThe workaround is now to explicitly import _strptime as described in the\nupstream python issue https://bugs.python.org/issue7980\n\nNote: This commit only applies to branches that support python2 (initial\ncommit in stable/train)\n\nStory 2008211\nTask 40997\n\nChange-Id: I0e7119d5a379f8662c4913d7232ef28535a29b32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a69da392f8951fa3716c81f81f1ecaf5dfe3480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5a69da392f8951fa3716c81f81f1ecaf5dfe3480"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
