)]}'
{"id":"openstack%2Foctavia~755486","triplet_id":"openstack%2Foctavia~master~I78d0aff8a187ad1dfeac463d1f0a7dea7caf76bf","project":"openstack/octavia","branch":"master","topic":"fix-devstack-tls","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-08-08 07:29:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I78d0aff8a187ad1dfeac463d1f0a7dea7caf76bf","subject":"Fix setting OCTAVIA_PROTOCOL when tls-proxy is enabled","status":"ABANDONED","created":"2020-10-01 10:59:08.000000000","updated":"2023-08-08 07:29:54.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44554757f284ea83b3d6a3e7c525e81e9d46a6ae","_number":755486,"virtual_id_number":755486,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-05 01:02:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2022-09-29 14:46:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-09-05 06:43:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-03-06 09:25:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2023-03-03 14:16:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2023-03-06 09:25:15.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2021-11-15 07:42:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 13:39:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-05 06:43:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-09-29 14:46:02.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-03-01 12:46:57.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9c30e07189509faf2035b9a03ff5fb4d9194dddf","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-10-01 10:59:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69c42866a9fc346aa2a2ac2250c9ab2b48a456a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 13:39:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af99e6836d934362ae3c67c3055bbc54 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eae2c04f7f554e7dbd6108755915e16c : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75d99575294493282386aff5234b444 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9721bcc719cd43339629ba91a0b6dfa9 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd607c56f4be4372b6d679229b0209c2 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f3cfbddb154ffdb19f7fab539e32d9 : SUCCESS in 11m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/070017edcf0d4e389d10244f0ae35eb2 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/53866ea3595140e8bd2ce484d8f19a9f : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c3411d3281f464da23cd4ca7efe15d8 : SUCCESS in 5m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98af7bb5990d41daa27845bfd8f170ff : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87b88d2e733c4649b2829129514d3699 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82067f3908ec4a28948feb1f869bf0c6 : SUCCESS in 2h 16m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d53ba0bf7a649b1b9a52511777ce23c : SUCCESS in 35m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/01f05652745e441698db44c1a749256d : SUCCESS in 45m 45s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f3d1c7bbde441d881305e082d6dffae : SUCCESS in 46m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b57722d2a6ed457682a0fbb09070d7b4 : SUCCESS in 1h 05m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1e72e04343c5401796cb4d84920a9012 : SUCCESS in 2h 15m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d211ee5a098c4c16bf5b78ea3d45c514 : FAILURE in 45m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/311e2057222e4e9ab77fba3a615cfe5d : FAILURE in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/25a9f6b333de49a0b1add59923f15e8e : FAILURE in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/28be44a2d73b4800bf458d05f50713d2 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b678a3e551114e2faeec1fcc091abd6a : TIMED_OUT in 2h 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aae8eff77c263141aa8b722a65edab60b33f19c6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-21 20:33:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bb764959a84b4a664c336a6179645954c52f2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 01:16:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a79189ae7f4646862d93d60a4e5cc4 : SUCCESS in 13m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2b7344805a04c50a62d54c0b240788d : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8fffac76ba45908a3bc53436b8bed9 : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da6d2f05add2417ebfd35c3215ec5930 : FAILURE in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf79942919fb4789a49dc520a8d4037e : FAILURE in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dedb844e6e494e45bcfcabf24a701b21 : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b265b921465a4f0388f2d80550d8cc4e : FAILURE in 8m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/84b9227b34204b72a0b4bc22de0d4be9 : FAILURE in 6m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dcf61505acd4489a928fe39d8488315f : FAILURE in 6m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b9a97673f6744dc7b28f0480ce633811 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19313746fd1c48fdacfe6d20bca50589 : SUCCESS in 13m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec682a7e5b824a298e74eef716bbe83e : SUCCESS in 2h 21m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5eb2d47b1144cbd90e3fc4e124b2370 : SUCCESS in 1h 59m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/00f0f9bc6a844f83be878527c5a326ec : SUCCESS in 43m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ace7163a5624a5b923dbe56300a9a1d : SUCCESS in 52m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d9d66fbecb14075ba85f04bbff3ffb3 : SUCCESS in 40m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8856bcd5361c4a64887b41a4cbca28fc : SUCCESS in 2h 01m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a6556191f3a44d4ba9adb81646b3aa11 : FAILURE in 52m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f5ff006c11cf46f8ae02d2204eed6824 : FAILURE in 2h 03m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bdc43f0c456a48a59d5072c22e16d3a9 : SUCCESS in 1h 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ff605983e6da4972a022ad50a2c966b3 : SUCCESS in 1h 55m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2e5b1a3760f54ae4bc569813a18ed03a : FAILURE in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cb5f54312ec030409f61540bef40f0537630410","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-23 17:05:46.000000000","message":"Patch Set 1:\n\nrecheck tox failures are certainly unrelated","accounts_in_message":[],"_revision_number":1},{"id":"ab9691260181b293e942af064cd2990fd549d41c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 19:03:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51cdeec000644c62aa6dc5a9da4f0824 : SUCCESS in 14m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be9a42fc15fd44c984c44bdf50c972e8 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c55f569453f4bd19ba7a21ca796d540 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da05fd58292f481dbecc38f712e466c1 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ad16769b1ec446d823108d8a15ec7aa : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b42ac2aaf7da40f786a6d54183a1a01e : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed7a2144ee24e498d892774c839947e : SUCCESS in 15m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/447b1c30cd174a42a41d2b4e56cc83b2 : SUCCESS in 8m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c36ccae52060469cac6b32776a269ee2 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dab63c46f17f49829d3e4cec09cf45e1 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fa17a147e6a4e3ca61bada876cfc73d : SUCCESS in 10m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/253654d9296f49eb80bcf3bf891eb531 : FAILURE in 16m 46s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9b8efa0bd8ce441382fd817ab03bed51 : SUCCESS in 1h 26m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f307d5f2edd4e7fb41a97346892ae98 : SUCCESS in 37m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cf1b31ef85b148ce9c73be58fb9c6c1c : SUCCESS in 59m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/99158390e34045a294c5c6dc93ff1325 : SUCCESS in 37m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0646e532369c4cff8f816c68fbdef9d1 : FAILURE in 25m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fef51daf6c16489b9b73439518d0bb6a : FAILURE in 35m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/81c13a9d8eec4797a813460d9f2f1611 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d10b1e54cc8d43719edb0055312f5201 : SUCCESS in 1h 24m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1d16ac0d4396409ea0a3303857b3fd37 : FAILURE in 26m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/777c133d7fc5495cb03959a0f40e6d5f : FAILURE in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd603d0db7ca933dfaf2826a759c0d00e82ee151","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-23 19:30:20.000000000","message":"Patch Set 1:\n\nrecheck another unrelated failure","accounts_in_message":[],"_revision_number":1},{"id":"00faca543361566b8bc21ee1d3f513135971a068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 21:39:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60eba60f27b946d7807baac3ca34121a : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e86e40687aa04c34abd9497ac0c2be54 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/117465d269e64b4eb60ae8746f675a72 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647b74dc99b248a08650ce6596634ce6 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965383f060934288a2b61291152b2972 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69e20e83f2f047cd993327bb7de750c3 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3ef53dc81f4f3da4cbbb81cb6d1781 : SUCCESS in 12m 56s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1186deef52d44fb3a9133b86ee08752b : SUCCESS in 6m 27s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/64018e4fbb2f4337b95e7acf4b48fb0d : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2a4f947507d84c49933c8d2e15f01e05 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/03dd96157a504f23a4cf20ad6bdfbccb : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f6975582af904e4982b13da4d8324a5e : SUCCESS in 2h 05m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a76550656384957b38a051cae320e57 : SUCCESS in 2h 00m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b4d16a98ad549789ac75743a1db2f79 : SUCCESS in 50m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99ee8814adb44b8ab0033be1c3423c55 : SUCCESS in 59m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f73e2ca6daa428282166afea15c5b74 : SUCCESS in 49m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/174295869e8e47afba35422171afeea5 : FAILURE in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9ea3c183134b4b9faaa84dc99bc83bef : FAILURE in 56m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/373d05c292044dd0890f3722e48041d5 : SUCCESS in 1h 37m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b867cba82fbb4a2ea1c66237de810e6e : SUCCESS in 1h 47m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/95e73553ea494a929526e48b93b3ce72 : FAILURE in 1h 50m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6514534864b046fcb4da2cabbaa0d035 : FAILURE in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfd8d33ef9c750518ea2babbaf62e423d4abae30","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-15 07:42:33.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0950f03585f4d757d4fd05854b63872bcb4b2614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 10:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0afb9a21af849a786b58ff067860912 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63f27e4b2d8647bfbfc73d18406cd506 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952529ba9339434f9704eab251b3fed7 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baee6a430af948cdb16554519b615ab9 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5392aa0921824a6083791ff2cb7e7d9d : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d48f3586b98c4a989e7baa65ee3c0c32 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad5d98e5e0245cbb3ced9ffd23ba6c5 : SUCCESS in 12m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/56bc6ae6cf7547c8b14a1b5e01766018 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/864d39a28e9243f0a832742abe037a38 : SUCCESS in 11m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ef458ae47ad4d51a62fd105fce9f63d : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db12d2f2500b4b62bdd6867caf3c7801 : SUCCESS in 9m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cf0c7bb98ba4d009106b47bf9db8477 : SUCCESS in 2h 03m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/92c810899a1f46d6a6fc6f7d9ad38eeb : SUCCESS in 2h 04m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/456cdb42b9784537b80b6ac86050646f : SUCCESS in 51m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6765f4599f9147bfa08d965999cd42a4 : SUCCESS in 1h 19m 17s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f4a2a2cbcdb4cb68a4d2d4bbd8246a3 : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c7a7758f65ce4968935c317109c8d367 : SUCCESS in 2h 11m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/320b658a9f0a467b8ae239c623fd67f2 : FAILURE in 55m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/586e8b0615434dbfa6f2e0ae2d170c0c : SUCCESS in 1h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86863d2606874e93a97ae9ed37a869c1 : TIMED_OUT in 2h 35m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1e731f851f0942489d9149c1d94c3fd2 : FAILURE in 2h 14m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/31fba1148ccf44ebbcebc0871880379e : FAILURE in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"870a8101cd45d7f41cb471e89f6f2652b6a216c1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-09-05 06:43:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5256f076105407bd925679acd05305ec66f89f82","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2022-09-29 14:46:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbdfa63f4eb2752673cc27e65de665ca05a00660","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-01 12:46:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91921691fbb676486a8fd740b3aa63a7898434d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 12:47:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dee1ddaa4343178cd6b9b25290e283063b32288d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 15:23:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8855d4c4f7f548f7bc0daa7634a6ecec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57340be3b5ff446fbb49b46542f2445c : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aac35d446284d9a8521637d36f299d3 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b106fd4dad594a138e8ab5272ffe29cf : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08455eb0785041a589df9f1907bce39a : SUCCESS in 11m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cb2fa1980a64c7388f7c99cbffa10f5 : SUCCESS in 4m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a60e8810d65f48e39907aae578b3ee70 : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/297c3279999846fdbeb3e8d96ac06afb : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2981c28ef1e4483c84bbcdc63d6c2bef : TIMED_OUT in 2h 33m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f09cbe88d6f54c89bf017e54d39a0df6 : SUCCESS in 47m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8dbbfd9fad7a4ff2aa4efe0c78b29663 : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3007abb3b605d755995ed0b34dc73aaf1e31d652","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-03 11:20:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9beea01144bfdeee88f8d999f9fbea873861a37f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:59:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e770ef310dd4efe82176aa743978c0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44c15458a0e34125bf2a2089fe623b35 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a87bf7dbf22b47fab8188674f91d8278 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/051fa248ee40404ba7b0ede13e53993d : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db727f606c8741ab89f6178c1f3596e2 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa32d07c61e4edc873b1e6bc3d7db8c : SUCCESS in 13m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/48805a9411c744efb932a188bac1a130 : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ef8c6f0ef64143319e56e0b2c4724115 : SUCCESS in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2eca389ee3634fac9a347a5c6b1781b5 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/48455aad5d3e4681b29703cd709d5787 : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52f83182875b424aa74906240bfd7ecb : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66468359abc24587a9b91fdf5c8aa454 : SUCCESS in 7m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/febe395c3f98418da4fba64243edea20 : SUCCESS in 1h 25m 51s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/897a9f1d379d42d2abc3280aa75f2ed1 : SUCCESS in 1h 44m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dcb890a84e4040c1977efe55f38a9bc6 : SUCCESS in 34m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/23d4dbc417d24740bf433bf318a09305 : SUCCESS in 38m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/54369c1e4fa0496f9ae7c2f67e9380ab : SUCCESS in 1h 12m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e69d018b55ae4a32a2acb0fc5593d08a : SUCCESS in 32m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3856d0350b946b9a0e67f473e575401 : SUCCESS in 1h 56m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ccf551984b73447bbd6b7cd2ecf2a763 : FAILURE in 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65da690fa6e642d2a173ee94a230aed0 : SUCCESS in 1h 47m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c3e34944d6e941cc8d7980bf3327a603 : SUCCESS in 1h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3b5af3bb6fc347e48182b09222f28f4f : SUCCESS in 1h 55m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1297ca5c0ac4a018b33f71219380ec4 : TIMED_OUT in 2h 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0aa484825973eddbaee3782e2ec892bed49b5dd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 14:00:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2f701d2b29f15a6d7447f8c7b1080d85cd409cd","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-03 14:16:29.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c020be067cd069f29b68a867f0b65f7c196eb029","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 14:16:41.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"3cec5af9f2f397b292399c0b54878953e89fb03f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-03-03 14:17:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04bf8ee2a8f5c7cc1fabad9df3955271c42b5918","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-03 14:20:57.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"38e408fe27692049c0af814ef69a81bcd9f568c0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-04 22:11:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a0f73e52d6069951209e1ab27cd628ef892b04a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-05 01:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d28b357cd3e341bea7d2d030086aefb0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed968eb447da4717b1e8e8a596aa6fa4 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6417d327651f46a48ff34ccfb59f5f48 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/550539b7780b420cb08ffc3e2813ad20 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7919d5052a0e481dabac89137cba0908 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6af55e025c4b868df1eb63706861f5 : SUCCESS in 9m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/252797a252cb4ca88795e0bcc1dedb7e : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/63fed77dc0904c7bb43fe81c61a04990 : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/38569ecb86dc485387f73c4ca91ab4fe : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f4552f5b1c5342d4ade207fdb9869457 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c44a371e798f4b3687a0e924110c1e6f : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6080eb2a2f4242f18cabcd013b615f23 : SUCCESS in 6m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99b539e7c7e74875931934b3ff72b4ee : SUCCESS in 1h 26m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7920981f5386482498ce0b9e3ace58a7 : SUCCESS in 2h 08m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcf236978b1642349f3e26d2f3d4570e : SUCCESS in 45m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a8b28d3697c4b358cc258a70a8a3dbe : SUCCESS in 50m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/60390313e10c43189b01d0beeef433ce : SUCCESS in 1h 13m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca395c7249f14c78824bca91c1ee1e49 : SUCCESS in 44m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f03266ca1fd54dd19e89878ab0182759 : SUCCESS in 2h 16m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c80f6a60cc944afeaafa757491a58ad6 : FAILURE in 46m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92d3cc42a13b493f9d2b9cc3830f52a7 : SUCCESS in 2h 08m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/542f1eea069144d9a365f13955f3907d : SUCCESS in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a659b16a0cd146178bd0c2a07372b24b : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ae946e6a4c94858a72801087cc75a4b : TIMED_OUT in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5160b897dab785c8a674dd416c6b841bd8a5b531","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-03-06 09:25:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44554757f284ea83b3d6a3e7c525e81e9d46a6ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-08 07:29:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68b7aa4ea4f6cc76cf844040c2bad166d1cd1366","revisions":{"68b7aa4ea4f6cc76cf844040c2bad166d1cd1366":{"kind":"REWORK","_number":1,"created":"2020-10-01 10:59:08.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/86/755486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/755486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/755486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/755486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/755486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/755486/1"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-01 10:55:45.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-10-01 10:58:13.000000000","tz":0},"subject":"Fix setting OCTAVIA_PROTOCOL when tls-proxy is enabled","message":"Fix setting OCTAVIA_PROTOCOL when tls-proxy is enabled\n\nThe fix in [0] was incomplete, the global SERVICE_PROTOCOL variable isn\u0027t\nactually set to \"https\" when needed, we need to check whether the tls-proxy\nservice is enabled like all other services do.\n\n[0] I0154b83cb64952844a28895721694d3e2ff82be2\n\nChange-Id: I78d0aff8a187ad1dfeac463d1f0a7dea7caf76bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68b7aa4ea4f6cc76cf844040c2bad166d1cd1366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68b7aa4ea4f6cc76cf844040c2bad166d1cd1366"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
