)]}'
{"id":"openstack%2Foctavia~755828","triplet_id":"openstack%2Foctavia~master~I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","project":"openstack/octavia","branch":"master","topic":"task/41009","hashtags":[],"change_id":"I7488f2fae61917f6d4a56cedd05bace7c5e2bc70","subject":"Support deletion of all tags","status":"MERGED","created":"2020-10-02 15:35:32.000000000","updated":"2021-01-29 05:21:16.000000000","submitted":"2021-01-29 05:19:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755828-task/41009","meta_rev_id":"310252d469ea95d535175dfeb104c85efa3829e8","_number":755828,"virtual_id_number":755828,"owner":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2021-01-29 01:18:18.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-29 05:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-01-28 10:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-18 07:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2021-01-28 10:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-11-18 16:41:28.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 17:23:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-17 20:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 07:49:10.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-11-18 16:41:28.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-01-28 10:54:22.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"b175a47b7ca5aed04f822bd4cb5ddcffb57fc3ba","author":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"date":"2020-10-02 15:35:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0855a9db39b64ebbd052a8a74bf6bb46430c6a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 18:18:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1eb9abca86f405882d4f812e9b7dc7f : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c14ad5bb1474630a3c282a8f11cab84 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97da3209e8984355a7e8eb2e388e5d82 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35bbab7a26df48a8afdb565d0fad0259 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83fb5557a0874e16a1603bac0773e838 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3082e03fd95341a589fdc88951c7d4ac : SUCCESS in 13m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/98cf8636359f4d2cb3e2abd4d3e34f34 : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/397e7397effe492f9c6510d67cb64390 : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50e8f33e581e43718f6e0fde81403142 : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/698332f615b5465bb9d41381144a9125 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8aa4cc4284c04202941ac59ed06c9ffe : SUCCESS in 2h 01m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b37a8c3b68744b6891a29fca6f76e8c9 : TIMED_OUT in 2h 33m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbf9ea2a345048bd8b45c9d1ac63afc0 : SUCCESS in 46m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6ec7dac465cf4e018e765f237ca229e9 : SUCCESS in 44m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c22ac897cac64e42a2f6c7393f75b0c7 : SUCCESS in 48m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b8331d6ef1f84d11a8e9c691871522d0 : SUCCESS in 48m 29s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/22b6c79af4d94417ab57167366260e16 : SUCCESS in 1h 41m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/14659db095534a67bf39258f7bcdd431 : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f4bf6bf9c6034b95b1764451e5dd37e7 : SUCCESS in 1h 56m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ba3a43c637d4467cb177826b8f5240a6 : SUCCESS in 2h 22m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3383364ef98b4de08da76d2b0f0840ee : FAILURE in 2h 02m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/050c11fa8d364a0ab33f39bae9d5d4e9 : SUCCESS in 2h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0742e7e4631c46ede2fcfe2fb648a84ea79c78a0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 17:33:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64dfe80d38d514af77485b794113bc6493ccf89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 20:30:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/314cc1a4b6f245cb8d4ddc9341eddbdf : FAILURE in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eba983c0c6049c7ad0170489ba4e5b9 : FAILURE in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a5b5bae72846b5949d507e171458df : FAILURE in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/387895fc211a48ff94b27c022f291035 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb5fbbf13dec4fc8bbdebe9fff41ce39 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66f9eaa336e540bfb0d3604c7c2f5f1e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8cd39f5e1334c5eadcd5c3a0f2b17bb : SUCCESS in 17m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/60b2067eca66491c811355b6f994a82e : SUCCESS in 5m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb57e9d8a3ee408284587bca5d294206 : FAILURE in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81b9d4a223214142ad4491c2bb4487bc : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/013bcdbd30514425a3a105cdf51d3caa : FAILURE in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/afde546e5e2845ca8765569789c4170a : FAILURE in 1h 24m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12a717e979084ebb8544c0f9c9121a62 : SUCCESS in 1h 56m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01cab2ef575f4cfba65257b05615ff0c : FAILURE in 42m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c645b30bbe914a51b404907f4f737604 : SUCCESS in 57m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40d2ba9cb600410487c494226c8ef28d : SUCCESS in 54m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97ad274bde974272b8b67690701fd3ef : FAILURE in 53m 47s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3d3eb4c518f42d0b48d8ee3528724fe : TIMED_OUT in 2h 43m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/756abe11ac7d4c7aa80ca31ee70a8524 : FAILURE in 1h 50m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/013eb5332d1f472cb00d649a2b66f893 : SUCCESS in 2h 07m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f4136c789dd24a5c9d2813be0c7a7346 : SUCCESS in 2h 21m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e5daf5cfc68a4f0fb2fd3578ce11238f : SUCCESS in 2h 20m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/37edaaefc1b948cf8b091e8b5dea5b54 : FAILURE in 36m 56s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py","accounts_in_message":[],"_revision_number":2},{"id":"ea0ddd41ea1fc34620fc9bfb8e6cafdf97b00892","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 19:34:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"83067c32c16be4eaa34df4b48e908ee797673031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:43:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4fc0879b3cc4fb7a66bcd14ee7bc9cd : FAILURE in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd41a257788141b993d8ab9822b64266 : FAILURE in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef8e1877d894de3b71c287947f2b2c3 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2f93fc276448a690e7aee64e45b1e9 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/253e78c383fa4af48a3c9f741ffe2e1a : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e84fb986d7ba4c1b9a591419160889e4 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9ca4bcc0c14ce7b9008a76e9b2a0ad : SUCCESS in 18m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7bb28676212242a6a02190eb10c80ac2 : SUCCESS in 4m 38s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7366afba07634fef874806198e7a34e1 : FAILURE in 7m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7fa00c33b9f14d99be4373f85eec2f25 : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb1b8f7972bc48d7a5c2db13c2d33cda : FAILURE in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d043439340f84445a5d90bfa132da27a : FAILURE in 1h 23m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9a6238c488ec4afe91f05a842b9b6734 : SUCCESS in 1h 49m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd689d0a3eaa4ebfbc881f9998371409 : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2719f207c85c4e3aba3cea0ddcb71bde : SUCCESS in 37m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/11ff5fded6b04e43a16c95fff872b973 : SUCCESS in 49m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce787b01991e4784b6821c0910c7b646 : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/118c998399b64d149cda62e83f9e005c : SUCCESS in 1h 38m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3d91d2f3fef14376ba0c450a6e3717aa : FAILURE in 1h 59m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7028de2d12004ca182a30813cb452bea : SUCCESS in 1h 30m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3e383fc9f8c149acbb7b29806f384473 : SUCCESS in 2h 12m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6a88782c7ddc491f962178c3b3b2364c : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1faacfb51d1a474aa505bcb3ae69c54c : FAILURE in 32m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b6d440aa9a2f8d461323b9d62b58525d71e476d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-17 06:05:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b5f9b3c55d46bb8f4a63d1b220c8609341d6088","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 14:14:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba052ad9adfcaeabc35059060abc011259c684fc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 14:15:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"876f2010cc874d0413254e70915e9154e9921f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 16:55:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18c4a633d90c46a6ac448fb5a30101e1 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e010b6c2b3564960a4101fc4c9e34c9d : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ae001bc1bb492c892beadeaa49d974 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad20fcd47ba8449c8fecc4a07e5b20d8 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39263a88d71a48778f24136fde8fe983 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6a2c9fc6fab429293a12f4478fa7ebb : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8404fffce1ad4870b12ee5e73f40334e : SUCCESS in 18m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9a67695e71f14e65a6074ac132430a39 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/28254cd124594e4da5f56134f7800a27 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/997f5b5d8b8147f38bb009e9404f5665 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39aba54e82564badac15e9764d007bb7 : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85a119142c7a4542a2cd428368e51d74 : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71beb81c836044669392b95cba7a302e : SUCCESS in 2h 09m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe1496a06b744093be2b589f68b73d82 : FAILURE in 1h 39m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/aac81bb0deea47429970acfd94fe3dc6 : SUCCESS in 38m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4007e4890cf3457f88e6b5508f25f920 : SUCCESS in 43m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7221940070643b3b22c3180c4336a74 : SUCCESS in 1h 04m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f861cde5121b4d49b33935ca9d8ba5b9 : SUCCESS in 2h 11m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/afdd7bf72c374360ac4c1fd2049c4868 : TIMED_OUT in 2h 32m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1e020fc22bf6434ab334702d98aae842 : SUCCESS in 1h 57m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f390ac83a7f84e22a76653590679fa01 : FAILURE in 35m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/66703bbe41494af6b2516dd071401f9e : SUCCESS in 2h 12m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6f81af2d1be94a188984619b396e4659 : FAILURE in 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0295c748c1570f3969691d39f2072abd849d92e7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 17:23:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cd76aba026fe8fb236f61f930cd184313e04baec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 20:00:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44a9b930d3444959a6898ae5ad790e17 : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4db4e69b5b7343fb888e7e82d3ac901e : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e870b679cc9433a9992fdfa2dd70183 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d883dc040df54f8f9c600cca75218390 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f29360cbfe0d4a328b96d67acbeefbff : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96c26f348cdc4bbe84d54497e62e4e4b : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9f02409eaf4f4dab411fa8031ed70f : SUCCESS in 19m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2d081070f63a4dd2a80afdb84eedabe0 : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a50e34fea4044f998ba5a2cb84ba7e40 : SUCCESS in 9m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ad7322f162e4a87b5810f781353e890 : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/326466696eb149bdbb96f40f0448f0e2 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/552218e06cdf4067a8bca7c7fb48bec0 : SUCCESS in 1h 18m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63b4178725ad42d59cbfdc7673736ea1 : SUCCESS in 1h 58m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b59f7230156b45d69efc8bba7fbb69bd : SUCCESS in 46m 32s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/45c2cf0a6c744ec684002f7c9628c5ee : SUCCESS in 43m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31a9271d085242bca826cee5c3769cd4 : SUCCESS in 40m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ec5e762d9ce6451e8a408e56a3d6795f : SUCCESS in 1h 04m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c457fe9872e4b5b8fa7c7bb7f769a47 : SUCCESS in 1h 58m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/799407d4367a4dc68cac8461fa2b5c43 : FAILURE in 2h 07m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cdb1edfed964d26b470d4ffadb7a7ec : SUCCESS in 1h 27m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1dc6d2f68021419f9b5eeb602b083435 : SUCCESS in 2h 26m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6328df324da2413f87a8f878adf06863 : SUCCESS in 2h 03m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/283713b280374ba3977ed1cfcc063f8b : FAILURE in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc68164491678f265b9a6ddfa4d9f6a11bb33349","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-18 07:49:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c204fbdbd9cf36eae25432f5a1401585d8112377","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:54:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"708b6185848bfb5ee7ed942a585eb94a20849e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:55:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c675233fecb108410fd8ed3e17f84c53d67cbf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:40:00.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/996bea2a061c436b9bb5f4b8bd4db3d7 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32394dd0fb884c97b751afbf5341982c : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b83e0c35a774f02846837f92da1a909 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1d060f673004d0094fa0655390db1ad : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50bb9e92606d48b49bd0cb3f9b0394dc : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2352ea4c8a694843bb358c7d0ea018d6 : SUCCESS in 5m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d8890ed82da40e182ec2618d514ed4c : FAILURE in 9m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/185244609a2c47a381190f80b3666235 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b25a44009d64e7094d05b9860d01fe4 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a255f302f73d46edb6b3b3b59784f7c4 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4ae6eaf31eec4f0795d725121cb3c34e : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd8f09c2f78e49478516e089c6ef0726 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d74c8d8bc8db411996e342ce3e9986f5 : CANCELED","accounts_in_message":[],"_revision_number":3},{"id":"ef1265c278f5212bef4c7f9f8ba8becd6f4b77a5","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 20:20:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8cc01b9f8ba3fcc19aac197842d587f113932be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:31:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8622e7b414849a5974a9e1e6bbe15d2 : SUCCESS in 8m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fe7dc9c548042fcaca163f7b114c02e : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ac0eab7c2b4275a42996449b8acf61 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d131281d8ef4dbf862dca41e50e5ae9 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bac37bc32ce4f5eb201ed2202598ed4 : SUCCESS in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/851c937647934d568afae7117689698c : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2b64a173dd4314a7f3053148434b49 : SUCCESS in 11m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1a81816bb84479786350022ab0cef0b : SUCCESS in 6m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3dc52ecb67c3462d9bbaaf6d8e2f7502 : SUCCESS in 10m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da08d09f6b8247feb5c8f94c49f6df24 : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d30aab5bd2da46cebdd939162b3176c3 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3a96d4b89fab4b1eb10e895594670bd1 : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5e390452561418c85fe4a054be70ed3 : SUCCESS in 1h 45m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22800c336a444f99b1886c783599cea4 : SUCCESS in 43m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0aef6b8e56924695b48962261a149a34 : SUCCESS in 33m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e58979be04af40c2bfe195d453f55690 : SUCCESS in 40m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9af3ef7467d540dd83462eeb984abca9 : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c4473891e81400dbcd64bfa944e36b4 : SUCCESS in 2h 02m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e6959a1fde346be8d06dd9c82855348 : FAILURE in 1h 52m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f9aed57d3e714332a1fd9eb5427b58c5 : SUCCESS in 1h 45m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7cf6cdf1f252466bb5cdf33cb489b29b : FAILURE in 1h 05m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/88d84ab16ab04846bdd3ceac9f173f99 : TIMED_OUT in 2h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e897cb59f43d4ed0a16384758d87b3cc : FAILURE in 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"437810ffa59c2fa945942b05b493c4f00a0af0be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:31:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ed75b08b29886dd302d12f2e0090630b9f403c4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:13:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70175de43cd448c2b3520d46353aea0c : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54da1167ee95476eab4c561c6d212655 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e01cc3412e240cca43afc5c1ebd2c15 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53a351a610064b7e9353168096563b31 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb59e4c0a8b4901b641f88bb6c82397 : SUCCESS in 12m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/501282a2a80a4603adf128bc790e379c : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5281f9595abb4b64a7d8ed4d31aba336 : SUCCESS in 7m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3908e42137de481f87990c003991746d : SUCCESS in 1h 11m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6dd408c507b4d8bae584393c753027d : CANCELED in 1h 16m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/65176136ba9e40829521b326329bceb2 : SUCCESS in 38m 02s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dff4c5f5cf2344e8873738ba1fd2c7a5 : CANCELED in 1h 16m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd9b39ef0b814571af2f3643f4e7d533 : SUCCESS in 38m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49fd2d223cd843bda9444884160b88c0 : FAILURE in 1h 16m 38s","accounts_in_message":[],"_revision_number":3},{"id":"fda2b32e982484c201a90c1a8fed00dbac949411","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-29 01:18:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5459203687c706e0515e2adb7eed39a9b51db36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:30:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e19e92263ae4a86b3e99d817357fe96 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2bffcbbe85848719ff45d78d80a3fd0 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e3506690d048feaffbfa58a8621204 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d0cbf760f84895aa6cbf2c8ce2a4b9 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddc57ab6957d49599aea9d58e5289ffd : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbfb30903c7d456cbecc974fc52ef204 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35da6ee1c2e4f0eaa89433c9ea74906 : SUCCESS in 13m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a18256e26eda43e994d9b8c786a05ab6 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0482ad889deb438dbc9702b4fb3a51a6 : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/329e9202e3954371b90a25ece2181620 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77cedd5195f5443ea1a1996eb1a9909a : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a07fdd3b5c194da1a699ae7ea70850f6 : SUCCESS in 1h 13m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce5eaf23ce644e11a0760590955a745a : SUCCESS in 1h 51m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67e9412a95e34723ba25177b6afa3fca : SUCCESS in 38m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/605ebda3c9a448de8e9d7dc0b59b3275 : SUCCESS in 35m 58s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6480a607e494afc8b9b03754585480c : SUCCESS in 42m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7131239451146dda180b380c1153e9b : SUCCESS in 58m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/676824d021694d43b17f6ccc80eae32b : SUCCESS in 1h 39m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1dbdbc54b6f4f869e8a769e386addf1 : FAILURE in 2h 08m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/32b5dc12fcdd41de8ea4cb554f83c560 : SUCCESS in 1h 45m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/42d8f7e694264f65bc298db967a56e7f : FAILURE in 22m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b36d0ea80d254f6b932d7d007b011175 : SUCCESS in 1h 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bfcf9caa68f146e5864272fa918512cb : FAILURE in 33m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a7b8b7b2309663a1eac23ac67c36a1f729c05add","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 03:30:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3eb76d864b7387470bfc7c0069d6ce0073c3d20a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 05:19:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07173fffdc434f34aa608993010791fc : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc9af267603446a2ab541add1bfc7b2b : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3794c23c0e8e4fe4ba371699979fcb5c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c7b96c9afec41f1855ea95f7a3183cd : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1987dd48beac44bf983e32ae2027bab4 : SUCCESS in 10m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6f3c02612efc4123b8374f549d50c726 : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9836fb3ffe2488ba9b603b30cd4369f : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f42a52b4b3a41e499543c232704db6f : SUCCESS in 1h 17m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f211b20d6b0c4b2283b0867ece506650 : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c9cf430f26e45c7ae73e7a3f0c5036e : SUCCESS in 38m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/56642b29917142b9a0ea4d5c4e29e2a4 : SUCCESS in 36m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8bc6a55bb864c089d918de72d6fd17c : SUCCESS in 39m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba86fe46295b491d904745dc0732be98 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":3},{"id":"eaef560670d90235fe89006d806b5a77c5cdd146","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 05:19:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"310252d469ea95d535175dfeb104c85efa3829e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 05:21:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb98c508375d491ebf02500ec91ff966 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7ad022379ff2e366d3dfd73113d4619b0137551c","revisions":{"58ac28575e3d61e92c094f8df10ce9f2210ce860":{"kind":"REWORK","_number":1,"created":"2020-10-02 15:35:32.000000000","uploader":{"_account_id":28514,"name":"Andrew Karpow","email":"andrew.karpow@sap.com","username":"andyk"},"ref":"refs/changes/28/755828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/755828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/755828/1"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:34:04.000000000","tz":120},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58ac28575e3d61e92c094f8df10ce9f2210ce860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58ac28575e3d61e92c094f8df10ce9f2210ce860"}]},"branch":"refs/heads/master"},"67f55569ba025f181856f1c6cd441546304f5b2b":{"kind":"REWORK","_number":2,"created":"2020-11-12 17:33:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/28/755828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/755828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/755828/2"}}},"commit":{"parents":[{"commit":"17ae2625da0487b7398b927e4e67a713b61569d1","subject":"Handle listener not found during loadbalancer status update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17ae2625da0487b7398b927e4e67a713b61569d1"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 17:30:59.000000000","tz":-300},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nAlso moved assignment after initial session query for the\nobject.\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67f55569ba025f181856f1c6cd441546304f5b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67f55569ba025f181856f1c6cd441546304f5b2b"}]},"branch":"refs/heads/master"},"7ad022379ff2e366d3dfd73113d4619b0137551c":{"kind":"REWORK","_number":3,"created":"2020-11-17 14:15:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/28/755828/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/28/755828/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/28/755828/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/28/755828/3"}}},"commit":{"parents":[{"commit":"17ae2625da0487b7398b927e4e67a713b61569d1","subject":"Handle listener not found during loadbalancer status update","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17ae2625da0487b7398b927e4e67a713b61569d1"}]}],"author":{"name":"Andrew Karpow","email":"andrew.karpow@sap.com","date":"2020-10-02 15:32:52.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-17 14:14:41.000000000","tz":-300},"subject":"Support deletion of all tags","message":"Support deletion of all tags\n\nAdd support for removal of all tags by PUTing empty tags array.\n\nAlso moved assignment after initial session query for the\nobject in the listener update path.\n\nTask: #41009\nStory: #2008220\n\nChange-Id: I7488f2fae61917f6d4a56cedd05bace7c5e2bc70\nSigned-off-by: Andrew Karpow \u003candrew.karpow@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ad022379ff2e366d3dfd73113d4619b0137551c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ad022379ff2e366d3dfd73113d4619b0137551c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
