)]}'
{"id":"openstack%2Foctavia~756505","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I37926e8547ba8b994593ff5a3838e9620153e999","project":"openstack/octavia","branch":"stable/train","topic":"train","hashtags":[],"change_id":"I37926e8547ba8b994593ff5a3838e9620153e999","subject":"Verify pool exists before checking its LB on HM Validation","status":"ABANDONED","created":"2020-10-07 12:55:49.000000000","updated":"2020-10-15 11:27:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b5c04b276533bc356e866861229baeb92af02698","_number":756505,"virtual_id_number":756505,"owner":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-07 15:34:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-10-07 16:46:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 15:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-07 16:46:22.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"5dc660c90077fcafa55bb29ed3b0057cfb2c116c","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-10-07 12:55:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15215f7830fd6496545620ff8c117e08cb2ae5be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 15:34:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10484b9ceff14453ba20e77a6d5b3d39 : SUCCESS in 8m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a2bf916a1b348d0bf005bc52466c1d3 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e1d71ee8ee47eab523c1b848ca39bd : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e96beb8195e4ac29b5c9923ea9aeff3 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17861bfdf7cd41aa89bd040b82f030ce : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e2593b04609402bac34fe4f0aaba396 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815acf531f6e409f9ab298c34757b635 : SUCCESS in 10m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/557be02494f24fbd85fd8178b5c032f8 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/573e5e76ba3f47eda40e2e25b352b8fb : SUCCESS in 8m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f7ebfceff4f34940b22338bcca25fbb4 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a99301664414d1683ecfcdf81ed775b : SUCCESS in 13m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cce8f573f0ae4bddb774f249dbc796a6 : SUCCESS in 1h 21m 21s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/85621445595243b2b67eb51ce8fb96ac : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b5da059e0c8e450da38d5218f79f0668 : TIMED_OUT in 2h 34m 07s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/525a017a82824554a0114014d3745191 : FAILURE in 1h 34m 22s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4650c10a1f3c417b8df25ff91a35e200 : FAILURE in 13m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78c9e19858314aa2a15f8ff88e63ca87 : FAILURE in 27m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c99b8ca627d4450194418c5eb690cdf7 : FAILURE in 44m 29s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ecc540abc7446e1aad4e32b8f6f48ea : SUCCESS in 1h 32m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/beda56cb53b2438a8d475a6860d1315a : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/926a42f038b643608bd67b3fa4a9576c : SUCCESS in 2h 23m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d6df8db3051049d39adde0c832ddefb2 : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/87eb684cb8eb4130a2ff143188858089 : SUCCESS in 44m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/1c7c3f4f93734d0ea3515f63a08d92bb : FAILURE in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f074722512999a42a9b4fe2e075bff078b52608e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-07 16:46:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThis is not a good solution and we probably need to fix this on master and backport it down to train.","accounts_in_message":[],"_revision_number":1},{"id":"f093f7dc3f9dc8342a3c94c49e31e1d5ae70706a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-07 17:05:01.000000000","message":"Patch Set 1:\n\nI have also added additional comments to the story.","accounts_in_message":[],"_revision_number":1},{"id":"8da724957ad4bfc16c6cc3355a5748811a821795","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-10-15 11:27:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6e4f6370baaeef6c2689d77411d5a7e229a9860","revisions":{"a6e4f6370baaeef6c2689d77411d5a7e229a9860":{"kind":"REWORK","_number":1,"created":"2020-10-07 12:55:49.000000000","uploader":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"ref":"refs/changes/05/756505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/05/756505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/756505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/756505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/05/756505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/05/756505/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2020-10-07 12:46:45.000000000","tz":120},"committer":{"name":"asarfaty","email":"asarfaty@vmware.com","date":"2020-10-07 12:55:24.000000000","tz":120},"subject":"Verify pool exists before checking its LB on HM Validation","message":"Verify pool exists before checking its LB on HM Validation\n\nWhen touching a healthminitor, _test_lb_and_listener_and_pool_statuses is\nused to verify the status of its pool and related resources.\nIn some cases the pool cannot be retrived, and the code can crash.\nThis patch adds a sanity check for this case.\n\nChange-Id: I37926e8547ba8b994593ff5a3838e9620153e999\nstory: #2008231\ntask: #41073\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6e4f6370baaeef6c2689d77411d5a7e229a9860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a6e4f6370baaeef6c2689d77411d5a7e229a9860"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
