)]}'
{"id":"openstack%2Foctavia~756597","triplet_id":"openstack%2Foctavia~master~Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","subject":"Fix load balancers with failed amphora failover","status":"MERGED","created":"2020-10-07 23:02:20.000000000","updated":"2020-10-08 17:31:30.000000000","submitted":"2020-10-08 09:02:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756597-1602147745536-8f1aa1ae","meta_rev_id":"8a19e814dfac518609dcdc6ec29e66846c454fdc","_number":756597,"virtual_id_number":756597,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-08 09:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-10-07 23:39:32.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-10-08 07:09:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":1,"date":"2020-10-08 07:09:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":2,"date":"2020-10-08 07:09:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-07 23:39:32.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-10-08 07:09:10.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-10-08 09:02:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"654098f7dc0b72d5569085c4fc8321779d195d4b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-07 23:02:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51cddaf96c2244b8973ffebb70cccc9b5cffda02","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-10-07 23:39:32.000000000","message":"Patch Set 1: Code-Review+2\n\npending zuul","accounts_in_message":[],"_revision_number":1},{"id":"0268c6bbd481164dcb47aeb531c39269d864c1c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 01:39:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d117c567a3a54b3397de62bc6c439bd5 : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/007708bbdb2b45939ea0632065b5efa4 : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584d343faf6e4595910774db8fab4ab2 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee7acfd602704ae38380f5249a9de1fc : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a0cd0ff40ce4004bf048a9546ec5c98 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5027e18f7b6c433782fc758efbb8a462 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f371e2ee0c9e41e3aca2daed2194edc3 : SUCCESS in 6m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3e234710e2db4a25b9b1dfc8cee568e5 : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b095a7aad3564d7498350b2b07d4d32e : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f92cd715f16545afa091494945f0d450 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b65e4116c5e744ebae1e12621a71ecff : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c23aa0266f59446dabcd248def355244 : SUCCESS in 1h 36m 15s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f47221e7fb445a5ae4bb78aff5d802a : SUCCESS in 1h 30m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/60fb1fb2d3094fdc8db421ac0fa56166 : SUCCESS in 51m 47s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f52d76584c284f838918aa620754ff6d : SUCCESS in 41m 28s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/075c16fe9fc04d3781a580a98ae3faa0 : SUCCESS in 41m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1b0ab65584944280b66449122de72461 : SUCCESS in 1h 02m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07dce6863de44370a5eecf5f69c0ffcd : FAILURE in 2h 01m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0f58f4c5fb9e431cbfc33acba655c13d : FAILURE in 2h 06m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bbe1012b7be64d4798999c0a020e5a6d : TIMED_OUT in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c87e177a07bb4c749e78561884782d5c : TIMED_OUT in 2h 34m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/548f5ed87032457faeaef05973e679ea : SUCCESS in 1h 38m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c0deb4fcef9d444e97149af62f05c398 : SUCCESS in 2h 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"359a1f96a12d52ff0b37dc71b697082f55bc38ce","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-08 07:09:10.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6cf6da80feaaed4187e67bda0fc55a57f56a381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 07:09:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98c8ed9a608972dc290557cb1004729407984e3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 09:02:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"55b37518a27eed171235c599969daf4eb2e7dc22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 09:02:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7bd6a4ade6f4b578a0caa88e2907193 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0e99bd1fb24269a919ace4e466e5a0 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43981ec4e5b54c40a35d7e957317cf04 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/462dda2f91344a34b7ac9a8062fa93c9 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b1b2e4857c4a4e833f55a646d163f8 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6714dd6da7304cbdab58e7758fa99413 : SUCCESS in 6m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46cfd644f29643ac9a6b4be04ed5bf4a : SUCCESS in 7m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5680e49dab34acba485cf1ca6000a59 : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16c37fbd4777478fb1225bd36365980e : SUCCESS in 1h 33m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9f776b91b8294bd4b54e07552cc50113 : SUCCESS in 1h 50m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/474d02fd2a3f4b58a58ac587f8e7e262 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e13eb135dbda4294b64133406ded0024 : SUCCESS in 42m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a59d49a189445188b055b9a4c3fcbc6 : SUCCESS in 45m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ff1702f77adc47bc8e3d2fca90f1bdd2 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a01d13d175f0a2995141886d50a9fee1c9565836","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 09:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028ddc1c9ce4401a86383741d704ffde : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7cd49894c954a72b1fac7d3d8d46800 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"8a19e814dfac518609dcdc6ec29e66846c454fdc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-08 17:31:30.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit a865c03c3b4573ce5855c1d90aad43b5950ea9c5","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77dc23fc26cbb8c6488b5205c63bf7cc847a9c32","revisions":{"77dc23fc26cbb8c6488b5205c63bf7cc847a9c32":{"kind":"REWORK","_number":1,"created":"2020-10-07 23:02:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/756597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/756597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/756597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/756597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/756597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/756597/1"}}},"commit":{"parents":[{"commit":"418ea0908c44c97261d353a531ef533613739618","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/418ea0908c44c97261d353a531ef533613739618"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-07 22:59:50.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-07 23:02:19.000000000","tz":-420},"subject":"Fix load balancers with failed amphora failover","message":"Fix load balancers with failed amphora failover\n\nThere was a bug in the load balancer failover code for the amphora\nv1 driver where if there were multiple failed amphora, with session\npersistence, could cause a load balancer failover to not complete.\n\nChange-Id: Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2\nStory: 2008099\nTask: 40802\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77dc23fc26cbb8c6488b5205c63bf7cc847a9c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77dc23fc26cbb8c6488b5205c63bf7cc847a9c32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
