)]}'
{"id":"openstack%2Foctavia~756902","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","subject":"Fix load balancers with failed amphora failover","status":"MERGED","created":"2020-10-08 17:31:30.000000000","updated":"2020-10-26 13:50:54.000000000","submitted":"2020-10-26 13:49:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756902-1603720159604-5b795edd","meta_rev_id":"783e98d1e72150e6ed1523484a14c25e24390dd2","_number":756902,"virtual_id_number":756902,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-26 13:49:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-09 09:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-10-26 09:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-10-26 09:16:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 09:18:13.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-26 09:16:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-26 13:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb3aef790bb4508fc8bfa3d4647077ffff207118","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-08 17:31:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"681bd74e8bbd126c65c58e79401bcbe81d34406e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-08 17:31:40.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit ebf190819e109e6a9eaac86cd80092ae34aa44a0","accounts_in_message":[],"_revision_number":1},{"id":"8daee1e03d7935053e8ce33474257ac08d947c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 20:20:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a79c49283a9f4d4d9c3df24c78ce132a : SUCCESS in 10m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd527301795748b099f0017013cd579e : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2267e8ae678842fea1d9b49035708d30 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c50563af5b9449258a1cfeabea5a9cca : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70cfb55daf1944fc9013f9da2e6c5691 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59594bec69c04567a31b60a560fd6b6e : SUCCESS in 13m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6934adac52074bf8b08638d93bb42391 : SUCCESS in 4m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6c37d554ee0541c8a6975e84f2abfa29 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bf6a28abeb0e4160a65f75f8c2250b0d : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5719b741e3f648f2a2c3e96dbaf6e35f : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5245db76dc4c4d72ae74c64e231c3d1a : SUCCESS in 7m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/06542a371504485290d7e9494d246004 : SUCCESS in 1h 48m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8389b53a25304534be951efd0c4c9a1c : TIMED_OUT in 2h 33m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e87a50192b94a838e145c77aaacc217 : SUCCESS in 45m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/be18b18fc4464dc38a2ddc712e42965a : SUCCESS in 35m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/761653ef1d4948b2bb8f00b3c9475b30 : SUCCESS in 43m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3a6a8d82c52f44b2a306d6c719ba1dfc : SUCCESS in 1h 06m 30s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bab38c84222249129bbcdd2b8629cdc4 : SUCCESS in 2h 07m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e7c3d0cde53047ea807eff32a27e6b5d : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4c7bb1cc8514e1e821ce995b7c34cab : SUCCESS in 2h 09m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/83c3ad927b9e4b81815e5d1bbb3dbf29 : SUCCESS in 2h 16m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/da146ca0f70a487ea488fc40aeaaf002 : SUCCESS in 2h 29m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c4138fe7f7b44b84983fecadb2014a5e : TIMED_OUT in 2h 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a83d41717fe97b2902ebf2fff9372300bb927c7d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-09 09:18:13.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6de88b15406e7d3a557aa5189cf0375cb1f0306d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 12:08:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1799695c64c44aa49ccc3dc2d48b28ee : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df99750192dc48b5a1b84c82535098e6 : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d2eccdf7bd46e28b173abd47c9d4db : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b292436bed664bb6a50f724c5f9d2aba : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d30f65277d2c4fa3b9fad1190ced9e17 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97440936db34cd18c9911a0004b7e9b : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0181cb588124f01834ab2a0eb5d86b9 : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c9e101b8f68d472fb08dc8442cd68c8c : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ed290b6523904d59b866bec30fe6d242 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c4de0101cb0943188f77c6e7fc0397f2 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d744cb354814fd49aa5e222caac8b42 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/677e98b39fec4b448c0a6a9cdf576666 : SUCCESS in 1h 35m 37s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b8c5a9fe35745d58ba4b0ff1b92c965 : SUCCESS in 2h 17m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6aa6693820a644f99029cbb4de653286 : SUCCESS in 1h 04m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/778db3dad7244f0d8eb83798226c4f6d : SUCCESS in 40m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0e2e956270d4c0bbac93b009bf4e8a4 : FAILURE in 28m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16e40c53f6364301a6aa6cf66b35c0c2 : SUCCESS in 1h 02m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a77324d718a3464cbadac6d1b2826da7 : SUCCESS in 2h 12m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94c98bd49b2545e2ad34dffabb449770 : SUCCESS in 2h 14m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/73b85a962a6d4d648ea95394beae3c68 : TIMED_OUT in 2h 35m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ad346c15db8040d4b98c759368ae7b30 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/df9a4856b1f44d7fb7f765e3de4696be : FAILURE in 1h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c8b47448d0ce47aab58c3ae3c64f2b87 : SUCCESS in 2h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09ef892fd864ec8c1a72b53c140ce9ac5af29c8f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-12 08:14:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"758d4b895836b42f9666faebc4bf3531c5c070b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:28:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3b0a542e5754e1bbaf4cab35040ae18 : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cab8fc47e7a740e1a640198ffacc4473 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6fa1d026f5745ed8d3c514409a39333 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f7ccf46851f470b80a53c19292ffafe : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb163b76e5a043b896c8264c08cdb4ba : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb3098285264fa3b2cb95b3519fafb2 : SUCCESS in 14m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d8796c3261492d8a155782214a0fd5 : SUCCESS in 7m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e18b2c8b85db4836bfe6e849a84fb542 : SUCCESS in 6m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/49ea92e6e3d547cda854ec3584460bb1 : SUCCESS in 22m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c11ad6d29b884ad8bc4b891abb9d6dcd : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/517753d595e84fc7ab8e370d6e5134c8 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c27ee75e99a6424db62f9e6f3527d99d : SUCCESS in 1h 36m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f34e8ba186854e909409809f0804b965 : FAILURE in 2h 20m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/86e7f39b1e874f3bb0a25c0086131b20 : SUCCESS in 43m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b2530c007ab2403c9770a8efe59f1adb : SUCCESS in 40m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee507f684d6e434f86e0c3e1e9b15fe2 : SUCCESS in 1h 47m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2e39512c6af495bb7eac57e6ee1f467 : FAILURE in 1h 33m 11s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ab3142472f04d8daf09062d08cbbcc5 : SUCCESS in 2h 00m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1f67b57c48784d67a0f9beec08ccb402 : SUCCESS in 2h 24m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8cfe8098a3a947a6ab179852b4e85562 : FAILURE in 2h 20m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b2e0fe9f9db545ac948bd9fe153065f5 : SUCCESS in 1h 56m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/edb10729c9f24c5887b5b3c43f234467 : SUCCESS in 2h 01m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c115c27f0fba43698238c6ee6d4c4a78 : SUCCESS in 2h 24m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"667ae9e33b5b8e4eeb3b4778699a3fc68b8425f6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-26 09:16:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b97899a23a14ea511d79282f252eaaad01e01fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:54:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34b8a539e5ca4060aaa83d5ca6d57c8d : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04b7a822f31748b19f5966391715bb22 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9173bcf2748c4eb4b9a73098c61a8dae : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4880f92329b742cb97e97d623a1a43b4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fddf97f3f24a73bd5b0d685dd63f63 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff3ff129bcd40f8ba9659f234683d5f : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0184c286c364ba7aaf04c64656fcb60 : SUCCESS in 4m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/af366783bb9b4a34bbd21c27b182a17e : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/11434a2efb9b4a97a5f34ef33a6e5c65 : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea905c82d30a48318ca4bc58034d2d4d : SUCCESS in 4m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2550bc2414164b469c5ff706286e5bfa : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/79b1315606aa4630b3fda2ded17e5bd3 : SUCCESS in 1h 46m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d8c6180a09844e68f49859d5a986331 : SUCCESS in 1h 35m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48a8ded156d7429f9425f1e1cf48acbf : SUCCESS in 44m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4521cce9a81d4ef9b49e3f17e13c82be : SUCCESS in 35m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd69b5a2b7c94eaeaa1202032b074e00 : SUCCESS in 43m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1a9b10252e1f4c5fb0cbf3996e7a81e6 : SUCCESS in 45m 15s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/99cf3036cf47404b9a58a13284efe592 : SUCCESS in 2h 00m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ac0bc222f1b4d5fb931542351728437 : SUCCESS in 2h 20m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3c7e6519c4e449a08f14acdab33e34f0 : FAILURE in 1h 59m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dd1c3bee61834268b5924cec61c96532 : SUCCESS in 2h 01m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d1f6f1a99a614047a0c980b5caa24998 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0d2a5fc82d83475daef1a5b10cc81d00 : SUCCESS in 1h 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5946387beb8fce610c1d194849b215da04173e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 11:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed90b43694fa9417b30bece3700e4704f5f78a45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:49:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fce36a0b263e410da1aa708b6fbb87e3 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c18c5f692f884ac2a0585d42f4dca0da : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e35c3aca864fc283b7b029aa8f0171 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/210d4e37dced48d6a98234e95baa1d72 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb73db9802e46888b0439fad7fc9c0c : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/474b035fc0e14a48968db9fb2ee86488 : SUCCESS in 4m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6f36b07cf4004c25ac01961f4d2b51b4 : SUCCESS in 4m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10f73ac3d5eb4280a5f5cd2c58380c54 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24c9be0e4d3f4a50893d868c0b191fe9 : SUCCESS in 1h 32m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a059805aee6c465a9a9a432384b65178 : SUCCESS in 1h 52m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9b96a0480d564824a57cd21e7454c9f8 : SUCCESS in 36m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/270fe038dcc64988ad8cc67c184c4cd2 : SUCCESS in 38m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63671f40f9764d29ba1639f8a918c93b : SUCCESS in 45m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79dcdb4bef6745b8a716c6868bdd9846 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"17d60f815c29b5e1f5787e0fc141562e9f167731","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:49:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"783e98d1e72150e6ed1523484a14c25e24390dd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 13:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2454bac1e804bf1bd5d7809fc02361e : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8285a9c376e54f5fb674d2fa8a47d040 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a865c03c3b4573ce5855c1d90aad43b5950ea9c5","revisions":{"a865c03c3b4573ce5855c1d90aad43b5950ea9c5":{"kind":"REWORK","_number":1,"created":"2020-10-08 17:31:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/02/756902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/756902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/756902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/756902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/756902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/756902/1"}}},"commit":{"parents":[{"commit":"ab0246b84581077c85e8462be6d396f96f07cc07","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab0246b84581077c85e8462be6d396f96f07cc07"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-07 22:59:50.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-08 17:31:30.000000000","tz":0},"subject":"Fix load balancers with failed amphora failover","message":"Fix load balancers with failed amphora failover\n\nThere was a bug in the load balancer failover code for the amphora\nv1 driver where if there were multiple failed amphora, with session\npersistence, could cause a load balancer failover to not complete.\n\nChange-Id: Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2\nStory: 2008099\nTask: 40802\n(cherry picked from commit 77dc23fc26cbb8c6488b5205c63bf7cc847a9c32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a865c03c3b4573ce5855c1d90aad43b5950ea9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a865c03c3b4573ce5855c1d90aad43b5950ea9c5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
