)]}'
{"id":"openstack%2Foctavia~756903","triplet_id":"openstack%2Foctavia~stable%2Fussuri~Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","project":"openstack/octavia","branch":"stable/ussuri","hashtags":[],"change_id":"Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2","subject":"Fix load balancers with failed amphora failover","status":"MERGED","created":"2020-10-08 17:31:40.000000000","updated":"2020-12-03 20:56:31.000000000","submitted":"2020-12-03 20:54:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756903","meta_rev_id":"513d1a73fe2758e2a370b54f25213ef1741ef25d","_number":756903,"virtual_id_number":756903,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 20:54:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-12-03 20:56:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-12-03 14:02:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-12-03 20:56:31.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-12-03 20:56:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-03 20:56:31.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-12-03 14:02:03.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 20:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 09:16:50.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-10-27 09:19:14.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-11-23 14:24:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2020-11-23 14:26:15.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-11-24 10:05:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-12-03 14:02:03.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"38c3f2c45d307f895af60a14bd45c5da0eff5f67","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-08 17:31:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"56163f33030123268bbad1377ae8be2d45b23100","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-08 17:31:51.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c33f6047f95b3cdde13db1995ef39037e5e41dd0","accounts_in_message":[],"_revision_number":1},{"id":"92cdd278dbaa54ee1c8d3633c4e875a52c9861e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 20:27:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3b19b8ba0ce49e5b6f3926ec08b9352 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010a6a9f71bd4e19b34caff6bb8f815d : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ed3d0962252461285716eccc6306b4a : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76b8c71452134af4aca4946b9f63d3be : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2528396fe16d49e7b0001eb8642bd82c : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02f3923792147cabba862545e8efd7d : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b0cd96299b5419c8b1208806f6b17ed : SUCCESS in 4m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c4574d1e2ba443a6b6e1623acfbd3d16 : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cb78998f15b742a1860594c97f62503c : SUCCESS in 9m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5bf3caa0444c4398b3fdd7e5558bf343 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6515ee5f763f45698dfbf14e19dd2c54 : SUCCESS in 1h 36m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a96954a2c2645df86233ae061db5e55 : SUCCESS in 2h 04m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c100b481daac4a8ba4c29aa887f68204 : FAILURE in 1h 28m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e13d55a51c4a48a99537557d48ebb411 : SUCCESS in 43m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c879927376f4429687043769fb8075a7 : SUCCESS in 40m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fcfbd2d6c9c142ababef645cc633076e : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/465c8d24797f479991bf65b22d7dbc39 : SUCCESS in 2h 34m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d33bd593121d400bb30d56fb87379c8f : FAILURE in 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/06e4ec6533a34c02a0fce50371aabb7d : TIMED_OUT in 2h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf4a2684b026873cd982058e46f453afbb715979","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-26 09:16:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5157aefe250091e86c2bbc9746c59124e175a34d","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-10-27 09:19:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"79af439f50b99530e99083750a7243bc6c37daba","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-23 14:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5b0b51cf9e6926cb9bd86c16327434048a3af90","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-23 14:26:15.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"74ce6c76b415795dc73ae034f1e2276441531af0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 17:05:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a0a0ec355be4d809dec7d11862ee334 : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35f5b42afa9b43e1aa020209f749292e : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3e1f47831e84bcd9a1707d60c871d1c : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1309adada0e14e93ad3ed378e12ce03f : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f12e5f78eac4f96b81eef5353c8f86f : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002c23ece0e04a758b0de3c37dea85f5 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09290e3e446c4d428abde238009957e0 : SUCCESS in 4m 14s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c47bafebc84b4d3bb304a255fb919dd6 : SUCCESS in 4m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1b2dead398684f3a99c5997ae2fb052d : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/54db0d196eda47f09c0d1c8dc0c39c73 : SUCCESS in 9m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4608c3f89fe4a62ad59737717262962 : SUCCESS in 1h 24m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c1a4fc7c30544efb46a7141300bf652 : SUCCESS in 1h 58m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27103d57d6a145a8acf009b152951d84 : SUCCESS in 44m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b9d0183fdd6f4a9c9a88b407f9158dc7 : SUCCESS in 36m 15s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bdcc5a29f3ab4727a69aed2d13fe8809 : SUCCESS in 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3ce60b98a07243fdb89fba40f8406765 : SUCCESS in 1h 28m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a262b295f3944b45a35fe752b48440b6 : SUCCESS in 1h 50m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/368a749fbff34d49a54b258f559781bb : FAILURE in 48m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b64fb83a3b15495fa8753ee5be1733cf : SUCCESS in 2h 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb26e724206395431df0a14676e83b487e16084b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-24 10:05:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fd84c78568d310364512dfd167e273842867961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:05:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d22efb9b03450a7f1bf87228222ec5987f06e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:31:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83aaa02b71674d7088a196135a14ff09 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3987a6ac3e05487ea982c4720fe7985b : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce4b10ea1ced4b568a0682d5f7bb4db5 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a38ea651dd1144c38c54b3a48efbcd37 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c64bd5f68046f89c8dd418a6d761dc : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/739c9595b8b0437abda7e401a799ac75 : SUCCESS in 15m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db8244f4845d48fe977b639fe75fe116 : SUCCESS in 11m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b50daeddfee04721be24af2199a12f46 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e686f443ce7410d97867773c1f52cfa : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c4fca88db1a41a0ab6b13e4913db600 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4f69239d3e2c41048001a357a97c6757 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82c36e6eb20142519000832405736fc3 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb732ed8226d47eebfaadd73a6a9edc2 : FAILURE in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8bbc301658eb794d42f0c513a649600731a816d2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-24 15:46:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59a1c2ebdd855ffdb010b6af6b961f0e54532d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:02:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86d8fcb88e874861a7b0686656cba874 : SUCCESS in 9m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ac0f501cb447aba5a58db09a284b0e : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d85166ac1b4139b619176dd1b99a79 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94b2f552178a4dcdb74f7f859e7fcf1d : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e3ddd307abd4881a87b890e525e630e : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a2c375621a434898c8407cac0cf9c1 : SUCCESS in 27m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57f261eb0869403085a371f4a8f5bb22 : SUCCESS in 18m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2262d4f35a214c65a17f034030853b5b : SUCCESS in 15m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8215b94dbfc405e9b863cbb009ab062 : SUCCESS in 17m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/12bd33bde593422b9f06afeb3d26f094 : SUCCESS in 8m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/677c965c2a254ce1acc5a624fde97bde : SUCCESS in 1h 23m 28s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0874ec758d04de88c5337868a4fa3f7 : SUCCESS in 1h 41m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/792c3afa10e84d859908f928ec2317c5 : SUCCESS in 42m 12s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/649d5643846f4b86ad1f8e416400330a : SUCCESS in 41m 42s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c92455f84ad4db7b58552e8a4caea87 : SUCCESS in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b38ea3bb37dc443ca374e40a34fd4f48 : SUCCESS in 1h 24m 17s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/61b19cb02e1c4fc1aee52ebb89c30fa3 : FAILURE in 2h 22m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/862b617b5a7f402889368434bd77c683 : FAILURE in 1h 04m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f4e0d96c6b0c42fa8bda40c5fb51f353 : FAILURE in 1h 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c131553966d253ba2e70040361a248a8bb060008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 19:03:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cab8c3f6f5ddad800e6989a644f8035804636fa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 21:07:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b18eacaa95e4b9999efba71608dfff2 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22c279e10da4aecb96485cf3ae7b40e : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9c3bf0e4474078ac14176a30e177ad : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0b4221566d44da19f8c2ba779c8f3d2 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384355556ae24d36a10ba0fc01918d53 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5863c4d104be46738ba8af6a12b42b1c : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e377fc99cec490686d27d84019630c3 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/80904508cfce48be813de03e3cab5e0d : SUCCESS in 1h 24m 22s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c7c375a1d4440ce89352ebdb62cc9c5 : FAILURE in 1h 56m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4736b2e53a644806839da805eda6ea5d : SUCCESS in 1h 08m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/954da9c8f8d44333a0579762fbbcd17f : SUCCESS in 48m 25s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/30e046ec25ce48cdb187b727e1ef089e : SUCCESS in 48m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/666171dbf6c044f980db4205590c8c11 : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9d231475f8fa1f89399f6bda758c358fc2cca253","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-25 10:29:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd6e4f5d54643c28f1d0d657f364424e89662fdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93fe09699a3e42f4b91c9a549f3e5a85 : SUCCESS in 7m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2291b721b656445fab59822b23480322 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1c9531f3724f4eb579c6cd47b496cb : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac7201decc7479aaee631849e1a1259 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e8b224b38774dc7912bba6351a9025c : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae07f5b4515487baf85fe2b5cdc2e09 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/411c3054967841eaaafcbcf6f34aeb24 : SUCCESS in 5m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9eee96ed5ea34a2a97dc9cbca14650a1 : SUCCESS in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/884f0fce6ffe44bba64d2f009341af7b : SUCCESS in 12m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56128b141c6c4c4686610e452f2753f5 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b64aa9506c149eb953d867e650c7b40 : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/121d71065ed949ae91fc190049041f55 : SUCCESS in 1h 47m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cbf72862c808461cbcbb4cca63b5c6a8 : SUCCESS in 50m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b42feee81d3b48bd983b92cf7d6746ee : SUCCESS in 45m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f9a43f4296e4fc7ba1a451999cd5555 : SUCCESS in 46m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/160b17dc63c840999e6861e984f7132d : SUCCESS in 1h 13m 52s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba6b2d204e60423d8caefb73dccc790a : SUCCESS in 1h 53m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16fcaf11553f4ad8a850f12c7fc22b9a : FAILURE in 47m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7a77a920e8e54082a9e42a14fe2a50c8 : SUCCESS in 1h 53m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"658d81a741c1b0bf74357bbeb5f115e4ba94b038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:50:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4537138c8a3ac4c866fe960052af66023c3c802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 16:53:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b75ddd243141f9adaf6ae63121f6a3 : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d22c706d02a949c192b489616bbdf25d : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b688bf7cf24e7cb36893b060f40ae5 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f9270e0d12545d2b27c41978db6e99c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7781cff775440319a73780a7affac4f : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d83223c02bbc434d9a79a69ca1c18c54 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78a7d31e05474837b56d316bbd32c970 : SUCCESS in 7m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8997614ccece42409829a8bb296b0d9e : SUCCESS in 1h 21m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c8a2f3b5b76247cca844dccf0e3d0d42 : FAILURE in 1h 46m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/692be9ba06f54715a39e21db82112c30 : SUCCESS in 44m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0b6f207aa41b48fd891015ef6f89c391 : SUCCESS in 42m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ace78e0fdbdc465984b7f6736f5544a1 : SUCCESS in 42m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cacdbbcbc3334059ae84749caf37bb0f : SUCCESS in 1h 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"75726fe9b7ebf29a2596a333cc40123fc1b6f3af","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-11-25 20:07:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e02dc8113d249503cae49ef8ba74af9f809fb7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 22:33:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a2fe6b97a44788a86671f913a35b12 : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/821c7588c2ea483aa5c9adc9a3765063 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289516225de44dbe8e56ca10c14f0ec2 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2430ffa6ae6d48f1ba6f2e45fd366800 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69d625e0f5fd49baaaaeeb71dc769a16 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2c8d0c1a064070942fa3c374e04d99 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f43bac8ff32044b6a6ef7d2d5e804446 : SUCCESS in 5m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/292ff0f0b2cd44f691afa0342d240a16 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e031d5f7156b4d1dba25c1b73c8aa218 : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/506bd82c0c6246309ba51364ce969d58 : SUCCESS in 7m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f8cb244ab5c4f859db38153f9c9f21a : TIMED_OUT in 2h 12m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04d98e0131f34e8fbde2ed48d458f2c7 : SUCCESS in 1h 36m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e0157c929b55412fa57a97aa1f92a6e5 : SUCCESS in 41m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8a75e34addd34be9b4fee49f0d425ea9 : SUCCESS in 39m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7fbcd377c9174703ba680b6cf545c8b2 : SUCCESS in 41m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/183d6eba700c455491767fd3e327984c : SUCCESS in 1h 28m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a64811ae3c8f41ce8293b7a68c4d4d10 : SUCCESS in 2h 15m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05d70be426304871be621d00df6e84bd : FAILURE in 45m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c761a7603fcc43139aae283fa7f2e6b9 : SUCCESS in 2h 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83d894869a8ba287887ac097bed8c6af16952111","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-30 12:21:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d743f1d6fb7c315563351979abb1c10f0ba15b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 14:34:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/560fd394161c4185ab3f4340ac9d1a80 : SUCCESS in 9m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/538b73ed4a6f4f70a8eddcf60f966a5d : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b876552e87445c7aa40af4bc803096e : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dfd6bbbd50040a3837985ef8b83dd18 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c4978e0131048739d1e56e215db5525 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0694f445af6945e6b0fba7ebff176a36 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a8b03c2649d4567b09ce1203689dba5 : SUCCESS in 4m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/662a729bbe2544f59bd0896747e8c568 : SUCCESS in 5m 22s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/159a0c3a76c1465484cf5301ff98ad4f : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e695335ece24fecbfaf168d0ce1c212 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4a3bd7286a547a1b58b88e387f2cb3b : FAILURE in 15m 40s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/569e70e94bdf46019d67892bb1610f12 : SUCCESS in 1h 49m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85683d4f6e4c4cd8adef76de78616d12 : SUCCESS in 44m 14s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8d834f482fae4a55842c03ce007bb211 : SUCCESS in 40m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50ae4875a92046c3881b09125231b294 : FAILURE in 13m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f308755bf1fd4b0d90c494a48901499c : SUCCESS in 1h 19m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/242f355c65234929b8c029970798129e : FAILURE in 13m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fea4e42a7abb42c392b60ca3291c572a : FAILURE in 14m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a304b423a4a34852930d27173c0c8493 : FAILURE in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a9904424bec35f6fb99f1981c5dc452e40c385b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-12-01 17:02:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"afafe88196fc13e7d1b56195810605c94dad4882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 20:35:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8701fcb7c5745db877dc88addc32f4f : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/effe6090c5994554b22b1d237834763d : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790a4b18b7154008a4408dfc48b211ee : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5d4e729dd64d029b782f0207b45bb5 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40a3dc69343348c38bae81cbdfc73189 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ef24d02362405fa6baef2bf13c931f : SUCCESS in 19m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c04363aec8da41628862e0283b2a61c8 : SUCCESS in 6m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1fdd4a106e3b444ba8f4bff55ff16620 : SUCCESS in 5m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/11912103399d4b939335bc27632c2301 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5f415d7730e420ca1156ec80ca7b454 : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/46a42b986e024c1bbb57a18a0de0e1cc : SUCCESS in 1h 26m 03s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dd514ef80ee4abcabff5a0e22056db8 : SUCCESS in 1h 42m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8359e6f4d6f04a7fa2dccd8f3f06e069 : SUCCESS in 45m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d6f3783ec6104e0c88de6a360c7f012b : SUCCESS in 49m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2262194585154110a108e87e183f42d5 : SUCCESS in 41m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e22e0e51554448e8807962b44fa3c9ad : SUCCESS in 1h 50m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f3a6e38ccfe3439abc8702bc93942827 : SUCCESS in 2h 09m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/13a2f1daf564486fa1ee5dee242109ac : FAILURE in 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d5df876b1bb948abacc2bf5a35f59392 : SUCCESS in 1h 57m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db1501ded6c50858c23351b005f2aa91ab46c390","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-12-03 14:02:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"899fac552cddcaa6ad0cb5af6e7da477b6b2a9d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 14:02:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6307ece8d79251467becf89c43d1edc92013fe5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:54:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abc5dab0dee54e8fb5294776f75d3d85 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cc61411262143cbadf7a90317563855 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e83da745dcb0440782f3c6221a13e48b : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bc12cfcd8ff4cfb9e01e21f98b6c97e : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18fe83db4f744e6ab141bff3a32afda7 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07ff729f5c4f4b1d8cf5063f68b244a9 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c24b4eddb70e4f12ac732dfc3f422459 : SUCCESS in 7m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8415e8c0576a446a9eb4a35f8d68faa1 : SUCCESS in 1h 17m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a22e12097244e61b78c9732a2721d2a : SUCCESS in 1h 59m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ec7cbfc01634cca96d6ed62fd0d0bf8 : SUCCESS in 54m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e09b4873e31e4b62812eb47dc412d2a8 : SUCCESS in 1h 16m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a764216d05cc43ef94b533f52c68d203 : SUCCESS in 44m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2831899a2da8467cb2dc9bf5fc1b9ede : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f658c7e99ab748ad23eeeae5451437eec4f516c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:54:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5dc2097469e0a59abd0eb324568758c1508b1841","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 20:56:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f6fbc5c89d4bcf874ecca8c7573394 : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777dc056977c4385ab10c4b3487601c1 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a258e108469f55a089600afeee1c666d6a647acf","revisions":{"ebf190819e109e6a9eaac86cd80092ae34aa44a0":{"kind":"REWORK","_number":1,"created":"2020-10-08 17:31:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/03/756903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/756903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/756903/1"}}},"commit":{"parents":[{"commit":"d653edde2d7827e6d5d21c6046b64df8f9bb346d","subject":"Merge \"Ignore DELETED amphorae when performing certificate rotation\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d653edde2d7827e6d5d21c6046b64df8f9bb346d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-07 22:59:50.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-08 17:31:40.000000000","tz":0},"subject":"Fix load balancers with failed amphora failover","message":"Fix load balancers with failed amphora failover\n\nThere was a bug in the load balancer failover code for the amphora\nv1 driver where if there were multiple failed amphora, with session\npersistence, could cause a load balancer failover to not complete.\n\nChange-Id: Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2\nStory: 2008099\nTask: 40802\n(cherry picked from commit 77dc23fc26cbb8c6488b5205c63bf7cc847a9c32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebf190819e109e6a9eaac86cd80092ae34aa44a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ebf190819e109e6a9eaac86cd80092ae34aa44a0"}]},"branch":"refs/heads/stable/ussuri"},"a258e108469f55a089600afeee1c666d6a647acf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-01 17:02:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/756903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/756903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/756903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/756903/2"}}},"commit":{"parents":[{"commit":"214c386277254131b6174f5822437c0374be1d1f","subject":"Merge \"Remove unnecessary joinedload\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/214c386277254131b6174f5822437c0374be1d1f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-07 22:59:50.000000000","tz":-420},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-12-01 17:02:44.000000000","tz":0},"subject":"Fix load balancers with failed amphora failover","message":"Fix load balancers with failed amphora failover\n\nThere was a bug in the load balancer failover code for the amphora\nv1 driver where if there were multiple failed amphora, with session\npersistence, could cause a load balancer failover to not complete.\n\nChange-Id: Ia46a05ab9fdc97ed9be699e5b2ae90daca3ab9a2\nStory: 2008099\nTask: 40802\n(cherry picked from commit 77dc23fc26cbb8c6488b5205c63bf7cc847a9c32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a258e108469f55a089600afeee1c666d6a647acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a258e108469f55a089600afeee1c666d6a647acf"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
