)]}'
{"id":"openstack%2Foctavia~757548","triplet_id":"openstack%2Foctavia~master~I98b31be1b1c0f61157831145ef9d19d481e2e0b0","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I98b31be1b1c0f61157831145ef9d19d481e2e0b0","subject":"Update amphora status on health update","status":"ABANDONED","created":"2020-10-12 12:13:20.000000000","updated":"2021-06-11 09:56:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"405e569bf509d80bc93d851860c0b4d96fa9e5aa","_number":757548,"virtual_id_number":757548,"owner":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-14 15:28:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-11-17 13:59:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2020-10-14 17:43:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-10-14 15:48:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 15:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-14 17:43:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-17 13:59:19.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"1ea8428b8dbe08d3ef6981d33278d73e759c206b","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-12 12:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2bac9a1deb44ae38a284d25ab012306083f465f","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-12 13:23:04.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"1969f284bb8453d9b2101d35f111f747ea08face","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 18:08:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7070a7fab59457d8b6db34c70a268f2 : SUCCESS in 9m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed8bfbb4fabc41089841626b49ffe7d9 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7254d7521e647c6a265a5d69cc553d5 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3d29bbd2c23400884a4d1d8e255b105 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df13b3c49154bfa859af96b59ebc522 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc4e6280123461c9934177a27974a7d : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ebcf9df059c49c1b01f376f9f9e6c4a : SUCCESS in 8m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/965c338e7ca844bbbeb87c32ea4cb37c : SUCCESS in 7m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3747d7c49425436ab336fc2fb8273ac2 : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3fb4aa6112eb41b8bb738c7a456064be : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d05e3116cd04270b6d5d35633df6c76 : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c8c53e5f05d43aaa592c7995f5a5452 : SUCCESS in 1h 39m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cafe4041476540559fbb0ca4f378f47f : TIMED_OUT in 2h 34m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6da166ac92ee4be8be19b60a1591edf0 : SUCCESS in 37m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/996966941724407199cc189c4ec8a688 : TIMED_OUT in 4h 01m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0cfbb6434a1b4b48be39cd497c8f7501 : FAILURE in 41m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2aefe9ea517048f09a4b3b1506d42e5c : SUCCESS in 1h 00m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ac3685385ac542e49ae8e4f62b087d08 : TIMED_OUT in 2h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6695e88cd53b451da27a1112d004bd83 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/999b60a0818a412fa5d6dd9142521edf : TIMED_OUT in 2h 34m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a8c76cee504f49e5a77b28c8d34b5ae5 : TIMED_OUT in 2h 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/25958499c29645f18316aa39b2ccba89 : TIMED_OUT in 2h 34m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d5933474ffbb4863a8d45337249b80f8 : TIMED_OUT in 2h 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2038d02abdc6e88d1485af75febcf7104a5f0be4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-13 07:42:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bfdae8be7dab4dd2f48011f1d61dfd929ccf611a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 11:25:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/433b51bd4de84906acc3e4f485966c66 : SUCCESS in 11m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea4cccfc978f40119a0e9cdc6d7ab28f : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b9141c0bf094280b63c48a32da3c9e3 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04d9597fd498444483bf9a1fdbab7835 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b173c0f33e64e4fb7b298a0a13db02f : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587350eb999a45d9bd0c7d995e40f22d : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c3da15f71cf4c9581f2000d7ae7a260 : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/599739ac9a304fe4ab01e9fe0eb23c4b : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d157dfa539e84441ab2d56fe6d3000fa : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4dee247047cd4e49aadc96c24a5cb7b0 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4e4cf20711e441f1b498595abec2a109 : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a6664b0838694c7ba5a8125301955307 : FAILURE in 23m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/503d03fbaf9043f5adbce25575a95f50 : TIMED_OUT in 2h 35m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93a56feef6a3496bb19b8e05f304be94 : FAILURE in 24m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/41c6a108fcb042f0aa67660a3e7d5086 : FAILURE in 22m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f26fec5b73b346c09103186593525612 : FAILURE in 1h 03m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9cdbcd615b443c18cdcaaac143ad6d2 : SUCCESS in 1h 02m 18s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/81cea90335c346a79bc2b07b59f854d1 : FAILURE in 24m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0782e479aaf84ac5baf7c03fc0b494cd : FAILURE in 23m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a6ae0447b3394c0ba0cf9482865082c0 : FAILURE in 26m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/50c785ddfaa14d599591c8e928d0a755 : FAILURE in 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e2315b97a4f243a3b7fe7c8e05725005 : FAILURE in 23m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bd3192adfe0a4ee284f672ae46916697 : FAILURE in 25m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ebe6302a604563a977a0af863df76f93d4e7386","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-14 06:23:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84155c4338651dcab7bb49cfac3e61ddc46f9d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 09:54:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1149e06395814aeda582d726aeab626c : SUCCESS in 11m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4708af46c6b47fda50ef5f4bbaf1d66 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27535e7dd206474cac39be884370036a : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c80b4ae700164951a18963c4b22df35c : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d03b5ffb7c4ae8b86cd329df154de9 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc83bfdd257469793ca916c3ff89a4f : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae85c22304fb4e11b1cc1267c6e795f2 : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d2f4ac2e48b54e9683d15eff761f0de9 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/297875a7a70a494b9face742611dfe4f : SUCCESS in 9m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2bed1f21235d4e20896dba91d13ea85e : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b46af244438546b9822258c3c5a36d74 : SUCCESS in 8m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c85d4e7feb04bdd8e32525d0a9bb670 : SUCCESS in 1h 37m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b17eb3effd24d519e4db9e3603dcf09 : TIMED_OUT in 2h 33m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/da717e5dbc9645ada7fbe56b667dd186 : SUCCESS in 44m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f1a935ba3d944820b459b1f94e06d0d6 : SUCCESS in 40m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43699147ebff4cbdb1f8f1b786816400 : FAILURE in 1h 19m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/41b871fe9ad44f3b9b28b82c11b8d66c : SUCCESS in 59m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5068165fb75145dcbc32d3c67abfb333 : TIMED_OUT in 2h 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8f929dcf04514d78b8541cbd8ba5e10a : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/74113d5b100d4be8a831fadcef80c953 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b3d545e338954e7b8e7f349786e3e5f0 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4fbc8c6a1a5a48858fe5427a90df10c9 : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b6258aba58134bb7ace47d253d929bf2 : TIMED_OUT in 2h 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3829081fee35f539753533a2dffcc68d5d3d211c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-14 10:08:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f841d1fc4bbe308aceb336ab517fea3eec4cacb5","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-14 12:14:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3b109e2348d3d8ca1bd8683bb43ccbf5cc81f6b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:28:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51bf3e0b1f654c2a989d4ab87aa4fafb : SUCCESS in 9m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e3b5f2ddc84d309ef9e40b22746f24 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c750115af341fbacc64f475487ada4 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f457aa10b4044ca84515ddb0a0191f0 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b15356babd84cbeb8ea6ba370bd72f5 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a74ca0fcd9eb49cc8249dd454b7aab1b : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83b9e2ec26bb496da54ef5ed13fafde1 : SUCCESS in 5m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/df0b0601cc1747a98a7d8d0f8f620d38 : SUCCESS in 5m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fde19464a4324f1d8b055fd3b4ea41c6 : SUCCESS in 7m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a59a9338b69b43b5ac8558dab15a24cf : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0877793d9b1a48ca9ec5ccea6adb5f35 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1dd033286314a44955098ca30569010 : SUCCESS in 1h 46m 18s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/930bb2c33d0b48679015301b51d1d32e : SUCCESS in 2h 10m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/623c0a044bcb44d5aa3b64c0d3e4398b : SUCCESS in 44m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/49e808e7989141769d816def5096ed14 : SUCCESS in 42m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/acd01d40f62d4e1faa0deb72375101a9 : SUCCESS in 43m 59s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4dea74d97b6740fa8a73fc2acd12e2a6 : SUCCESS in 50m 10s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ff989b4be3a0404ba2663955e7ff7dcd : SUCCESS in 1h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a0c104448f024a4689416b5572622ae1 : SUCCESS in 2h 25m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/80d6d90c573f49b2908a01a2328826ae : TIMED_OUT in 2h 36m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6b7b9f91e6144ddea58be91a91aa2271 : FAILURE in 2h 12m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/47e0f2d1c50a432f9c1ea11c4641f69c : TIMED_OUT in 2h 36m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/59001eb85ca6419c8f66836fc32a6812 : TIMED_OUT in 2h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7906d449267616bbfcb3ac2864a2e3059586ddd6","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-10-14 15:48:18.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"c18760a9a7253631d179b0617b9ce771638789c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 17:43:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e54fc6d38b4c6d6f9a7897daa9d26a34c274ed2d","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2020-11-17 13:59:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"39adf72bb2a24b7693bd30f31027e811f7f4c1a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-08 12:23:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"c9033f1e5caf6ad8138e1178511a799941898f26","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-06-11 02:29:32.000000000","message":"Patch Set 3:\n\nHi Ann, is this not necessary any more or do you just not have time to work on it?  I could get it re-based if it\u0027s still necessary.","accounts_in_message":[],"_revision_number":3},{"id":"405e569bf509d80bc93d851860c0b4d96fa9e5aa","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-06-11 09:56:50.000000000","message":"Patch Set 3:\n\nI don\u0027t have time to work on it and it requires some additional changes(amphora can be ALLOCATED), not just rebase.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"60abe10e833395fa424988a9af6529f34a1683a1","revisions":{"cee20641b6ebe7215c7eb1f636245cf9356ae18c":{"kind":"REWORK","_number":1,"created":"2020-10-12 12:13:20.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/48/757548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/757548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/757548/1"}}},"commit":{"parents":[{"commit":"77dc23fc26cbb8c6488b5205c63bf7cc847a9c32","subject":"Fix load balancers with failed amphora failover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77dc23fc26cbb8c6488b5205c63bf7cc847a9c32"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-12 12:06:57.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-12 12:11:35.000000000","tz":240},"subject":"Update amphora status on health update","message":"Update amphora status on health update\n\nWhen update amphora health, if amphora status is not ACTIVE -\nupdate status to ACTIVE to align amphora\u0027s state.\n\nChange-Id: I98b31be1b1c0f61157831145ef9d19d481e2e0b0\nStory: 2008248\nTask: 41095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cee20641b6ebe7215c7eb1f636245cf9356ae18c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cee20641b6ebe7215c7eb1f636245cf9356ae18c"}]},"branch":"refs/heads/master"},"8b72d0dc70ff83973690ceb9aa10f2442858764b":{"kind":"REWORK","_number":2,"created":"2020-10-14 10:08:45.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/48/757548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/757548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/757548/2"}}},"commit":{"parents":[{"commit":"77dc23fc26cbb8c6488b5205c63bf7cc847a9c32","subject":"Fix load balancers with failed amphora failover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77dc23fc26cbb8c6488b5205c63bf7cc847a9c32"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-12 12:06:57.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-14 10:08:27.000000000","tz":240},"subject":"Update amphora status on health update","message":"Update amphora status on health update\n\nWhen update amphora health, if amphora status is not ACTIVE -\nupdate status to ACTIVE to align amphora\u0027s state.\n\nChange-Id: I98b31be1b1c0f61157831145ef9d19d481e2e0b0\nStory: 2008248\nTask: 41095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b72d0dc70ff83973690ceb9aa10f2442858764b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b72d0dc70ff83973690ceb9aa10f2442858764b"}]},"branch":"refs/heads/master"},"60abe10e833395fa424988a9af6529f34a1683a1":{"kind":"REWORK","_number":3,"created":"2020-10-14 12:14:30.000000000","uploader":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"ref":"refs/changes/48/757548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/48/757548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/48/757548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/48/757548/3"}}},"commit":{"parents":[{"commit":"77dc23fc26cbb8c6488b5205c63bf7cc847a9c32","subject":"Fix load balancers with failed amphora failover","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77dc23fc26cbb8c6488b5205c63bf7cc847a9c32"}]}],"author":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-12 12:06:57.000000000","tz":240},"committer":{"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","date":"2020-10-14 12:14:15.000000000","tz":240},"subject":"Update amphora status on health update","message":"Update amphora status on health update\n\nWhen update amphora health, if amphora status is not ACTIVE -\nupdate status to ACTIVE to align amphora\u0027s state.\n\nChange-Id: I98b31be1b1c0f61157831145ef9d19d481e2e0b0\nStory: 2008248\nTask: 41095\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60abe10e833395fa424988a9af6529f34a1683a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60abe10e833395fa424988a9af6529f34a1683a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
