)]}'
{"id":"openstack%2Foctavia~757841","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I9283d9804feb83a1d5a160da48da6146b19da88c","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"I9283d9804feb83a1d5a160da48da6146b19da88c","subject":"Fix pool delete race on load balancer cascade delete","status":"MERGED","created":"2020-10-13 11:43:02.000000000","updated":"2020-10-22 11:15:12.000000000","submitted":"2020-10-22 11:08:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757841-1603364918897-2212603e","meta_rev_id":"35acd3818ea500df24903c785906b22753d652b5","_number":757841,"virtual_id_number":757841,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-10-21 18:14:57.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 11:08:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-10-19 18:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-10-19 22:19:00.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2020-10-19 22:19:00.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 08:00:31.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-19 22:19:00.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-10-22 11:08:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"90f755a14f089ddd7ec5e9db8340893df927d33e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-13 11:43:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"021e68b62512888f524be6909a7debe1d374a502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:38:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc2395aaf5349ddbd87fc6e35583926 : SUCCESS in 8m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e4f517f7ba148b4a8e0bb6ef521ecc0 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1413ed9fbfd1477d951b6a7e1e72f685 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a4f4e644fec43bc84ac965cc57670dd : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7177e772b73442e9e935e75ef3d1dab : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53b1f43a2c4e422ab1f2e0b284902eb7 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16bef62821834c86a0a03ddb69fd209f : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3f6351d5bc4d435097e67e3d2e678e90 : SUCCESS in 6m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/162e5f8910b24bbc90857a4d16848f80 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/81a456a2944341f9bff64e48129bfa49 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5a82bd0091e4d778bf22d9bfc142dc4 : SUCCESS in 6m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c023946354a48fd8bc997cffdb89b8f : SUCCESS in 1h 45m 05s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/c51abcada521417489b87cde4fafd85e : SUCCESS in 2h 11m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41a17cd374d141228712f28add325441 : SUCCESS in 2h 10m 39s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0fa29fb38de84ab6b07c2fc1206d9d4a : TIMED_OUT in 2h 34m 49s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/21697299129b43b6a51be7829fe77433 : FAILURE in 14m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8c718ea47fb34fb2b538b1c02bf8bdee : FAILURE in 30m 53s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ca93541f34c4d878585862dab40b501 : FAILURE in 48m 47s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ed8c8cb43e314c7dae4b2f73ba38db3c : SUCCESS in 1h 31m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bed68b4efb35448a911b09be42b8d7e4 : SUCCESS in 45m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/bf62c204c2b84704a9add208b6c51b8f : SUCCESS in 2h 00m 20s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c00c12d85f734480817fd957a81cb0d9 : SUCCESS in 56m 12s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/543edfb602ba4a39ab097e906152b85c : SUCCESS in 49m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5a699a1ab26d4b8f946f128d2dee6578 : FAILURE in 9m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50e58f41ebbaf0d3e6257d0961af658daf8ebb1d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-15 08:00:31.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb9067dc02e29611c42e719a43677df615cfa786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 10:41:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9620f98c5b7d4f21b43bda8830a5c4a7 : SUCCESS in 9m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13606f8422d046938c46717b362d61e5 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ca782bfdfb4fb7922229caa6804a90 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/969692221960499a960155105896e667 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aefcfb23dc548c3b36e9940ede7de51 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28674f9dce8e437c82a63ccfcc53a91c : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4476f2a26141b38332650b5a83a257 : SUCCESS in 13m 01s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/90f618fe12a949198fd17794c85ce123 : SUCCESS in 7m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ecb90fbcc4fd4e49acac6e6263c3a9bf : SUCCESS in 7m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a54171b7a74a4d29a09b83b03b040512 : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/184f81b7da2e4bd58204e4073e870a1e : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4b212144be241ddad0af82bbb734514 : SUCCESS in 1h 39m 09s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/253f64e5a48d48b9acd716f90f3af84a : SUCCESS in 2h 01m 00s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a855972518114dcd9323d5933fceca4d : SUCCESS in 2h 18m 16s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/0c5f4df02f724652afe0e89bddbf9071 : TIMED_OUT in 2h 34m 35s\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/cae3fe51112248509e518c96f5426ee0 : FAILURE in 10m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edee20408fbf46af9375a4140e5bfcee : FAILURE in 29m 13s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e0035cf67a04ee4b6c70eabf5982c46 : FAILURE in 52m 07s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c4f1ee9edaed430386d3456320ddf057 : FAILURE in 44m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2739261af3f44a6b874ae289cefe3bca : FAILURE in 1h 12m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/940431bd8b3b42e3a695a3b727532470 : TIMED_OUT in 2h 33m 58s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dafe1c01dab247f69dd30aadf57b9d7c : SUCCESS in 45m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/b06ce00378d74d0a8de74e1d27dc5050 : SUCCESS in 35m 33s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/79314dbf0c7c42328b478aed8349d8b4 : FAILURE in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"579b266179e279ed9fa0e84671baa3e03ff25dd3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 18:34:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b4fd347bb333d8bb50466ce2bd885d01879a8912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:59:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5c97bbe331e407495332f3b64f9c05d : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff0275e533484f97a4afae0c748b85ec : SUCCESS in 12m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99dd94383894b67bbcd4b8c12481b29 : SUCCESS in 9m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e59aab9d1bb544d1aacdaa4156c6a2c0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dedcb0938a584a8eae66e66df127ed7f : SUCCESS in 6m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e40980eb055d47c7950072f83c64d249 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef5056ee53a44af9310174f6096ffe8 : SUCCESS in 14m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6f5f88e9cee3438d89e567eda268251a : SUCCESS in 9m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/35e93facbff64135919fdeec41c0463d : SUCCESS in 9m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a5e66a0e2fe3464eb5c273894ab33ea9 : SUCCESS in 11m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef55d5b8e63f41b5a90047fa8e727beb : SUCCESS in 11m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d547dcfd5f44a5cb13bf0a29054c837 : SUCCESS in 1h 33m 19s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/70c5252d2ee44bb09c4cd897870bfd1c : SUCCESS in 1h 37m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ddb542c50b074492a6d0036f46808846 : SUCCESS in 1h 50m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/90c7b73037d54ae2a563e69f98b043e2 : FAILURE in 1h 37m 38s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9e10a484216b4541bf39c04abb09a19e : FAILURE in 12m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2ef075f4a6824000a472a73f3fcaab6e : FAILURE in 29m 04s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/0ca4091c5f2948ccaf32c9a2e9ffb155 : FAILURE in 48m 30s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a0a175aab6e45159812ca68f52af56a : SUCCESS in 1h 20m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/15f209ea48c640f59f1e73c21266da78 : SUCCESS in 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/74747e5f2e6443d590594bec135ea222 : SUCCESS in 2h 06m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd7bbe83e714402ba933a471cde0e597 : SUCCESS in 47m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/69ad0ccde9fd41ccab17cc023e3f56e3 : SUCCESS in 45m 48s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8d8699f09b5f46e5a72c5bbac2340548 : FAILURE in 16m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8a371bed7ae0bdbf870246b2ac4a0f984bbbca9","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-10-19 22:19:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb990a902ba1910cd2a0523705b102d2fb6850f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 11:56:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc0543ac6b9908d1922506106dbe23ce771bbffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:55:26.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"748402092a34635aa54a9f2b04c51ac0e2730065","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-21 18:14:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"db1482866932b04a466e9437aad98185f393934c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 21:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/228ebc7b52564b728d5ccfa44b0434db : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/766fcdc315344d7098d727d67283183d : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbe40fb54b244b6b4ae1d132e5a0d36 : SUCCESS in 7m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76104c1da7a24d029d2127c1859170ef : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73a0367fbc2145a69c74c7886993d9e4 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3eb6ba9c03e54c128a5c1da84b7009fa : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45bfaa5351a4079b4cd837b1a0f4b45 : SUCCESS in 12m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e1d6e3cec5f44ccab91899dd72b9adac : SUCCESS in 7m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/405836ce71fc427d96a3690e234f20d4 : SUCCESS in 9m 10s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8bdd0c3b663545b982a3ca71b90b7f86 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82eca6cf623b4107b559bfd800ce4f7e : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ae77b01c49b84f5289b73843def84e09 : SUCCESS in 1h 45m 43s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/49234ad8ef1a40af81aaa287024f918d : SUCCESS in 1h 38m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/116f981ce54d4ec7b22d50a91b0fb992 : SUCCESS in 2h 11m 06s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d81b934550774ad6bd52ec93e426ecca : SUCCESS in 2h 21m 44s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/a6972c78f91c42c2be8b00808a05872a : FAILURE in 15m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b39b3ebcd184281b07e2524ef33ba27 : FAILURE in 33m 20s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5ccb2a29e2149a0b58970580b6af171 : FAILURE in 53m 33s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f713eba0278455986df430bb6ccfa94 : SUCCESS in 1h 43m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3736bf37b85c42769c59c5d317272059 : SUCCESS in 41m 23s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/57ead4572fc94665962a0484f9de622f : SUCCESS in 2h 06m 08s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74652ec5f0fc4952a605d493530c57f7 : SUCCESS in 45m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/d589dda1dbc54d9b908ebeb25e6aac8c : SUCCESS in 47m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6d025bcde94240efaa092dadc51d428a : FAILURE in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6049d9cf27ed62febe70c629aba6baaf0181e00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 08:23:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc33e534350ab5b44fd5a2d01b7f194ebce79983","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:08:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"21b39082536d8eb912b68daec350e19a5f5012ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:08:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69994b517fff4c8cb3200d4cffaf39e1 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d1edde190443b397a378fe289f30f7 : SUCCESS in 8m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b64a03194af248ddbef694c4ac3abc89 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc053784f1be489cae0f5f939135bb9b : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/118cfe045051419cb0aec98a987efbca : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d1bd6900cd4450b1c06a5ac20a6489 : SUCCESS in 13m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fcecc8c65e914f8bb133cad85b3818fa : SUCCESS in 10m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b3107a3578c42f38efdbb8f2b864a96 : SUCCESS in 9m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2910b88326494a34ac957c888cdafbe5 : SUCCESS in 1h 55m 30s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/43c4885ac6a640f5ab5ed9f117bb6cec : SUCCESS in 1h 48m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/052337698a1244d8a756e8d379e1df5c : SUCCESS in 2h 32m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79a534364a9b4ae5854162def8145b75 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":2},{"id":"35acd3818ea500df24903c785906b22753d652b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:15:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b142c8b6f2f44d3bcd21653b2d35ce3 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11f9694453eea6f79f541c4d26e159de1fdc7279","revisions":{"84260d3667329fabe7687afc3e8f84e276c63548":{"kind":"REWORK","_number":1,"created":"2020-10-13 11:43:02.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/757841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/757841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/757841/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-12 15:24:14.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-13 11:41:14.000000000","tz":120},"subject":"Fix pool delete race on load balancer cascade delete","message":"Fix pool delete race on load balancer cascade delete\n\nIn what appears to be a race condition, load balancer cascade delete can\nfail while deleting numerous pools. The load balancer can still be\ndeleted with one or more follow-up cascade delete API calls but this is\nsuboptimal.\n\nPer local testing, the ValueError exception is harmless so we can just\nignore it and continue the pool delete flow, with no orphan resources\nleft behind resulting from the exception skip.\n\nTask: 41096\nStory: 2008249\n\nChange-Id: I9283d9804feb83a1d5a160da48da6146b19da88c\n(cherry picked from commit 332791ee2b3cf29d01d94a9bb53fda4043303fd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84260d3667329fabe7687afc3e8f84e276c63548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84260d3667329fabe7687afc3e8f84e276c63548"}]},"branch":"refs/heads/stable/train"},"11f9694453eea6f79f541c4d26e159de1fdc7279":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-10-19 18:34:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/41/757841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/41/757841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/41/757841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/41/757841/2"}}},"commit":{"parents":[{"commit":"f0dab4ce75350657a6846378835bbdcf6950686c","subject":"Temporarily set octavia-v2-dsvm-py2-scenario non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0dab4ce75350657a6846378835bbdcf6950686c"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-12 15:24:14.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 18:33:36.000000000","tz":120},"subject":"Fix pool delete race on load balancer cascade delete","message":"Fix pool delete race on load balancer cascade delete\n\nIn what appears to be a race condition, load balancer cascade delete can\nfail while deleting numerous pools. The load balancer can still be\ndeleted with one or more follow-up cascade delete API calls but this is\nsuboptimal.\n\nPer local testing, the ValueError exception is harmless so we can just\nignore it and continue the pool delete flow, with no orphan resources\nleft behind resulting from the exception skip.\n\nTask: 41096\nStory: 2008249\n\nChange-Id: I9283d9804feb83a1d5a160da48da6146b19da88c\n(cherry picked from commit 332791ee2b3cf29d01d94a9bb53fda4043303fd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11f9694453eea6f79f541c4d26e159de1fdc7279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/11f9694453eea6f79f541c4d26e159de1fdc7279"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
