)]}'
{"id":"openstack%2Foctavia~758855","triplet_id":"openstack%2Foctavia~stable%2Ftrain~Ifc53049e7135e2623bcf7bc2287262ba2f70382d","project":"openstack/octavia","branch":"stable/train","hashtags":[],"change_id":"Ifc53049e7135e2623bcf7bc2287262ba2f70382d","subject":"Temporarily set octavia-v2-dsvm-py2-scenario non-voting","status":"MERGED","created":"2020-10-19 18:34:08.000000000","updated":"2020-10-22 11:13:26.000000000","submitted":"2020-10-22 11:08:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758855-1603364917739-271b27e6","meta_rev_id":"c0e2b3903c97fcdb3162eec8bcae238cac25529a","_number":758855,"virtual_id_number":758855,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"date":"2020-10-22 05:47:50.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 11:08:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-10-21 11:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2020-10-19 22:19:43.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-10-21 11:55:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 22:19:43.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-10-21 11:55:44.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-10-22 11:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4732e86de5c0583c6168357266a522236c6d065a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-19 18:34:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f503e8094610c7ffb11a9c39f03028f873702634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/669cea56dbd649ccb27f46d06fbb46b9 : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a13975568d14647b84c1ebe4e1ef9d4 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bed935a36a14067bc68ba19d9b15629 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/371770bc1c374300a1f6c02b5ba9386a : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf9604cf4f6c421dbf94bcb14469f727 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ae06c0f9d534e01a9d608e826a00a85 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802ca261d0a14b67a81d8bc236beea3b : SUCCESS in 14m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27a2451367ce4cfdb004dfed484b83e1 : SUCCESS in 8m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bf6a1a8c4a7b45efb58ca827009ff1ba : SUCCESS in 9m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/247ea61082c449fa8b7d34c0b3a8bc9f : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29c422f2e4c040db94aa98ba797d9d4d : SUCCESS in 9m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1a43fedd9af849a7bfd1e37af5ab2097 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/5f6faf410a5244918ae61e3885aa8122 : SUCCESS in 1h 44m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43f072215672431ebfdaadab0937787e : SUCCESS in 2h 00m 48s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/1b07640e28344bf3842bc8c149c5c4e8 : FAILURE in 1h 29m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/6f4ca6b710c44b089a40314fe08660e4 : FAILURE in 14m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fc0e030a9a1a471392ef70cb6cf12215 : FAILURE in 38m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/15aab1894d6449c9933f74222c3f72a0 : FAILURE in 50m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7f2ca0314644fd0b5261e882aeda4f7 : SUCCESS in 1h 36m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b271f7e30604ee38905b0677dfd3b83 : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e7bd18b00f42498290a262b30b19fe7e : FAILURE in 1h 45m 27s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9671cd837a44317a90a41a71cc48246 : SUCCESS in 45m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/25370d594f2d4b42bab316815d5810f2 : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/060fa5b196eb4c63ab563b07792e2a36 : FAILURE in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1053d876076c6e5b5129956360335cb3eb60465","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-10-19 22:19:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d69d01bd688b63fcd18c3cda1d833d44bd17c03","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-10-21 11:55:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9748404d6e3dff7a519728df1e06452760fcd9e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 11:56:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1e8d1e854b6faf9cb876e5437470ca8e98cf594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 16:55:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aeb4f88c37c34fe8b73a8a937a77e1d5 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04df5a2fe4a64c10887f996e9053e721 : SUCCESS in 9m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ccecf3643304b3c8f76dbf6f5969824 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e55fcb110dda41f29c5aef8ea809eabe : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/289a476a8e8f4c7b91ca2e3aaafc026f : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68345dfeeb204af49a1f2f8d710b91c7 : SUCCESS in 15m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/409da111981c428d923d4eb868e34f10 : SUCCESS in 9m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aad6e9bdd86e433bbacab162592e0bba : SUCCESS in 12m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a2ad996608bc452bb697f35690da7a7b : SUCCESS in 1h 51m 31s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/42b2cd0b46ca4dee83c62135e782510f : SUCCESS in 1h 55m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6150aba3aba74caf869509053d5e2cc7 : TIMED_OUT in 2h 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b60b0f92911146c9974077dfcf400160 : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8f2d4cc032042181dee0d5970968992a849a1575","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-21 18:14:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"076917a8d86941c558b8dc4f7439e91736a322d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 22:55:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05348a2337de4d2aa10c71c94234f496 : SUCCESS in 10m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1638a6ca03f4770b7b21166635be5d8 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df97c02b1b1b452ab4e26896d3d99239 : SUCCESS in 8m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2f52681ac384aa790e7b3278a5c1b8d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06d49ca83b9349f9835bf9ce351f8b2a : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d849a37074fe45e08c5a37f87892a73a : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77fbd309cae45a48a4bec288513d841 : SUCCESS in 11m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/76480e83090b4959924e7be4dfd1cf42 : SUCCESS in 7m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/61656800ebf94640a7454db9eece08dc : SUCCESS in 9m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c9c56fda961441c1afaaae7d78838615 : SUCCESS in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f8150033868e454aaa8b3177692998c4 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22d7304384784682aed68417fbf83682 : SUCCESS in 1h 46m 20s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/fbab7ef96e1a4f26925f7fa0e1513e04 : TIMED_OUT in 2h 15m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf55c136a4cb4d868752ea02192a92d1 : SUCCESS in 2h 13m 55s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/4790c7f67c724aa8bcd1ccee840de6f6 : FAILURE in 2h 20m 10s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/558e3025be5341c4900f7f4db4e254cd : FAILURE in 17m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d478583f17c2473bb7c6dad90d02ed67 : FAILURE in 29m 51s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/2297c815a3f344929a5d50a5b1607167 : FAILURE in 58m 20s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/00f671de7d854c63b8b13d5261d7d7f1 : SUCCESS in 1h 32m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64513c70abe04212b949248e3ab3beba : SUCCESS in 47m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/798417632166495280c695f5ea26100c : SUCCESS in 2h 11m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4daab6e99302413592d6e990b106b835 : SUCCESS in 1h 04m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e8549d1619e24e19b2a263b0d9188561 : SUCCESS in 47m 29s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/9b33e06df9774d5f9b88f44347790711 : FAILURE in 14m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"271b3ecbc3ccb2ea5c451a378ac6289dc9a1dcc3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 05:47:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe70176ef68679f0cbb56a432cb4790e17ce6818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 08:23:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d461d5a4e73a46a6ab07271375e26dfc : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f2c985a537b473c85451484257df167 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c443bce31774b7ea60aeda24ee4eede : SUCCESS in 7m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8a7f2bb55b444afb1b8fc4beada15cf : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a97bbfdc5bc4e6a8fb922cb275626f6 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/367c61d1b66647a9ad3cc9a8d67aef71 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cc201f3c9a41e585d2a3452d5504c5 : SUCCESS in 11m 30s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ffa9b366d6304089b003dac6f0bff8dc : SUCCESS in 5m 42s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/78c0180b2bc44da3aebd392388da2cdb : SUCCESS in 8m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9d75ca1f6e2042ddae976ab9bf740aab : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65f2c39880b543b29d932741a833e6e8 : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73c640f12d8e464a93eeeb6c4f5c3344 : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/86b32eff52f549f6a2ebb2b5d3b1b82d : SUCCESS in 1h 57m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/455b5cea513443deb2ba43ec0ccc722f : SUCCESS in 1h 52m 27s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/f2d6e854329e4a6b89a7dc861e0a5076 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f71d4559173e415982c25d4e14492ea9 : FAILURE in 10m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5439c9d6d2546b2a4cdb68c3767810d : FAILURE in 29m 31s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e83445eedc446b29e0d19326b7ec36e : FAILURE in 49m 41s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4aee7f90f32541319698aeda48bd15a7 : SUCCESS in 1h 39m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3cef85b99c04605880c39001104719f : SUCCESS in 48m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b0589d346b24d5d9bbbeddcc4b8a1d9 : SUCCESS in 2h 01m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17dac68863fe4b37afee1c866288b751 : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/c85c98347bec453888ce68cc4e9fbb39 : SUCCESS in 1h 06m 41s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/179f4309106e4897a373c532c6c044d4 : FAILURE in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7c218e8d4f661425a02459e5218439f830c52d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 08:23:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc65cf89f578a85ecb54c480e43d0a1495fda1d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"94f030db4dba25db2f8dd921a8f5030d627d62fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:08:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c057d395c3a541d3bea57cdc5499d1e7 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2baca917e2b242e492977ff9f96ea429 : SUCCESS in 9m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3b0e8bef78e45a4bf16c1994c8a76e4 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26481b38298f42558eb9f42278f5c3f8 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3c5cbce37fb4a2489ee8b439020a0ec : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5efe0a3c2214403693ad3958586e9d78 : SUCCESS in 12m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f45d42f3b0b64ac1bd6ee969317bf9e7 : SUCCESS in 9m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/37114165614a4b7f84080227a4f87c70 : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0658b37eb8564f4fbc4e3513cd943c18 : SUCCESS in 1h 50m 51s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/83ee6ea0a0c44f7988d7fb56af1f1649 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/243bede949394a12aa38dad0e437f6e0 : SUCCESS in 1h 53m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c00c4262c75426ab7ec0e750d3ca3c9 : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c0e2b3903c97fcdb3162eec8bcae238cac25529a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:13:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78578b21dd234cdeaeed6ecec45f8eaf : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0dab4ce75350657a6846378835bbdcf6950686c","revisions":{"f0dab4ce75350657a6846378835bbdcf6950686c":{"kind":"REWORK","_number":1,"created":"2020-10-19 18:34:08.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/55/758855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/55/758855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/758855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/758855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/55/758855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/55/758855/1"}}},"commit":{"parents":[{"commit":"8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d","subject":"Fix API sort key for complex columns","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8c32d2e6024d9cfa3e6ab8bff43f2a261aaee82d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 18:24:01.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-19 18:33:36.000000000","tz":120},"subject":"Temporarily set octavia-v2-dsvm-py2-scenario non-voting","message":"Temporarily set octavia-v2-dsvm-py2-scenario non-voting\n\nA recent change lowered the swap space on nodepool instances to 1 GB\n[1]. Octavia scenario jobs were impacted, particularly Python 2 jobs\nto the point -py2 job builds constantly fail.\n\nTo unlock stable/train and stable/stein gates, we need to temporarily\nset octavia-v2-dsvm-py2-scenario non-voting. This change will be\nreverted as soon as [2] and [3] merge.\n\n[1] https://review.opendev.org/#/c/750941/\n[2] https://review.opendev.org/#/c/755777\n[3] https://review.opendev.org/#/c/757841/\n\nChange-Id: Ifc53049e7135e2623bcf7bc2287262ba2f70382d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0dab4ce75350657a6846378835bbdcf6950686c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f0dab4ce75350657a6846378835bbdcf6950686c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
