)]}'
{"id":"openstack%2Foctavia~759098","triplet_id":"openstack%2Foctavia~master~I6693be3e2d2f0e34f19f07f55b0429f8fda317b2","project":"openstack/octavia","branch":"master","topic":"task/41009","hashtags":[],"change_id":"I6693be3e2d2f0e34f19f07f55b0429f8fda317b2","subject":"Handle listener not found during loadbalancer status update","status":"MERGED","created":"2020-10-21 16:14:40.000000000","updated":"2020-11-18 19:28:14.000000000","submitted":"2020-11-18 18:30:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759098-1605724237431-28503777","meta_rev_id":"b28f9a1728aee4f3feacd4c48b347d2d3b1ba3cf","_number":759098,"virtual_id_number":759098,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-11-18 16:12:53.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-18 18:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-11-18 16:33:54.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-18 16:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-18 16:34:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"date":"2020-11-12 17:33:25.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 15:55:32.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-11-18 16:33:54.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-18 16:34:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-18 18:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e351fc7ab35f78ae2aa33b4e40a23aa6b9b1ae6b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-21 16:14:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5803d520b85adeb63e3a8097f4915f4466a0d226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 19:02:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6c466e22bf4729a14a45a421d59fcb : SUCCESS in 12m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08cc64fae1fa4c0992e08c50de684b08 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287a78a8293444e4a1c85dc45069b000 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a90f9400054f417ba8faeea51a6119fa : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/441d453722dd489a88a25bd1d1054575 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc31fcb55f44872bb0a13c0e8cfb9a0 : SUCCESS in 12m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c2221e0c83d549aa84eaaf9c7823af45 : SUCCESS in 9m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6846ddad22c74840844ab9af8081f878 : SUCCESS in 10m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/68d4d92dff48408ba120394746adf60b : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/df42f9a01b3943e6a92904b38105a464 : SUCCESS in 8m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58d8ae3d4e64467eaa1446d1b86078e0 : SUCCESS in 1h 36m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/105533e688d940c393c0fe805ee8c409 : SUCCESS in 2h 00m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aa56ca2dff3143638e038fdbd4bb20a4 : SUCCESS in 49m 54s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2a815d162eb34039b590d696b9a68523 : SUCCESS in 43m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/344404707c00440d829da18f83b8213b : SUCCESS in 41m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5bfa93520e6645e1a4e8e50158f15496 : SUCCESS in 56m 28s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c9708f5e16da4ce8920f3e675bb70fd2 : TIMED_OUT in 2h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/025c4adfb5a643659e9a3199e0d861a4 : TIMED_OUT in 2h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c83cd255d1e944dd92e5a1570ae51d03 : SUCCESS in 2h 06m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10b9dd0e0512465ebb38fcafc22e0c87 : TIMED_OUT in 2h 36m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/573021a12ee84f8f9207949a5c769a2c : SUCCESS in 2h 12m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f822407e42db41808d43c7f914fd9c5d : SUCCESS in 2h 26m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a934ac25990765a4adbd2f7ed1d5b3e9dc4064f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-11 07:22:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de30656bda39207d2d2525256c603377aeaffca7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-11 21:51:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c7fb582c0e972eb0446d49bbe4db6b887c413be","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 14:49:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f23c296ab8c8c7f0a135386d37dbf88142a226fd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 14:54:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c405d379efd583ef7728821b637d9963a6b36b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 17:23:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d52db7ac0dee4940bfe7788f5dfdccaf : FAILURE in 9m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/029fec5f45c946a38533ca3bee68ae76 : FAILURE in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8969709071a845a0b13ca14d8d1c4ee5 : FAILURE in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/982cc645420f45a7a7192decb651090d : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f0bfb096474017bc7e196e806f2065 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc0b3617139b47d883788e5a3d9bb455 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4835c4628d174173a15fe190d2c63d2b : SUCCESS in 19m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8940cd005c5b466fb18ec9243fd804ec : SUCCESS in 6m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/71f600afbc214e5aa2294fd02dc24d50 : FAILURE in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b7eefb4c02b46f28dac57ab34af8ee0 : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cc298c21838b40eabcdc6aa646128ff5 : FAILURE in 9m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df6be760b67d4364a9bf42dd453d7801 : FAILURE in 1h 27m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d816023d90a48789c996525a469223e : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59a81ecdda7a4cf9b460bfb5b6c5d656 : SUCCESS in 41m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4141ceccdf03423386d95f6ff4a1b0ac : SUCCESS in 38m 23s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a8f1abf6bd04dbea52455dafab85f08 : SUCCESS in 42m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0caec88a2db949f4ae337ab9e3474e73 : FAILURE in 43m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/59c89a397704473cb06c7d25c821acb7 : SUCCESS in 1h 22m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6594da52c86b45bc94a933d0c3f94842 : FAILURE in 2h 15m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0897d3121d0b4efd8551a83f2fb542b8 : SUCCESS in 2h 04m 46s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/627a5f1b85dd4432bb7200d648d34c1e : SUCCESS in 2h 18m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4f6f44a4861142dbb2e0051ef49ba4f0 : SUCCESS in 2h 23m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a7a517424fc84833a62d1a150cf96852 : FAILURE in 30m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py","accounts_in_message":[],"_revision_number":2},{"id":"3eb00937abfef6fedae69be062a6f8453cad15fa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 17:33:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"762b081f8a33788952fecc27c9558101894edba3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 17:35:18.000000000","message":"Patch Set 3:\n\nRestored PS1 and based another patch that fixes an issue with tags on-top, as there was already a separate bug for this.","accounts_in_message":[],"_revision_number":3},{"id":"f8f8e67ca27e419488d5109111eabab7d4f3080b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-12 20:14:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91a7caca6f0d43399bc4907e04946f1f : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c02594c0ccb748a1b0e6d3795d721bf7 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26011940aad4c4d8cbfe6a055952f28 : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6303f2460c1341abbbb0ec64a66ae936 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48929d9934924feea1064249d5ac45d6 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd2dd5219afc48e3b68ddf627082eeb4 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0239ce20477e49abb885b2e381de97b0 : SUCCESS in 20m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/85118afa6c9e4fdd8c9d1bbb7133cfda : SUCCESS in 6m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8ce4cedabfe34f5589ba84fd963127b5 : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a3b6063f1f734f14b8963c606d10da69 : SUCCESS in 10m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1f672a84427463db78e1c881afbe0c8 : SUCCESS in 9m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d1741da72a5b4a809bb57fb734352830 : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ab0417842b24049879b4c8ca85c6522 : FAILURE in 36m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5d7a35015364ec3bfa77cd4eb8fc0d1 : SUCCESS in 44m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8d8370cb62274d82a0d198ee5338adc2 : SUCCESS in 53m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d19f1c8dfb446a6921a271a2d7014d8 : SUCCESS in 53m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/29254ecdafa647168bc1e3b0665b2171 : FAILURE in 53m 35s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8f195d5a40f485ca48cde57c92bd03d : SUCCESS in 2h 00m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a8f53f01545742fa8de3018b127dda7e : TIMED_OUT in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7a28e71aab8436cafc3eeeca07fbc22 : SUCCESS in 2h 03m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9041274c26a643aaab800cfab1a795fd : FAILURE in 1h 41m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/63e7e2e4885b4e37be9e23af635c2d38 : SUCCESS in 2h 02m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7a498f3f04814d429c36fed1b6fa1cdb : FAILURE in 35m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py","accounts_in_message":[],"_revision_number":3},{"id":"4d95bf261eb03ab48bb578ec0caa3fd2353916e2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 19:33:56.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3c21a4c4b09c9932fc06af2c03d1212939a30f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:29:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b0724e8ae0d4dc3981a2d02023da11c : SUCCESS in 9m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0bd0344f2b54380982d661de2fc55ee : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d451b6bde0a74add90f02461d7c37c9a : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e55c7238fa3e4bb3b0f445466ec4021e : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dea4f4c150484958800d4d84e491db85 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a0c62908ba14a05a30192236d63c2fb : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313311f162c246fba1bc88163e6a4da5 : SUCCESS in 18m 36s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/621481d92f994b6ca21c8a12ad257d97 : SUCCESS in 4m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5caa780dff0e41039fe07a95e7354fd0 : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4cd3a27c09e435ea70991ccc9e9cad4 : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c30ba00df071449ea4adefbfff0d764c : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/041612c7b53545039404dce864109309 : SUCCESS in 1h 52m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4e22309c19544688a2e926deb877977b : SUCCESS in 1h 53m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d09126000b7746618c7553b31443813e : SUCCESS in 41m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f5a756f760e6457dac1bb6b2cf28abd2 : SUCCESS in 39m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8435b06fb24748d58a4a9eb921611dbd : SUCCESS in 44m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/99b8d1e0151d4684b90987254fbb1a38 : SUCCESS in 1h 00m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dbbb6ec21379456ba6f7ea8903586946 : SUCCESS in 1h 41m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/117659a05a9a45b0bcd368ea1ab6dcde : FAILURE in 2h 23m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f872702640cd4841b011c7de01a1d51b : FAILURE in 1h 50m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ead997cea0154fd5b7e0c3f8dcbd4160 : SUCCESS in 1h 48m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3686dacae4d8488b8fbcc740a01384a4 : SUCCESS in 1h 40m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/87c0743b7a4c4884accb339c66cb2ffa : FAILURE in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55ea2c10a495653528081a1f2c15f6b6bba89b50","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 16:12:53.000000000","message":"Patch Set 3:\n\nSo the issue with tags will be fixed in the child patch fixing a tags issue, https://review.opendev.org/#/c/755828/","accounts_in_message":[],"_revision_number":3},{"id":"a725fa6f3f5b7b2695e0ae6d121826e34dd46c42","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-18 16:32:40.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"ecf9cd861e45a43ebef35cab5eefbc17c8f0aaa0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-18 16:33:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"adb7a2bad8e5cac4d0e77427c181918e4af578e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-18 16:34:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"d5a1f64f19065f655f86ca7d7286441143341d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 16:35:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f76198fafef2d6374d1309402731c1006b255b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:30:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c9402f1c324d0c2a167a68582138d784f1590966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:30:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67b71caad3c14f2aa2a868b6a1da1b27 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc72a61470654799b912b3862d7b73c4 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e3f0c9c086d43db93958f48be19c8dc : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9642a241109745ff861c0d5ab50c1130 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844b950aaa9145c28165a065619c80f9 : SUCCESS in 19m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11402adef3f14ceba2115d3e47ec369d : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8a0afdd90da4274aa4897551ccff4c4 : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/334d5e48efd8494697fe192c0b32f7a8 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d65a73591e2487fb22d0859887eebc1 : SUCCESS in 1h 52m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6762901332d74b289bb9fc247ad0a11c : SUCCESS in 50m 09s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1e76b360f84c45a1896e9f614b70304b : SUCCESS in 39m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ed44451f5a8948749f9f631fdb19d44a : SUCCESS in 44m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b7ab9233038143278e30e6757fbf6a6a : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":3},{"id":"310bf81201503028a3719f7841591160fe229aaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 18:32:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7c86b46c8749deb42a47bb7fab6108 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":3},{"id":"e8b404cb5a16e968919d5525fb6a754c6134a99c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 19:28:14.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit d893700c89ebff62df2e214dc9a17bb3fd83c0cf","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"17ae2625da0487b7398b927e4e67a713b61569d1","revisions":{"b50f38b5b80e62f60391ed59b8f67fd4d3005ff7":{"kind":"REWORK","_number":1,"created":"2020-10-21 16:14:40.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/98/759098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/759098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/759098/1"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 15:59:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 16:13:53.000000000","tz":-240},"subject":"Handle listener not found during loadbalancer status update","message":"Handle listener not found during loadbalancer status update\n\nIf a call to update loadbalancer status cannot find the\nspecified listener, the API server will throw an exception:\n\n\"Error while updating the load balancer status: \u0027NoneType\u0027\n object has no attribute \u0027update\u0027\"\n\nChange ListenerRepository.update() to check the DB query\nreturned a record, and if not, raise a NotFound exception\nback to the caller.\n\nChange-Id: I6693be3e2d2f0e34f19f07f55b0429f8fda317b2\nTask: 41108\nStory: 2008254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b50f38b5b80e62f60391ed59b8f67fd4d3005ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b50f38b5b80e62f60391ed59b8f67fd4d3005ff7"}]},"branch":"refs/heads/master"},"54e88f01137299a8fbdfbdd63f55bc598a0cf08b":{"kind":"REWORK","_number":2,"created":"2020-11-12 14:54:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/98/759098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/759098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/759098/2"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 15:59:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 14:50:26.000000000","tz":-300},"subject":"Handle listener not found during loadbalancer status update","message":"Handle listener not found during loadbalancer status update\n\nIf a call to update loadbalancer status cannot find the\nspecified listener, the API server will throw an exception:\n\n\"Error while updating the load balancer status: \u0027NoneType\u0027\n object has no attribute \u0027update\u0027\"\n\nChange ListenerRepository.update() to check the DB query\nreturned a record, and if not, raise a NotFound exception\nback to the caller.\n\nChange-Id: I6693be3e2d2f0e34f19f07f55b0429f8fda317b2\nTask: 41108\nStory: 2008254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e88f01137299a8fbdfbdd63f55bc598a0cf08b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54e88f01137299a8fbdfbdd63f55bc598a0cf08b"}]},"branch":"refs/heads/master"},"17ae2625da0487b7398b927e4e67a713b61569d1":{"kind":"REWORK","_number":3,"created":"2020-11-12 17:33:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/98/759098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/98/759098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/98/759098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/98/759098/3"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-21 15:59:17.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 17:29:30.000000000","tz":-300},"subject":"Handle listener not found during loadbalancer status update","message":"Handle listener not found during loadbalancer status update\n\nIf a call to update loadbalancer status cannot find the\nspecified listener, the API server will throw an exception:\n\n\"Error while updating the load balancer status: \u0027NoneType\u0027\n object has no attribute \u0027update\u0027\"\n\nChange ListenerRepository.update() to check the DB query\nreturned a record, and if not, raise a NotFound exception\nback to the caller.\n\nChange-Id: I6693be3e2d2f0e34f19f07f55b0429f8fda317b2\nTask: 41108\nStory: 2008254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17ae2625da0487b7398b927e4e67a713b61569d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/17ae2625da0487b7398b927e4e67a713b61569d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
