)]}'
{"id":"openstack%2Foctavia~759164","triplet_id":"openstack%2Foctavia~master~I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b","subject":"Fix provider driver quota handling","status":"MERGED","created":"2020-10-22 01:42:36.000000000","updated":"2021-03-15 00:32:15.000000000","submitted":"2021-03-15 00:31:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"759164","meta_rev_id":"9f2fa3334b1e31df5210ee9fb831479ba3cbb648","_number":759164,"virtual_id_number":759164,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 00:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2021-03-11 22:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-14 22:21:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-12 00:33:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-03-14 07:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-03-14 22:21:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2021-03-14 22:21:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2021-03-14 07:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-22 21:45:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-26 06:17:54.000000000","updated_by":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"reviewer":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"state":"REVIEWER"},{"updated":"2020-10-28 20:55:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-08 05:27:01.000000000","updated_by":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"reviewer":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"state":"REVIEWER"},{"updated":"2020-11-08 08:37:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-01-28 10:48:02.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"7eef8310156e8525610bd3ce725214e1095bbebe","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-22 01:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d97559c3789112d624762b94b1bc6baa2afe879c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 04:21:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efd829c5e43b43a1a16e9fdc134d98f0 : SUCCESS in 10m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2be50dd67c940eb958c4ef0cbd77ae9 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb6e7c76ab34538872cb869619b7fe8 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdc3430687a94d5ba21377c95cbd1b24 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c831d13d3cd4fca831103a7a1fcce64 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8147ec53673f48f8b11e0d3ca12bbddf : SUCCESS in 13m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa5bff2ab8f84fd29000fdc5a5ae07f0 : SUCCESS in 6m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/03a5569a2a0c48c5b5cd5f3ed4b4c067 : SUCCESS in 6m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6cfaacb4a0b3444b80691b2d60f8d41d : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6d94a4919c18402ab4a981debb0e5422 : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46d51aca747a4aaead3243d801261841 : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b281c03745fd4359815762eb9d625639 : SUCCESS in 1h 44m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18b292e0681a4656a04dabd844f15f14 : SUCCESS in 1h 58m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7fecb18a62bd4e568abde3cb6b92fe9b : SUCCESS in 39m 11s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/645af30c5c9440a2946459dbe057a15d : SUCCESS in 41m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62e141cd11ec46808c4cff7ce7234cc8 : SUCCESS in 41m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd289a8457c64019ab9e56f4d1f52628 : SUCCESS in 1h 04m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/50d2c4cedd00454b881a3e529e7d29c4 : SUCCESS in 2h 12m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a85624cbb63e444b94010a4543eb897b : SUCCESS in 2h 09m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f499e86051344abfa505d4c57b04c711 : SUCCESS in 1h 58m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a56ff39cd7f246d8abdf904ce2caea6c : SUCCESS in 2h 28m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b0b3d59b84aa4ad1a93d4c22e5947621 : SUCCESS in 2h 11m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5f38f1d938384014b6c1b662ed9af1c1 : TIMED_OUT in 2h 35m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f47ffc8a1f55eaa28f6700314ed9f8bbff0d1af9","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-22 07:47:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c26dfa6af4ed3f410a8664cccd7b10335cdaac8a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-22 18:11:34.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ee750a70cd30e7d9e96fc206955df0ca1586ea03","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-22 18:52:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31737dd94b3456ea435dae83100c6be3ae90b8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:31:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87999961fb2446f088fddce0fa3b2e5d : SUCCESS in 11m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25b9eabdb1bf43b9b94e9c7092d7bbab : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f10442271304362aaa28310c584358c : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c397afb5a98b432e9f94dacc821d9854 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cad7f32608114f54bf3dea31031346f5 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96703eede5e347229b5574ff8a43b052 : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae695f373d8e4733894935b16970ffda : SUCCESS in 7m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/771d15d1ada540f39e59aaf229e30b88 : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2bd0d13836ce457295631b4685e84953 : SUCCESS in 10m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d924fa28d7f4473fa5287f33619b5784 : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87e5ada29029477b8e5f2f076e873dbb : SUCCESS in 9m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ece8cfba05a540968ea50572c61ca86b : SUCCESS in 1h 55m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/856831da6edb432a8a775c832bb5fdac : TIMED_OUT in 2h 35m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2669163ef8574788ac6a99972f82e142 : SUCCESS in 42m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/830b9a9189be4e8eb5bd90b622b55960 : SUCCESS in 42m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5c33e36b87f24be2bbaa9994e7d0fa25 : SUCCESS in 40m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e6b38192bc2e4b35a8698bafa98e19bd : SUCCESS in 1h 02m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b3aee5341cf74eeaa0054a75c5846b00 : SUCCESS in 2h 06m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/046af51a02f049a0b7049ffca27e6320 : TIMED_OUT in 2h 35m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1942db8116a24fd58593b0ad697cc737 : SUCCESS in 2h 27m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/67857d4807ef4d0fb196b993beaa35d8 : TIMED_OUT in 2h 35m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/69228dbe4e7f48ba90f9e62172db9e49 : TIMED_OUT in 2h 35m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/92437d646b824a0ea4b9c9fffaa21177 : FAILURE in 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb6d5091e2345afe783b5e7c972c6a16d0b25b5e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-22 21:41:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3bb80a31220bf9a639a6a9e82f0a17f6971a3e19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-22 21:45:23.000000000","message":"Patch Set 2: Code-Review+1\n\nTested this on a live system with the OVN driver which had previously been un-patched (and had 8K \"invisible\" listeners) - was able to verify it didn\u0027t go over-quota while running in a loop a few hundred times with only one quota slot remaining.","accounts_in_message":[],"_revision_number":2},{"id":"d53317bea8ab5f8f22cba7ca5bd782328c894173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 00:21:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27a0ae99fefb4668b13a8cddc69b13a4 : SUCCESS in 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b57dd5b821e482f96312efd7772a1fb : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc082da67e84e339e928c369eaba9c7 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6b830803c2e49f69566c542f8710bbd : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab81b4da19ff4b2383c82e793eff4569 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d759ebcbb18441efa6d192f3d5da43ab : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0fc086c8ace460b9206bf8df95f3f6f : SUCCESS in 6m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e20e6362b44c4ac7ba5c45a51b0f3a10 : SUCCESS in 8m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6ced6839cab94b92a4b29ac1a0ee138c : SUCCESS in 9m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3de2f22b7404cb0b1d07d14464f1543 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34efc82b018a44478204480593a15566 : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1919b15bac8f4ea5b58dae1301e8107f : SUCCESS in 1h 42m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e31969fe6ca3406394a1f4f648fa0f85 : TIMED_OUT in 2h 35m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e2b8ad481f774ce6b3a4ce9ead7811d7 : SUCCESS in 48m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ac285fbe946497aa0dc63dbdcc05daa : SUCCESS in 45m 32s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46f7728888f7448bba9d6cbbaad798f4 : POST_FAILURE in 54m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/47b5414a06064affad6e6ffdc7fd5686 : FAILURE in 1h 03m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de9d8558da9c4902bffbd17d041eefc0 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df97171744c84d5593500aeb243d34ad : SUCCESS in 2h 24m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f7c84da450d84425a1ea69e542690bab : SUCCESS in 2h 21m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0ff989407b9b4a9f836a77c2074818aa : SUCCESS in 2h 35m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/93f7696c8c1943ea8b1fea8da9248c8f : SUCCESS in 2h 30m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/79672aba5c844c8e872b640c15252efc : FAILURE in 31m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8a8fc268231027a9898257845e13b5585adbdfc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-23 15:47:44.000000000","message":"Patch Set 2:\n\nrecheck\nPOST_FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"8fdb42719b88d8678ff3c1b7083601176d238047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 17:39:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfb0436f2d7f4d838602e56664f3682e : SUCCESS in 11m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72fbbe1ac55e4e3d80f67f296dfcd731 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533b80025ea74b7499a452e979ad20f7 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/972e6b76b8f44e328fb546b226891b0a : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe4863422194d71ae76ba3f1d2dbc06 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465a8aaf4146464690d7b8c73988c2c1 : SUCCESS in 14m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02131ebbf31e47748e4268862edce1ef : SUCCESS in 6m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d4f0c7f29ca7411b9773fcebbe269f64 : SUCCESS in 7m 19s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dc1203a5c4b9430abfba083b4f4e08b5 : SUCCESS in 8m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75b361d684114c2c98464d64efa7ce2b : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84ea2ea45fb34098aaf4a502e82f4e1c : SUCCESS in 9m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/96b8fb26c4214e8cb13cce6f60bcd150 : SUCCESS in 1h 48m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2828ba97f9d54d458facd777d91dd36e : FAILURE in 35m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4b4bad203da04b1da1c2075ac02bd8dd : FAILURE in 34m 08s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b05fda66121b4c7db7a840336b71b2e9 : FAILURE in 30m 27s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe05fd5a6d3d4d78bfc391afc4c38a17 : FAILURE in 34m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc456fb57d8749a9a6ededd01758e4c7 : FAILURE in 52m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb720469c1ad4942a094fc1801f90e52 : FAILURE in 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/402d44b862944cbc8f558c69684b74b5 : FAILURE in 36m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cab5d76cf5cb4763982c6bdadc046f0b : FAILURE in 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/97890b6b4c8349ca96481a6083eb2844 : FAILURE in 40m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/833cc08889994c599fb019ff6eaf038b : FAILURE in 29m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/34af4c1e75df45fea33949e959dd0414 : FAILURE in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"878704e3d82eed3d0688c8bfc1ebae164a24dec1","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2020-10-26 06:17:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dbb118d87e5ac7c1c35a0cf87fc256a818e52781","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2020-10-26 06:17:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8f55b1ce24c1ddcc83f7b4e86d68d526eb3e1795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 07:58:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71f8f4098da04fe781816577d6db4efa : SUCCESS in 8m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47f16878dbc343e5a8ee19326547d1d6 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9715ddd684458fb91bc2ecd8e1300e : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8683a5285ddc42abaf7f76a003810f8d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ec92ad11624d97b75768d6dc0642fe : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eaccf9bcc8e46d08faa6719383c1f52 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0aa32a841c54d0896c9bf5a49815147 : SUCCESS in 5m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/0d139113fb934eb0bdc7842993f2ce38 : SUCCESS in 6m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/55f9e668ba574912bd3fed75761f02b0 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20d4bc95e9984ac197abaf4dabaed32e : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6e3252acfc424457a35cdde49731d4d9 : SUCCESS in 6m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ef22e19c12b54ccdb0ce1168be655c2b : SUCCESS in 1h 33m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6268d88e755a4d1b93b41fe3a865471f : FAILURE in 31m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9085375eed540a28a5106fa15c1d253 : FAILURE in 33m 28s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a55dd5f87f5c4d239291703b9aad2895 : FAILURE in 39m 40s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19ea9696de3d4587a38ad863453a101a : FAILURE in 40m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19f4e61152a149d28d4615ef43a37ff1 : FAILURE in 1h 02m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e65b110bca7a471b819f30e5e2109562 : FAILURE in 29m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4bb3cbcad3ab4655a92e26b3c554f164 : FAILURE in 28m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f0322226d8e46c582df2daa8897b5ec : FAILURE in 38m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8261b4d381dc4ba3b3077e3f9a333e35 : FAILURE in 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1a9ea6b9171d47f6a7fe574d54425f72 : FAILURE in 48m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/92326416f42243c980c76468b5d57cc0 : FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82bf259fc9c01baa4b24147d5d2b1c52d4a4b8f6","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2020-10-28 16:34:46.000000000","message":"Patch Set 2:\n\nrecehck","accounts_in_message":[],"_revision_number":2},{"id":"e2ef80f157d0c6665be232a43c1bdbc9b04e3695","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2020-10-28 16:35:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5ab67abb1c0b5c61d22a088b85a248103f365b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 20:55:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51f7a40923774f538038e80cb2e87581 : SUCCESS in 9m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddfe3f57d9174e6f8e72d96ad81807ac : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da08658e54840deba902ba4e5762870 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b7637a43e04d9694a8d67998c0acda : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac1dceecd6794d79934337dcdeb9cb4e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02e181617ec400189b401e45e0854a3 : SUCCESS in 12m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aae798f6d0be4f72aae2ecaf987c81f0 : SUCCESS in 5m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3cb1cdb4c8eb443aac0e0f346e8456f3 : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/795b15b5026c492289cbb96e1806226e : SUCCESS in 6m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c1af242ca9054d488b5116dbcac83ddf : SUCCESS in 5m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/415fbd49edcc4378bfad4952efb5d28e : SUCCESS in 7m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8b772ba1c97b4e8d802c1f6186933f07 : SUCCESS in 1h 20m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/18fe19c4226b4e9aaa764e26d8a61d6c : SUCCESS in 2h 14m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a3e5e4e14434615910c3af44badaddc : SUCCESS in 42m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b3e1f3030fb74434bd80db77740d242c : SUCCESS in 45m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f103870d9794c9a9effa838f05c636f : SUCCESS in 42m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f7d9b2305fa441539730a691f6627cd3 : SUCCESS in 1h 03m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd1a3e4601e8413a8dc312b27ce37f49 : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d5b418a623042adb5f9f63bd3dbe97f : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6dbf0df0f6449738bca4d97c77189a5 : FAILURE in 1h 40m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/dceb41215d10429f85f2b4d9ad586637 : SUCCESS in 2h 21m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/80cb18cc92fa4aa7a852684774b98cc4 : SUCCESS in 2h 11m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/4878ea2e174e426684c63b188aa7ff1c : FAILURE in 45m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c2817de73d38c3fbe18ba44f3beb2642f27a075d","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-11-08 05:27:01.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"d1d426a5aa3c423df5b4d5daafc6755ef0d33098","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-08 08:37:14.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfc49d1e741bb2b2d6de5b872b7541dc1c004591","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2021-01-03 10:51:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8d761fd31f330f247ad0a4c122fcf54c9b73b05a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:48:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0c5b323f4d45fe9e2a7f8e63d62e7882d512f807","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-11 22:22:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbbfc7f87fe4eda042e41f11b5eadcb513ad5ac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-11 22:43:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"864fd2fdce8e78523a87940d4274145543b829b1","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-03-11 22:48:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f41c1dd1538687ddebb0fb55f120ac41751ffaab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-12 00:33:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2b8b4eb58f1b7e49249f62a6e047e0cdada3ed59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:53:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf7af441ee464760999cd61b0f0d3341 : SUCCESS in 11m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c84482b9860c42df9922a37d364fe904 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b47811696b4ab4b947534232ec7a35 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f49eedb968948e6b36170a5891c07ee : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71fe1c05408c45bbb7a821f43078b1ec : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b92677f0eeb4fdb80c1c88f7a072f04 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8785db18d914a968b93fc315d7670a7 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10aba2f787404d7e8232584ee58f4951 : SUCCESS in 6m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ce6a37e2a7f24283b6394af5fa6cc4ca : SUCCESS in 9m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fd2009b7cb4d4f4480bd8273b3906d03 : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/32a55d211de84240b12668b93bf39f74 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1221b4caeea045d2bd0b95c340b83a90 : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/be540b8537eb4b1f802dadcb0a05618f : SUCCESS in 1h 21m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74fb7ea6240345fba708e1dd5e4260fa : SUCCESS in 1h 37m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d2cc62ce3544308b6a7dbe6cf990c02 : SUCCESS in 39m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4e377c44a5ec4a918ba03e9bd456fe64 : SUCCESS in 36m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd4b26435b884d61a0ebce2c4adc01ac : SUCCESS in 42m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13b23e74c2cd4fc89c5568671ab62d0a : SUCCESS in 48m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a67d0de321b4823a7e0283872b8c4ef : SUCCESS in 1h 56m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1b78f2c273b48e6a34e20600b2c49e6 : FAILURE in 51m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f40361a956774a3582c07acce4c53540 : SUCCESS in 1h 38m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f9f27d3662914867aa937ec85d017c2d : SUCCESS in 2h 06m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7d2f2e2c75164885a2becfc86540c8e8 : SUCCESS in 1h 30m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/0967302b2b8f4c56b9fe4f6643796894 : FAILURE in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"159fe15e82fe56c7b48dffa78ed5155392841190","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2021-03-14 07:31:44.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c722803af7dfd098cff379744efc71a948f0f4d5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-14 22:21:48.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f2b32771b8d13050ef958d4eff55dbad600c176b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-14 22:22:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5ba75a9524f5b11827f3fbaca055d3c53a3c201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 00:31:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d8650b2b7b8402eaac2fd31089377c0 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ab16a52afc4cdd9260b4cde22a6fdf : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/478630d085f24bdfa58499461cc8a406 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7217126d615645e191302f72719b6b14 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556c83f320364e269d2b05e2332815cc : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70561bf40db648e8972e4e8e3779d83e : SUCCESS in 6m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d56c2f05826245dcad3c9706e13c751c : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76f9ca1ef1374f28bca004a23006aa45 : SUCCESS in 8m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/043484b4816a4ce99365b531ac440895 : SUCCESS in 1h 14m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/34c69f9546fe4c21829503f17be9a37f : SUCCESS in 2h 04m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59b7350a0afe4cb7b4e2fbb6742d16a3 : SUCCESS in 48m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e001ba8846cc4b6cb997064ec8ef769c : SUCCESS in 50m 57s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/055985d4d307401787510be6b1705d08 : SUCCESS in 51m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c9d4be209187489a895a3e363a299e93 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":3},{"id":"51f540018fa8c187d44cc9c65f8062259ad20d57","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 00:31:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9f2fa3334b1e31df5210ee9fb831479ba3cbb648","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 00:32:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb15962fb97348a5ae5e8d903344fec5 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd4d2a0bd6bb45bdb4f0bd489d7db8e1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fc8ee42dfccfd4b0735185896c30b0899daa4c55","revisions":{"c39192d0b85f7ab69784afda2b38ab7feaf26728":{"kind":"REWORK","_number":1,"created":"2020-10-22 01:42:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/759164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/759164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/759164/1"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 01:40:46.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 01:42:35.000000000","tz":-420},"subject":"Fix provider driver quota handling","message":"Fix provider driver quota handling\n\nWhen provider drivers registered a load balancer object delete,\nthe driver agent was not decrementing the project quota.\nThis patch corrects that by decrementing the proper quota\nwhen a DELETED status is received from the provider driver.\n\nChange-Id: I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b\nStory: 2008268\nTask: 41133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c39192d0b85f7ab69784afda2b38ab7feaf26728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c39192d0b85f7ab69784afda2b38ab7feaf26728"}]},"branch":"refs/heads/master"},"edea3ded8faa9f6b9cdf2baf0126aa54d3bc02bd":{"kind":"REWORK","_number":2,"created":"2020-10-22 18:52:15.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/759164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/759164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/759164/2"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 01:40:46.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 18:52:09.000000000","tz":-420},"subject":"Fix provider driver quota handling","message":"Fix provider driver quota handling\n\nWhen provider drivers registered a load balancer object delete,\nthe driver agent was not decrementing the project quota.\nThis patch corrects that by decrementing the proper quota\nwhen a DELETED status is received from the provider driver.\n\nChange-Id: I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b\nStory: 2008268\nTask: 41133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edea3ded8faa9f6b9cdf2baf0126aa54d3bc02bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/edea3ded8faa9f6b9cdf2baf0126aa54d3bc02bd"}]},"branch":"refs/heads/master"},"fc8ee42dfccfd4b0735185896c30b0899daa4c55":{"kind":"REWORK","_number":3,"created":"2021-03-11 22:43:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/759164/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/759164/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/759164/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/759164/3"}}},"commit":{"parents":[{"commit":"217cdbb719a8d303c486ed6526eedd9ca6fe83eb","subject":"Merge \"Add aarch64/amd64 support to amphora and devstack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/217cdbb719a8d303c486ed6526eedd9ca6fe83eb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-22 01:40:46.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-03-11 22:43:16.000000000","tz":0},"subject":"Fix provider driver quota handling","message":"Fix provider driver quota handling\n\nWhen provider drivers registered a load balancer object delete,\nthe driver agent was not decrementing the project quota.\nThis patch corrects that by decrementing the proper quota\nwhen a DELETED status is received from the provider driver.\n\nChange-Id: I7d705c9f4f0217c6fbe332f45b15892bf1d4a90b\nStory: 2008268\nTask: 41133\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc8ee42dfccfd4b0735185896c30b0899daa4c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc8ee42dfccfd4b0735185896c30b0899daa4c55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
