)]}'
{"id":"openstack%2Foctavia~759383","triplet_id":"openstack%2Foctavia~master~Ia176d06c69413aad64801cf6777f91da72cb9924","project":"openstack/octavia","branch":"master","topic":"tox","hashtags":[],"change_id":"Ia176d06c69413aad64801cf6777f91da72cb9924","subject":"update tox","status":"ABANDONED","created":"2020-10-23 09:41:52.000000000","updated":"2020-10-26 01:03:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c6f876a7912952b195f5cda912cd44ee4effa85","_number":759383,"virtual_id_number":759383,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-23 11:40:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2020-10-23 12:31:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 11:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-23 12:31:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e9b740fc3553bfd09948aa4aa03b4890add64651","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-23 09:41:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f8281dad8366bb96ecd886d82437c7f1c615c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 11:40:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e06001bca43b4ccb9e25a9593851100b : SUCCESS in 12m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ec7182fbfd34055b4f28afe49498b85 : RETRY_LIMIT in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb19f79f86b244d7beda300fe7a796b7 : SUCCESS in 9m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7a99beddc046629e54f5350041801e : SUCCESS in 10m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44a6789c680a417792ea09f7e9dc0ea6 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dda89614b8d4ca0bddfa6bc9b6560e8 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc2e91b9739e451c93b5fad5facb743c : RETRY_LIMIT in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb977dc2fd81438a8796b594701fe9fc : SUCCESS in 6m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ca4457c56a444cd987e450ad3aadd817 : SUCCESS in 7m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bda379cd7d874ae19a9627bfa3b29349 : SUCCESS in 15m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8385acfa84234ffcad41d72df8d732a2 : SUCCESS in 10m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/80c823c27e3046b687441ae5a5d05995 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9247be56b02a4ddda0c93c9ced8fb462 : SUCCESS in 1h 50m 48s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f0834a147c364c20b7dee3cdf0b14d7e : FAILURE in 5m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca45a9ad0ffc4c529d0254cf760029e0 : FAILURE in 40m 06s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/6bcc6855e4e74d94b0d136785c543b1c : FAILURE in 39m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/769acabd4ee44dce82c182bbafb3c380 : FAILURE in 5m 49s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/75c2d46f890547ffba2ae562931af53c : FAILURE in 41m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/20dff492a4ac40388d481c7d50d02679 : FAILURE in 38m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad7cbad9c05244689ab75f724a1d5267 : FAILURE in 36m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c4b5492568c0405c8585ebadd1c3d89d : FAILURE in 5m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/84a2c3d92cfc421aabcd6d9f32a1e310 : FAILURE in 36m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/72e750465d0142e69879ae3d4c882844 : FAILURE in 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f079bdf9cac2718be8746fba132369eaa143a30c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-23 12:31:03.000000000","message":"Patch Set 1: Code-Review-1\n\nI have seen comments in other similar changes that this is a bad idea, so I don\u0027t recommend it.","accounts_in_message":[],"_revision_number":1},{"id":"f269876b369332c284b098366936d4a8b0605cce","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-26 01:03:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83a47968dfd8f13997074b20d8f3eb35ccabb6ab","revisions":{"83a47968dfd8f13997074b20d8f3eb35ccabb6ab":{"kind":"REWORK","_number":1,"created":"2020-10-23 09:41:52.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/83/759383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/759383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/759383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/759383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/759383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/759383/1"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-23 09:40:25.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-23 09:40:25.000000000","tz":480},"subject":"update tox","message":"update tox\n\nChange-Id: Ia176d06c69413aad64801cf6777f91da72cb9924\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83a47968dfd8f13997074b20d8f3eb35ccabb6ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/83a47968dfd8f13997074b20d8f3eb35ccabb6ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
