)]}'
{"id":"openstack%2Foctavia~759527","triplet_id":"openstack%2Foctavia~master~I6a45d1256d24876923f82460beb7eae2511fa2f7","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I6a45d1256d24876923f82460beb7eae2511fa2f7","subject":"Fix ifup if the VIP secondary IP is already up","status":"ABANDONED","created":"2020-10-23 21:46:38.000000000","updated":"2021-04-23 20:48:09.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bd92b36fe53fddcf05de37db9e8f0ce89801c052","_number":759527,"virtual_id_number":759527,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-10-26 19:03:56.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-23 23:29:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 23:29:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 19:03:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ea2eb806950ff9fc9d58b0b66dc808c47cf8dcb9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-23 21:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e274366a4679380105d58cce482a8fcb8c6aa610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 23:29:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d29123ab9240edb582d662d0572f24 : SUCCESS in 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eecfdf2fe55a47c18defd54fb1374ea4 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab9e3fb2300e4334b67d93a8230f3f90 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3e46caf97ea4b3b906c1e639195472a : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7604d0335e74c72885bb0f8cdc1240d : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af1458f261b47c7942d6c16d02e76e6 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dc21a4d606143a8a64c9c155abc3938 : SUCCESS in 6m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dd4a31bb661c454facbf1e4e8990256c : SUCCESS in 7m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c17ab42d4ab34b329ba636e4a4629240 : SUCCESS in 14m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a2711483be694eec95766f6599bc4e0a : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5dba5e628ae94f09bf30b674d88993f8 : SUCCESS in 9m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d905d585cec4ae598ecce619b9e553c : SUCCESS in 1h 40m 39s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bd1f282224f14642901e07e3c9f31798 : FAILURE in 28m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec6f2dc7ffa14114b7a62af24759c59d : FAILURE in 34m 24s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/528ad12226214676bb95b4b444778684 : FAILURE in 1h 03m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/46f268a312b54fc8b1c8e713f0cf84a2 : FAILURE in 34m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aa464faf95d94c00ac56f2b0bc17c496 : FAILURE in 51m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9698c732e5f48b684b69605a2ac79e1 : FAILURE in 27m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bfc0a063e93f40028fa7af771481f648 : FAILURE in 31m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4914ec8fbbb4415c9ed35b4fb96d1931 : FAILURE in 32m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/657dd1ca5c124e9487feda20c7fad845 : FAILURE in 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0bc320a9d84c4bef859c16ad2c084eea : FAILURE in 31m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/06fc80d7e89044fca386a952ce42d9ef : FAILURE in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41a25baa048481806cb375555731e17682b2c477","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-26 19:03:56.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bd92b36fe53fddcf05de37db9e8f0ce89801c052","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:48:09.000000000","message":"Abandoned\n\nWe are moving away from ifup","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e9cd48add7284dd7a00b9fa1a54e965e64f48fd","revisions":{"1e9cd48add7284dd7a00b9fa1a54e965e64f48fd":{"kind":"REWORK","_number":1,"created":"2020-10-23 21:46:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/759527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/759527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/759527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/759527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/759527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/759527/1"}}},"commit":{"parents":[{"commit":"71440bdbdb6e66f154fe69eb43ce116d131506a6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/71440bdbdb6e66f154fe69eb43ce116d131506a6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-23 21:44:04.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-23 21:46:37.000000000","tz":-420},"subject":"Fix ifup if the VIP secondary IP is already up","message":"Fix ifup if the VIP secondary IP is already up\n\nIn some cases, when we call ifup on the secondary IP for the VIP\ninterface, we get an \"RTNETLINK answers: File exists\" error because\nthe interface has already been brought up.\nThis patch will log and ignore this error as the interface is already\nup.\n\nChange-Id: I6a45d1256d24876923f82460beb7eae2511fa2f7\nStory: 2008272\nTask: 41140\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e9cd48add7284dd7a00b9fa1a54e965e64f48fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e9cd48add7284dd7a00b9fa1a54e965e64f48fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
