)]}'
{"id":"openstack%2Foctavia~760307","triplet_id":"openstack%2Foctavia~master~I8f191a720a3d38a7f579ed5f587563bfd40cce15","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I8f191a720a3d38a7f579ed5f587563bfd40cce15","subject":"Remove useless proto parameter from sample functions","status":"MERGED","created":"2020-10-29 09:16:14.000000000","updated":"2020-11-25 14:05:40.000000000","submitted":"2020-11-25 14:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760307","meta_rev_id":"93c5a2c9e17cd2efe8c2c65937a443483e18e116","_number":760307,"virtual_id_number":760307,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-25 14:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-11-23 10:16:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-11-20 09:16:43.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-11-20 09:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-11-23 10:16:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-11-20 09:16:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 16:12:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-13 00:21:39.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-16 12:43:08.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-20 10:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 10:16:57.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"}],"messages":[{"id":"d8d6932d06673342c29bda4733bb07e43a153dd6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-10-29 09:16:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca780a57420b0ca531e3f73b7dd791c66863621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:35:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d36fe5c990a4eca85e6bedfa8ffac7a : SUCCESS in 9m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b51f88bce5b41d3b9d9516cd5200a3f : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2bc712c63c408e9078be8232b1a078 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08cb2fdac18a4648a26208e9c4808488 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1e9f38fd21546a9a9b789ae7d52127a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e80b52eaa1341f694cc9ee955f9b70d : SUCCESS in 14m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ffc25d985e9e4dae84eef9397efdf36b : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d2bc449a015440179d3249d907e807cd : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1},{"id":"986d96a18b7d6253b37de4470a93a62d8cd3664f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 00:21:39.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"bceb3c2e25f4e113047d37a3ba9d8fce93df782c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-16 12:43:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68986327d1d73d7d33332c7ead780ff6259508b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:43:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca25c8fa89c47813e31573f16d5e8a3e7d655146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 12:54:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/None : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf36debe5d664a37a8b8597ee1c52db6 : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa39b414d8bd4969ba70b1859e947c75 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f17961441e4c15907bb4582591c3ba : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3855f28cc54e0c9facf3ddd36cc661 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ece3ae499c5438c91eee8be27d24f1d : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94107991220343d68be58b34240ac0ab : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ca1aa5e3ae1943269afaf651529f8c2a : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e7c2714feee7430594af0ced54d567cd : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a474833351a7441dbe4593a7800a21ed : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/92a5be2c2b66475bb9fd18df7fc474b0 : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py","accounts_in_message":[],"_revision_number":1},{"id":"30b39d5d9bfe3b85e37db971d75b18386358262d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-20 09:16:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"892d1564ae405bdffb17b3199a1a65d26b47d5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 10:16:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0ba7580b5ac4fc08f6e1767450da4db : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52918ade59334b019047743f095f09a0 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c20faded464fbf94536393620821a2 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9adef1940ef4c14abb70d44fea5ff68 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66343db7434e4a01afaf33f404ad6edb : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06d012589f38426e9f904353ca23592a : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc560f05a4048b788afd09c1894eb84 : SUCCESS in 12m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e833e1b8f4e44901833492e44b51d6c6 : SUCCESS in 5m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/deaa9c03fe8943fe8b4a1adc47b6fbc2 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":2},{"id":"be7768687ecb00fe1d30cb9a8227a89c1d126a4d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-23 10:16:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d806e8e54d0a0644556f7ad4c90b44a1fc57f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:17:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"747cf5ce12aeae8923e98e8897f30d37fa4ea9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 12:09:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a48417cc0be466aa2504d71685a3a4a : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e52cccb3576b4d98888ea2f0a8c9d38c : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/922f1cb306ed40098f8bfad37cc449d7 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43a851f4433341e9bd354dfe3511e743 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d82e47dde014de39d5101df1aa0b851 : SUCCESS in 11m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fdc0fb13b17a4c3aaa8aa8b0a411f118 : SUCCESS in 1h 27m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4f418bc1f1884b6d802917788443aa42 : FAILURE in 1h 50m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2bb7f2dd11884d8dbd28181b2efd774f : SUCCESS in 39m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/90558c39599646b49f028ecce5e8903d : SUCCESS in 41m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d441fec12a24414fb8de9de8cee38f88 : SUCCESS in 36m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2625552f682746769677bb81f8d4b915 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c08336722c4888c08212d4e8208baf23e00ea85f","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-24 10:06:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0b42ec85e29f1d3130ac554972b3914b5fe5fa54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:30:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4117f9b4b4344a4a7df55eac815212f : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c036ad8b6b9d4e4e80366310cccdd983 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f92cfdf4bd9401680dd7fa5448c69f1 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22feac021886427faffcaa72370fe091 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b26f38e953447b4acca3a2c14643e8f : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e92a5f4108140d5998dd7f998475e4b : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66569f1da6944c0c95bc51c1df882e76 : SUCCESS in 9m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/5fad6c441c5145d9bdd8c7d3b380c27f : SUCCESS in 4m 59s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cadfdec2ac5348d5b902c69aa3a692e9 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":2},{"id":"2f6da0a9f1041515bc3e6317f3c0f6851c8066da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 10:31:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6cb8d9f87351be937800a06af20f7c108900bf62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 13:18:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/230febe605fb4dcaa1594b306c06fc59 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a961b48cfca247558e78eb3b5bc79d91 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e4d9bf0992f4979af5cea12a766cd56 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d42ccb3cf9d4499b1e1812bf1268dc0 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6958f7359bf4365a7372b91e6d6c0c6 : SUCCESS in 11m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ebedf6c1c2b46da9e10f8700d8ffdcb : SUCCESS in 1h 13m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3221dca515454a97b240f0f3a1d5adf5 : FAILURE in 2h 20m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/143b3705208449eba21736cfcdbd6c7c : SUCCESS in 40m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8b554c6bd259435ea80ab25c951d2aa9 : SUCCESS in 1h 31m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8934e491ea34789a498d2ea774d3fce : SUCCESS in 41m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e34ded87f4d4493d9cf01da00aade7fa : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a6d060cf3cf31e468d689d6a1593ae80904a0f25","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-25 10:08:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c186a4cd26fe744d651327294a8f6a40d91a49c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:02:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69e2b4d2bc604289a32f795f67e27648 : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a579c30adc6e461b97d0f6af09117abf : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2bc5c4113b8421caa06c6576736fee4 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0718de7aad16473da416ba88cdfe369f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bedf089f9814d1b8dff5fab04c8d6c4 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f73385e2b9a4981b347ec7f06e1da40 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b1d93fe07c4099a1bb15c612e93be7 : SUCCESS in 16m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc5ebbe45c4f458d8b834a7e5c420f48 : SUCCESS in 5m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ddf95e88a21e460ca51f72a12433df9c : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7208bdb6acf747f683e32b7d5cd7b1a338f65e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 12:02:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4b57028eadea15ecf0fa7a1ca6a11a300bc8286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:03:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80a7253a2f92478688759c106958807b : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869c17e86da7470296ca1c4b79cfd931 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3984bf4807c547feaa375bcdf4080302 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a079780b504b1aa79d844494216e8e : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ced6bee945045f3bdf7e9e21d4447a2 : SUCCESS in 11m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da829d8379e54d2cb93ecea03e374fc5 : SUCCESS in 1h 30m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a22a72e8efa446539eac43d9b4f3c6bb : SUCCESS in 1h 57m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91d33b9e816b4c468524b6f782fefb08 : SUCCESS in 40m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5351ee5031994faf8f2b6f8689aea5a4 : SUCCESS in 52m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7bb97a1619443c38c55b097910b0a9b : SUCCESS in 45m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7cf8bc2a45542468990c37966cb0e18 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c34662f5d15ace2b2ef9f4a7c4a184afdbe3d8c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:03:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"93c5a2c9e17cd2efe8c2c65937a443483e18e116","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:05:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7fe40d7fe224116a4376aa0e8c2d8f2 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9653920a04e43a8229bcb710b51a110f2be93aea","revisions":{"9e84b6d1b2a55d8516b48a6933ada47658002cdb":{"kind":"REWORK","_number":1,"created":"2020-10-29 09:16:14.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/760307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/760307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/760307/1"}}},"commit":{"parents":[{"commit":"009af5b0d8c7a4037bf81eb1b129698ba222d54c","subject":"Merge \"Correct a typo in the document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/009af5b0d8c7a4037bf81eb1b129698ba222d54c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-28 10:02:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-29 09:12:15.000000000","tz":60},"subject":"Remove useless proto parameter from sample functions","message":"Remove useless proto parameter from sample functions\n\nSome functions in unit tests are creating namedtuples that represent\nloadbalancers, these sample loadbalancer objects contain a \u0027protocol\u0027\nfield that doesn\u0027t exist in real objects and isn\u0027t used.\n\nChange-Id: I8f191a720a3d38a7f579ed5f587563bfd40cce15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e84b6d1b2a55d8516b48a6933ada47658002cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e84b6d1b2a55d8516b48a6933ada47658002cdb"}]},"branch":"refs/heads/master"},"9653920a04e43a8229bcb710b51a110f2be93aea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-20 09:16:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/07/760307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/07/760307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/07/760307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/07/760307/2"}}},"commit":{"parents":[{"commit":"58bf439bda1cdeaa3dac8b453e05485092bd2174","subject":"Merge \"Fix setting None in tls_versions and tls_ciphers in pools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/58bf439bda1cdeaa3dac8b453e05485092bd2174"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-10-28 10:02:29.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 09:13:11.000000000","tz":60},"subject":"Remove useless proto parameter from sample functions","message":"Remove useless proto parameter from sample functions\n\nSome functions in unit tests are creating namedtuples that represent\nloadbalancers, these sample loadbalancer objects contain a \u0027protocol\u0027\nfield that doesn\u0027t exist in real objects and isn\u0027t used.\n\nChange-Id: I8f191a720a3d38a7f579ed5f587563bfd40cce15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9653920a04e43a8229bcb710b51a110f2be93aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9653920a04e43a8229bcb710b51a110f2be93aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
