)]}'
{"id":"openstack%2Foctavia~760382","triplet_id":"openstack%2Foctavia~master~I53cf76182d2ba9f1aad81edc9da68bedd4a81aba","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I53cf76182d2ba9f1aad81edc9da68bedd4a81aba","subject":"Bump the minimum coverage to 92%","status":"MERGED","created":"2020-10-29 14:57:48.000000000","updated":"2021-01-31 02:05:34.000000000","submitted":"2021-01-31 02:03:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760382","meta_rev_id":"b08eecf4b33b1e262455a55554ef662a260b4bc0","_number":760382,"virtual_id_number":760382,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-31 02:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-01-31 02:05:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-31 02:05:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2021-01-31 02:05:34.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-01-29 23:02:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 17:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 10:49:14.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-10-30 14:57:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-28 10:49:09.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"}],"messages":[{"id":"f7f84f75ea8b1c824ba7ae424f0a39d53ad1a6c2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-29 14:57:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0568625267d3d55efcfeee2b0ac992dea82eb069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 17:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92c32fa317c944eab8dcb089f676e65e : SUCCESS in 6m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58731104c5db4cd889debef8e88db2e2 : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e075c8e22f840c9bb5e1ba611a6518b : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b6370cea6e4d3cb445f090bf6b9e28 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1abf670dc87f4fc0bce1064dae526136 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fee75b543224debbf0696c1bb5183c9 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae6e2470fdfc4bde8541e5e9021bf974 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c454d979eed41789a205ebb2b827d50 : SUCCESS in 4m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c2286bf19950416f99ce0e2dff6d1141 : SUCCESS in 7m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4df82a47c26a45159aad3b0ac6b01a52 : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cb320ad485fd46a98fb57cbf1aa46c4a : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cd415b06f9f24696abb1aa4020bdfa1a : SUCCESS in 8m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21e2525db8c547e183e99f11fc2afc21 : SUCCESS in 1h 37m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a3a99a246a64e95b03d33f7e2817cbe : SUCCESS in 2h 05m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/439898c0191b408c9dbe2a29606a1016 : SUCCESS in 46m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5ea5501d92c34e0ca4518fdeaab350f1 : SUCCESS in 49m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9a92016d01444a185f696c19049f1a1 : SUCCESS in 41m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d92f45b106794341910c5a16ee777043 : TIMED_OUT in 2h 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/29d9e03541db4b3ebb5bd680b050042c : FAILURE in 2h 29m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5da052a740ee4516a1356f92e45dee04 : SUCCESS in 1h 36m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2e92d89a44844cf7ac50a9de404d90ed : SUCCESS in 2h 17m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d6f7077b4d354913af679b4d13ba879c : SUCCESS in 2h 11m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e5821f61cd2b4a5abb241dff259b59c0 : FAILURE in 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d5fb20a6f43abc0034583346de061b84289d48c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-10-30 10:49:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f10516a7c525642597a0b565201bfc837e61425c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-30 14:57:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dfe4f2807b769233afacb79c2e5c8ea36e6fb24d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:49:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2374e7d010c1a8db4a3e8d6129b3c42417f98be5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:49:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d50680b103aab3c39752ceb61734bc35bca9e6d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 12:18:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e6608f0efe9945d6bc5a956bac45bb6c : CANCELED\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e8ea09ca99b44b8b1eb5f01944f92fc : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8af916e6e04168a9317285869f8340 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaf3b575be2147cba24bc0ab2854be84 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74032817431b4caaac6cbf920bc3a982 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78f81f96a484faaa219e653a181e309 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60cabadda1674ddca9d7af3f7ae24edf : SUCCESS in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c34a9d0a77754211a9aed132126af1a3 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7a085d4cbb2491898e95b22285859fc : FAILURE in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b364acdaaa034e219a3877e8ea958b16 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dd87a1c92644247bd6d9f2b42555e55 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a738bc3d8d0b45e38715842edf297fa3 : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7a9d805157704c118ce1e43f2d095b0b : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de9792b4910449f091c84351d02d5be7 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4cf6650b48404093bee02e4cabdd49c9 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"73648d633eafa3f965812d65983899cdae2384da","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 20:20:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0abacb01c8c3fe4611b14fc46138b67f5a828172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:21:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fde7d0375ef2483fb588218b2e70cc5b : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40057ecd2ea6459d8f5e07102f356448 : SUCCESS in 9m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b885d84590ae4c66ae98713d5be23de5 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a668e03794a24b26ba8a4784d47e0588 : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/541709a474124faaa6b219712e22d03e : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b89954209bc8488fb1c49d70b159510a : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ff3d6a201f4231bdf22db8c9e68296 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de4fe1e06e14857b90053068f7a9034 : SUCCESS in 26m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d504a9abab843418efdb53cdc8b52a4 : SUCCESS in 13m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/341411cdb93f4809819049158885873d : SUCCESS in 12m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4e24b63dd67b4100aa99f073564fe097 : SUCCESS in 16m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce788ef50c9b4b46a5f53609f9bacfa1 : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0fb6a799f6d546c59db256b25bf4baec : SUCCESS in 13m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a12ac98e37754c8abb532471320f96d1 : SUCCESS in 1h 30m 14s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b5cb1b239d646929d6059d49f423d59 : FAILURE in 2h 11m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91c1797d574840c0a1134c50b6b65805 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a115c38fe5f94ea392dce522099438dc : SUCCESS in 37m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59150754b408489da34ac7669b3552dd : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/99733b68bf274f7e94ad90e0f6015028 : SUCCESS in 1h 50m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3740c26e95e4422a9d3bc0ee3812b1a7 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fe4f243b79624a629c9db19162bc9339 : SUCCESS in 1h 35m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/331be7ce3abb43598a501679a7a0c650 : FAILURE in 21m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/59b41be84a3848998bb75e5b018d98fa : SUCCESS in 1h 46m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ef883a688395427cbb0195cde3685fe0 : FAILURE in 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb838ae494efae495b5f0aa6f5128b074aa5f24c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-29 23:02:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5f21d301c9ebc6d169a509a653846678e37188bf","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-29 23:02:24.000000000","message":"Patch Set 2: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7c21d2bc3cbeb54e57f41c035f796fbd06a8abfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 01:26:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/add55a6f65d54203b7dafaadc88a1909 : SUCCESS in 7m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d13c951e151b45788f0c216bff9f548d : SUCCESS in 9m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b471c11c78485ea05d39647c57cb26 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75db98e464564facadd7fbc54a9b5f17 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79301ad685654dad9aa20f15d2532fe0 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c0d5486f3e24e7991995238d2aef12d : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f96320c7bc1644ca8988622c3fde55dc : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfed2f38e9e42de93c213989250d87e : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/913937efea7746cf94a3b733dc3f02d7 : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fb44b63fbaff4ab281e2181c04923e59 : SUCCESS in 5m 03s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c2e4a26e0b854fd9a4b2b8139561b25e : SUCCESS in 6m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f727b6c686b941c59bae3237f9a1f4a3 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b6f8da99daf943ab8fcfc30ed64d39f1 : SUCCESS in 10m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d310702ad861400ab4cc3fc198e1aa14 : SUCCESS in 1h 10m 20s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68389cc0fbd848fdaf74f16e74d2ce13 : SUCCESS in 1h 40m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9884ed2881f0466f98a17bdfd8858952 : SUCCESS in 1h 08m 20s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cacc07e48c694c28ab95912ce2803143 : SUCCESS in 44m 29s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75b482e6897a4069acc011ececc97d3d : SUCCESS in 40m 24s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f23d9a5403c42efa82f9ee13a9fcc94 : SUCCESS in 1h 45m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/58ae5c4c762c48a095a38bfaede80514 : FAILURE in 2h 21m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a50201e46d8495fae5801da5dd4d797 : SUCCESS in 1h 44m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b608a4788cd74ed2987deebe93673dd3 : FAILURE in 19m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/de4d78fa436c47b99b30c5ebaf8d9659 : SUCCESS in 2h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3022ec689a1948e889c6c4852954f517 : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78e4ebd44ece367bcb45e1c3224ee4cb013f19af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 01:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4562367f462ef845ec39a86922b0bb13a84d6ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 02:28:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f9c31eb598bf45a59d72ddc2f1583a1b : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c38a3ce2ab454ce58d70a4c78570c4ab : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fede416a827040d5971286bb6e44bb51 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48421e3660234a84a27cd4a4a6115903 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42630e06a7754207bb88f075439ae92c : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a925a1a404c4fbe8a38fe65cc382c06 : SUCCESS in 15m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2379eb479b1d477b8092beef3d9e5054 : SUCCESS in 4m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3a00e4b176304a1790161f04a76e1e8f : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ba92b2c4aa24aae88b081480072f296 : SUCCESS in 9m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3054d2a9f3f419a8a24604f45a88e54 : FAILURE in 59m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd873c92b4df4592acceac031aa16c14 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/96dd495740c44835a19b95cc0080676b : SUCCESS in 39m 26s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/29be2ae83cba42019d81af31562b0ba3 : SUCCESS in 39m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ef78e48a67814567aedf5cbc7f58fcec : SUCCESS in 40m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db22365858204e3fad842d6b2c7591c9 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2a355c50774806ab9eed1da64f25d3f8d89f931e","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-30 02:46:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"24b6dd8ead55a817ab0bdb7516ad0dd4976ec3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 04:35:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d900da5d35bb4f248498309abce26939 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb0339741cc41a691d0c4622a659026 : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55316313f9a24832aaa9be38e3699bb2 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c520dd034f43433c804d813613a140c2 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9344d4f2ada471dbe5dae1030edd63b : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7310d802134f87b474f6d94fdb303e : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e45169b0b844e3c8e8055d153f6ea53 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5c0affa6714dc3afe11c7b53b18f06 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c9b5978c7604372a209ec57fd8c0902 : SUCCESS in 9m 09s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f842e68b816944d4ba601b9cc0726df3 : SUCCESS in 4m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/cd8227ef5e52424ea5398bf29d87d826 : SUCCESS in 7m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/80cd64573df545288f291ed3d8930701 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/381d118740b74671821c5fdb424d8e10 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2597037e8a2045e49a02d7c123d515c9 : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9cbc9d69f14c487abd1549d34f218365 : SUCCESS in 1h 44m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11e87ea022f84f96a06027bb96b7c9fc : SUCCESS in 39m 40s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bb267419abeb49bc8260ff95ae3f8b73 : SUCCESS in 35m 20s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d06436a930d94173af29b885447f4b5e : SUCCESS in 42m 32s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16247af3bd4242aea565308fe2b9ebcb : SUCCESS in 1h 43m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dfa49e6b2a124124b68a8c363b0cba85 : FAILURE in 1h 46m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/53dd94e18d024dae8b1007c48a775c9f : FAILURE in 1h 40m 22s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2377820f6b7f4df4a9ea7832348b1491 : FAILURE in 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8e487198f2a7474f85f9c8b54b3c37ca : SUCCESS in 1h 44m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2028ffb9f6c745d9b785d80eee34c693 : FAILURE in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b6d25de637acbb110b4ed76c489082944dd9742d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 04:35:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f7a6ddb660055f4912f6179d467964e17ffc5e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 06:33:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d500700b9b8e4fa59bdf857838daaed9 : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14b984ff8bcc4cd3b9fe7496b6d194a6 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc34ac8da63d483da909b32232c6880d : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/727d820f933a47e7afc4887e2fcb0d4e : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2436e1db34984e7b8fe7e566614ff3bd : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a951437065045468edeacc492076d14 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ded5859f0e064e6ba573db294f3c0315 : SUCCESS in 4m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e1b7a44cba6846ffa92e84de29d1e2d6 : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7ae3e20c67b4604864a4ebf840c85ca : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/26ed138adb94437c96e7a7dcfa59937e : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06eaf93f2cfe4131b523d7e201b28b72 : FAILURE in 1h 56m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eabca383dbae48489c9affea8de7edeb : SUCCESS in 41m 16s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da77a1d773594f20866fc5a3d53fc77d : SUCCESS in 39m 21s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fef45d69a80b4d3185577464bc0bded5 : SUCCESS in 38m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b3735948d49744d5aa56c3200c2467d5 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0addb53ab59d4ffd6a97180f4211b180711c191a","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-30 21:16:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3d6830cb25ae0e98e7e4347fa174cd519f9d3b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 23:12:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c29d74e3b8ea431a9e93f348cf175c5f : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1524789aa647ed9c2498335914e8e5 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b96aa5f25d33455a9e2020ae1ed6c448 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06474f3cd18c49f99ef5989e1833f7f4 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69da3313ebb64270bacda5df6000cc1c : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c958fac26fa4fb4ba9e2c21301d66cf : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bc0866957614b16b4633ce96d2cf226 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b568fdc20f8c4cebbf86d296f9ea3374 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b7281110b9e48ada4c35e76401a913d : SUCCESS in 4m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/baf10ef55156494d8c6770298cd8ec89 : SUCCESS in 4m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb48ad83844a43dea8ddc72f20429763 : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/716e660826284878ae2d4a71baa21a7c : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15b70a2c07cb49ff9b6647f97be7e51d : SUCCESS in 6m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b8bcd7cfe08442cb87d68b1f8b3c9bb2 : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/650d2014fd60452b9322bcf6fc74b866 : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/505cbdd3e04d425e857703e3d630715b : SUCCESS in 38m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/b14196a838b0460eae0454609271fefa : SUCCESS in 47m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dbe99f756fe148ca8b13968c77db4c4a : SUCCESS in 37m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d5b50bc7a044f6588b183cbbcbda210 : SUCCESS in 1h 46m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5f6db609c454f08a9d9c5e4861ca44e : FAILURE in 1h 53m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/84de38e70b614d8996d7b2ca8786a775 : SUCCESS in 1h 53m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/103b4e2e69f347a99ea5e82f4c0af24d : FAILURE in 21m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/863544d6035c4b62ba66f336dc5226ff : SUCCESS in 1h 49m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/a66c3127c3f3459e9d4d28e449caf798 : FAILURE in 30m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be5e214c256a0ffc67c6348c78ffb90d03fcc084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 23:12:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f6003f75d2a42e717193dd9d91cea558ae28dbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 02:03:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e633ddf428564c72bb3a44b7459c0b8e : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f483df731004be09faa2c84a23a0534 : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca2cc5696b0469d9ad461ba14d78a5b : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed7dd6ac31dd4eba889ea445ac79758e : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f9688a88364822aeee6b3aa26d5509 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eabd483c616e4a888742fbbf072aad38 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9335b740cd674087b44eb49588914e17 : SUCCESS in 5m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b52283f60d34465c93240fced986d6c7 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2c357232761249e991e0154b365ad942 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/00dea44c5ce041c7a2e3cdaf992aed00 : SUCCESS in 1h 17m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17260d4f9f7f45e3baf924bed9d3bea9 : SUCCESS in 2h 07m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/85107fc4127b47149cbf9d981891f559 : SUCCESS in 41m 46s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/80e369577db34efb96a5433d0651b1b9 : SUCCESS in 39m 10s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb4f9c71ec2c4ed4a60e87ee6730f326 : SUCCESS in 40m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/95469d9b00544910af96daffdb4177ac : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":2},{"id":"104f8d50ae9be529e333efbf110898d705b501ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 02:03:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7df4e61383010c9cc659e35d71cefdd2e0c4a5ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-31 02:05:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5b1103cf109246d2a71ae6f37cb6d346 : SUCCESS in 1m 03s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a04b58449749fb943894bcaf5dfe50 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65f6f3c6f8f7437c9a9ea102935ed77a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b2c6425a3edb6ef26caab3ab4c62a12e57311fa","revisions":{"0399cad84d53cbcb652040163b7815708da4d70f":{"kind":"REWORK","_number":1,"created":"2020-10-29 14:57:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/82/760382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/760382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/760382/1"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 14:56:37.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 14:57:47.000000000","tz":-420},"subject":"Bump the minimum coverage to 92%","message":"Bump the minimum coverage to 92%\n\nAs we continue to improve our code coverage we can now bump the\ngate minimum to 92% minimum.\n\nChange-Id: I53cf76182d2ba9f1aad81edc9da68bedd4a81aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0399cad84d53cbcb652040163b7815708da4d70f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0399cad84d53cbcb652040163b7815708da4d70f"}]},"branch":"refs/heads/master"},"9b2c6425a3edb6ef26caab3ab4c62a12e57311fa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-29 23:02:18.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/82/760382/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/760382/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/760382/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/760382/2"}}},"commit":{"parents":[{"commit":"673a5691c19008d680423f4af6200cd4d9f8f757","subject":"Merge \"Support deletion of all tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/673a5691c19008d680423f4af6200cd4d9f8f757"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 14:56:37.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-01-29 23:02:18.000000000","tz":0},"subject":"Bump the minimum coverage to 92%","message":"Bump the minimum coverage to 92%\n\nAs we continue to improve our code coverage we can now bump the\ngate minimum to 92% minimum.\n\nChange-Id: I53cf76182d2ba9f1aad81edc9da68bedd4a81aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b2c6425a3edb6ef26caab3ab4c62a12e57311fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b2c6425a3edb6ef26caab3ab4c62a12e57311fa"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
