)]}'
{"id":"openstack%2Foctavia~760461","triplet_id":"openstack%2Foctavia~master~I736816247131715f5c385b4680614ec3218a2ad7","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I736816247131715f5c385b4680614ec3218a2ad7","subject":"Fix pools going into ERROR when updating the pool","status":"MERGED","created":"2020-10-29 21:54:45.000000000","updated":"2021-02-17 08:28:28.000000000","submitted":"2021-02-17 08:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760461","meta_rev_id":"a35a491e488cd46a81dfb6fab5496ea2a28306dd","_number":760461,"virtual_id_number":760461,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 08:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2021-02-17 06:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":1,"date":"2021-02-12 14:22:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 11:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2021-02-17 06:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-11 11:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 19:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-30 20:09:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-03 08:24:09.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2021-01-28 11:03:54.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-02-11 11:01:22.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"f09af1683cfd5544cb1ac235b1c80c2ce351b5ae","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-29 21:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b1d07b2da332fcd30c949483b21d9421875b8a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-29 21:55:14.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"cd24a98bcea7347500e887eb3ebf009eba7408df","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-29 23:35:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c05bcbe666f3cf6895f3743dca1ab7562a49be3b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-30 01:23:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8abb8d700e4f68a547094487cb3916078b8cae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 04:02:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/039eb3c926a849de9d7f3cd5cbf1eccd : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc1a3e28f3e940c08c043dd18ac26d83 : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3055505bde0e42f69615d1cef82359b0 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f032f4bf810439694d336feb227ee0e : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ef41f870bb74408a5578eeddd2e4b03 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8f0d009da041dbac3067530e14ce0a : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91420968deb240b599b88b42d6d2ac33 : SUCCESS in 9m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/91d952a0f4f443aaa035389069f90827 : SUCCESS in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d089640641254ba58cbe561c4e3fe711 : SUCCESS in 9m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa5e85e8f49340ac8bf89135d32e8a11 : SUCCESS in 5m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e235326cc12249399a6e5063a6768b58 : SUCCESS in 8m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffe615ba18024e6692b53cc6f1017ce7 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b3fe108cb9845f989fa2048dee148e8 : FAILURE in 2h 05m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39d54fd9ff884cdcaac7fa6b34ea25e1 : SUCCESS in 48m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c5a30ca8913440449bdd606ee3044366 : SUCCESS in 41m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c81765923c22467b8609ff6bdd09f9fc : SUCCESS in 44m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/533936cb3d3045a7b60727b7b4ec7327 : SUCCESS in 58m 57s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b0ac83b2b824313a190431e3bbb38ae : SUCCESS in 2h 17m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b902b96b08694f26a44849ad1804f1b5 : SUCCESS in 2h 13m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f586512cd631414da0aec945b8689fb3 : SUCCESS in 1h 51m 51s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/05390d28c28d4de3983429ad2fceb282 : FAILURE in 2h 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d2dea610c3f04634ba772bd7a0e6243f : TIMED_OUT in 2h 35m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f30cf37f87f44d7ba212ead134c5d491 : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2274ec4786f3487eb802857b53176bc79a397a65","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-30 17:26:15.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"229decca270a11fb956c641880cdb6090145eb60","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-30 18:06:46.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"4ec2e0cb1b90ed82f759f1b24814bc13c32a816c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 19:59:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d9a1982a78a42fb8ae36061abc8d097 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6af6dc47e54a4138a83cc5ecff568f02 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a1d7ea7c214b0d8665849668204180 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9408766cff134e2ca2930e77632aa2bb : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dfef70b867a4412943494e3534c2b7d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e471a41f66b41c6aec9c7de4dd64fe0 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a17fe23636974c3b87046bf7371b4883 : SUCCESS in 5m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fa225d7845c34b6f906fbd5a9143f676 : SUCCESS in 5m 07s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0be974b97de472bbe7af1a88d55d103 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/252e41aca89f4f07a9b1087c850f8754 : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa297295e65b403689cb4f2fa2261475 : SUCCESS in 6m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c25732d1298f4335a359574da2bb0e05 : SUCCESS in 1h 27m 55s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f86124fbadea447dad5d3871e39ce46e : SUCCESS in 1h 33m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5da3514cc5034adeab2fe2afe652d6f7 : SUCCESS in 36m 27s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/ddd847cbe6e34d8b9c37887e87ab8a1e : SUCCESS in 38m 08s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f14ebe2bcfd249d6b0fa59b76732c983 : SUCCESS in 34m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/011fe01fb34b418cb71ea5436566ad11 : SUCCESS in 1h 03m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a676e24a3ae24f7fbb84d6ae2f280414 : FAILURE in 28m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/63b28fb217cf44c8b7331e4731ab1b09 : SUCCESS in 1h 54m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/32774b5247df47a281c7f7925f50202c : SUCCESS in 2h 06m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/af8b986acdac4ef4a188b4bb5968185f : SUCCESS in 2h 24m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5dd83398a71a4f308968143236358c08 : FAILURE in 2h 06m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5a4b3edcefd44da98eec9b2f3c821c3c : FAILURE in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36add411e5afae5df041c26334c7434b2d052f7b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-30 20:09:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2cf5e986e4c6fc743d391c5f19a9372e50a30764","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 08:24:09.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eaf93ac0997e51974b12817cddcb339d6c5cfa10","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-14 18:19:15.000000000","message":"Patch Set 3:\n\nrecheck\n\nThe dependent tempest test is failing, just wondering if this change is still Ok.","accounts_in_message":[],"_revision_number":3},{"id":"d0b3f54fb93652646b174f3b3350c7abff81db11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 22:41:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/016ef84ce06846af8aebff85da2d3ac4 : SUCCESS in 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/242f72cda2ad4a5da2971281b115e253 : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d1a96b35654c248bdc45e991c9ff9c : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef6931d2b324f0f996246e5883606ef : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f37e0a841e9c4e5d9c5ac22a0a1a0454 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f1f35217bc041acadaa4281a927c3c8 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f490bb8c563e439b9922667b2f0b19f3 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3a8cb0406e40ec89e9ffe82d45b8c2 : SUCCESS in 5m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/77115aa0c65d489ebb8bcd52388e72e5 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/011375bdb27c48d7adad6c37e431a7c3 : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28a656a650464c849ffa13950938498f : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a200eddc850744f5882e252fb513df80 : SUCCESS in 7m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ad8899eac46e49d8a1866a7c87852350 : SUCCESS in 1h 55m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/55bc0609894b40e49acc00c2a3b26be9 : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcb8c5d2305c4c74b499a83715e65751 : SUCCESS in 43m 31s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c0c3775628c64c6bbea5d563be2686eb : SUCCESS in 39m 09s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1b6406a61944db4bfd7687f2a750326 : SUCCESS in 40m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8cf2037ea3541c48d54fa37e92fa125 : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ceaf6236fe2641afb7c887de3efdc4d7 : SUCCESS in 1h 30m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8896752b535d46fabf3530708bc8b686 : FAILURE in 1h 57m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db8ae73fe62e428d97ba20114f30d547 : SUCCESS in 2h 07m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d866e1d64c3140068bf7865ee32f6644 : SUCCESS in 1h 53m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16bcd1d1f64a47139a706ce02eb8b15a : SUCCESS in 1h 53m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/8d4bf955556a432781f17b88ae67e960 : FAILURE in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b6bc8dc007f716291a41836010079148f3f087b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 11:03:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d23963ff207b0cc82f1cd98de6fd84317514ce4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 11:04:13.000000000","message":"Patch Set 4:\n\nJust rebased...","accounts_in_message":[],"_revision_number":4},{"id":"e7700a3916c2b13c8f179510cca5799296d83262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:55:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b6dcd67cbcf408bbdb0c24ce8e07189 : FAILURE in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32736832568b49098d126c290e7ba611 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fbf211a5c8b4b228852f01602fd5b18 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5fda653219949c4b3a2a0955bd37226 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b4752860c1e4c39976782f0878a9dbc : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc9a97761db14120abbf537fa6d4975a : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88a2802b844a4202b2ae714605eb4a1e : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cb30f81836542319f2036fe84477941 : SUCCESS in 4m 55s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9641e74358e44494af105644fb17a404 : SUCCESS in 5m 57s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c0fbd8ba36934d9a8b6aede48b70c7dc : FAILURE in 7m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/27a9207cd4944a59b0a9afe56ddc7d19 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4cd02877ac0447ab06655e99aeef45c : FAILURE in 8m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5f4999b427f465caff71ae38ba26a13 : SUCCESS in 1h 41m 23s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9f0dc48b0e842178e3d1fcff039c8de : SUCCESS in 2h 03m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7f44e822527c49d0bb90451e4de75aab : SUCCESS in 40m 19s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2d3b090b3cc24cc1807161dab32b91e0 : SUCCESS in 42m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a4060558575f46ef834d1bf227bf0497 : SUCCESS in 42m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bb114c565d1e49aa9dda466aa7b97e42 : SUCCESS in 1h 03m 53s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c0f161e64ab4fe5a9b852f9a5cc9e63 : SUCCESS in 2h 14m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e160c8b952b477a904c96ee5e938b1c : FAILURE in 2h 21m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/82450dced93644758c0a427287c89e10 : SUCCESS in 2h 12m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f2b9003be9b4b889d829dc9386dd3d1 : FAILURE in 24m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/739647b16f3c478ca522f1101a31ecc8 : SUCCESS in 2h 16m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7ae2c1c87a354606b6b5a1b9e1810eb1 : FAILURE in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e0ce20e4d77db8fbf03757bc5d9e6446cd42dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-11 11:01:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"29c04fd2cb106a7954232960c04e487c6c2839af","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-11 11:01:44.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c6338f0f4a241a251a98ed05e89185d4c82f58ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:15:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8d1b5212d594f4d996d6a6db73d153b : SUCCESS in 10m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5c4c8d597fe4828a4c314baa6732556 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b409d3b929e14ee889b27b0d739fb9c9 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f836a2c74645729b2bcc7070ad3a7d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/086399934f63493194bec4975d173d36 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d9f833a9f8c4aef85c806e972c2bd2b : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bdf26e252174d2fb00619adf143636e : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ec0c7b06819495f832a3a9e08a6afb6 : SUCCESS in 4m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95009e1a082445a28e41fc99cc65eea3 : SUCCESS in 5m 14s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fff2f17958674911abe59a20495c2a70 : SUCCESS in 13m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3fa1c33de9ef45f4a87eee59b11fcde4 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/597e9b390b7b4b728835412666030aa5 : SUCCESS in 9m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fb0537d02acd4cc798afba025d1c94d8 : POST_FAILURE in 1h 21m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/14a162a4aefc4af7be79029ea1ab9f7c : SUCCESS in 1h 33m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/161637123d7d4571a77bd816c7b41737 : FAILURE in 38m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/68286f44a67d401cb5c0d85e0696d4d6 : SUCCESS in 45m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b105eb5f50a2421d9667cd2841cf645b : SUCCESS in 55m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37a726ed72ea44d7a8077c33c13a61ff : SUCCESS in 1h 17m 40s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/47b0c7e48edd433a96070882a392000c : SUCCESS in 2h 01m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/838130da833d461aa6d8b419d44014be : POST_FAILURE in 2h 36m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1acd94db94944f1adaeb067c13f950f : SUCCESS in 1h 54m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1b8c275703a04c9d9bdd266acc17eb57 : POST_FAILURE in 1h 53m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/698d5dfa661b48c5ba06a4c49c82d529 : SUCCESS in 2h 02m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/03a4f2af4aea4078b0ca6425f752c351 : FAILURE in 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92c38718e07004c6565539fac229e09319aa7790","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-02-12 08:08:39.000000000","message":"Patch Set 5:\n\nrecheck\n\nPOST_FAILURE \u0026 timeouts while downloading some modules with pip","accounts_in_message":[],"_revision_number":5},{"id":"c523a2afbc3b0a5563a10d55f6049366097404d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 10:11:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476f5e7b44cc4b4ea27217a22051f4c3 : SUCCESS in 9m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fcd0061e3244d20950cd7f764fa95a2 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38565de36ff84c00be74fdb087ec6732 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5817a39e9574c4982c51fb9f23f3bd2 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ecd6dd9f354d12895076fe684f0496 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c3b9d1d47dd495a82bcf8b8e392b8fc : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321ef5884f7d4cc3a5de6e444a65bffb : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4f0e5e89d584ddcb463f356f3a0dde7 : SUCCESS in 5m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a2cb340c17904a24bd443f9c7d6dce05 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b8a17f9bd6c4da29b664cbecee6b182 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5131a3f336b4b1180eb803500ac6428 : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b30fc7090d3462c9361d509a927dc5a : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4f72938ddc94991ab682e292d3a7342 : SUCCESS in 1h 20m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7597133d4fa1478e938f44b3eff0e852 : SUCCESS in 2h 00m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d515a58ced7643f6b417e0fc6dd8ce0f : SUCCESS in 37m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/bf0325b43439447d862d019ed240cd17 : SUCCESS in 37m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fbe259038a904b2e9e84c3fabf8d36a8 : SUCCESS in 41m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c766e21f770448a9e0f2f8d1c906e3c : SUCCESS in 48m 41s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4525611ecd324f2ab97b535b8317d891 : SUCCESS in 1h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4511ff0b71f548c2b126c852f4d17226 : FAILURE in 1h 50m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/533ca8e5d4674e97ba5851037302452e : SUCCESS in 1h 20m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d6c355f9d8a846d59a12510cc9a32bfa : SUCCESS in 1h 29m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1aeff10299c041f2890105d5fdf7906f : SUCCESS in 1h 41m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/cad5de6622e44317b4962f45dbd473a2 : FAILURE in 26m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"888bdd28b2692cf162e56513d717ad3595c8753b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-12 14:22:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"59de34737868c7885a9512f8bc42306c27a9e119","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2021-02-17 06:48:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"163412c4d456dc8a0f9dd256249cb846c3d87de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 06:48:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"84e417750dc4a53764cf0228f0950f7e7b37eef8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 08:26:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/095e29752f8144ca8cbe95e475ef847e : SUCCESS in 9m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfe1f6ee4e684419867f17b13a45f07f : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89d67cde4ad34e94b572aa7478670c77 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f4ffb18d85433eb71b269f3392c3ce : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7072679671ad4966ada83b495dd2f02a : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ac9654943d4c6982414f6ab514f2b5 : SUCCESS in 5m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/40f7ebaafcf34bab89f3a21bc4b6d7a1 : SUCCESS in 5m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a9035a3d631409e8ae7fc76234a8fba : SUCCESS in 8m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d3c50bcc7ab44a5b05d7f4d7c263700 : SUCCESS in 1h 25m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5afe9f67687a40da8f58aadd70521185 : SUCCESS in 1h 34m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1bc49964c92646ce80cf208832db3b93 : SUCCESS in 44m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2a6af19b01514de8a4f6b1eb0ff02fb0 : SUCCESS in 38m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cd76ece0bbbe4a2e9194a415e99e0046 : SUCCESS in 37m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/591e00d1e02f41ccbb0691f56302add5 : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":5},{"id":"65fb1ffcebd7d87c6b0d5cf0105337ac3641df0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 08:26:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"a35a491e488cd46a81dfb6fab5496ea2a28306dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 08:28:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be51911849864f2faf414d5414466974 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b5cb93a5e34ff6b8f3faa4eac41f19 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"370aa4e61c421311b06d986aafdf123cd6737fd8","revisions":{"9f8a7819a86681227be65e49199e297c0dc8805b":{"kind":"REWORK","_number":1,"created":"2020-10-29 21:54:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/760461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/760461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/760461/1"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:54:43.000000000","tz":-420},"subject":"Fix pools going into ERROR when updating the CRL","message":"Fix pools going into ERROR when updating the CRL\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f8a7819a86681227be65e49199e297c0dc8805b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f8a7819a86681227be65e49199e297c0dc8805b"}]},"branch":"refs/heads/master"},"f601d8a3b7a864ccbb5925b8067297a6d70f9e54":{"kind":"REWORK","_number":2,"created":"2020-10-29 23:35:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/760461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/760461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/760461/2"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 23:32:19.000000000","tz":-420},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f601d8a3b7a864ccbb5925b8067297a6d70f9e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f601d8a3b7a864ccbb5925b8067297a6d70f9e54"}]},"branch":"refs/heads/master"},"af4a5a57ae7a7464f49abb65fa08ed1ca1137b3f":{"kind":"REWORK","_number":3,"created":"2020-10-30 01:23:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/760461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/760461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/760461/3"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-30 01:23:01.000000000","tz":-420},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af4a5a57ae7a7464f49abb65fa08ed1ca1137b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af4a5a57ae7a7464f49abb65fa08ed1ca1137b3f"}]},"branch":"refs/heads/master"},"cfcbe46bc14e2352e55978b089540b37df15e507":{"kind":"REWORK","_number":4,"created":"2021-01-28 11:03:54.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/61/760461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/760461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/760461/4"}}},"commit":{"parents":[{"commit":"fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33","subject":"Validate listener protocol in amphora driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fbbc5f9024b0408d5ff70bb8f2c4c690f0304c33"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-01-28 11:03:42.000000000","tz":-480},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfcbe46bc14e2352e55978b089540b37df15e507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cfcbe46bc14e2352e55978b089540b37df15e507"}]},"branch":"refs/heads/master"},"370aa4e61c421311b06d986aafdf123cd6737fd8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-11 11:01:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/760461/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/760461/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/760461/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/760461/5"}}},"commit":{"parents":[{"commit":"a822f30eb135b31f1e5fb212810f99f5d5adf477","subject":"Bump oslo.policy version to 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a822f30eb135b31f1e5fb212810f99f5d5adf477"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-10-29 21:52:32.000000000","tz":-420},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-02-11 10:50:41.000000000","tz":60},"subject":"Fix pools going into ERROR when updating the pool","message":"Fix pools going into ERROR when updating the pool\n\nThere was a bug that would cause a pool to go into ERROR if you attempted\nto update the CRL or client certificate on the pool.\n\nChange-Id: I736816247131715f5c385b4680614ec3218a2ad7\nStory: 2008295\nTask: 41180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370aa4e61c421311b06d986aafdf123cd6737fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/370aa4e61c421311b06d986aafdf123cd6737fd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
