)]}'
{"id":"openstack%2Foctavia~760552","triplet_id":"openstack%2Foctavia~master~I8020ed9299b4dd39f943ce4768bcea44ef64c0e6","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29157":{"account":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"last_update":"2024-08-22 00:01:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8020ed9299b4dd39f943ce4768bcea44ef64c0e6","subject":"Replace md5 with oslo version","status":"ABANDONED","created":"2020-10-30 13:24:43.000000000","updated":"2024-08-22 00:01:52.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa741addac3ccd5b3f098f67d456a937ce7a085d","_number":760552,"virtual_id_number":760552,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-10-31 11:00:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":-1,"date":"2024-04-30 16:46:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 15:02:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-10-31 11:00:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 22:24:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REMOVED"},{"updated":"2024-04-30 16:46:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"ec21747e8b0e9e6ff0dee297d2c330c70f368427","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-30 13:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daf88d2b9ada0812b2e3cb77fa36ee9233ac0e6c","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-30 13:28:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"490af24785cbb51890ea16483be235699ecc4a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 16:13:05.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d360d063d12479eb0f7af8957000df3 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f293d14aaaf8475198991b873d5084d6 : FAILURE in 5m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28f568ae4f0f4e5b895d9417260755a1 : FAILURE in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43552e7d4a34ada8bc376af77911520 : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c31fad738cc9443d8451230cb79e47b0 : FAILURE in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af40da55aa7542e69a270eeab237c85b : FAILURE in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3592fa23554e4f43884847f3e49a042c : SUCCESS in 12m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3e48240b868e4c499b0b530d945016e6 : FAILURE in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b8a0332f973456ca8041953bda2cd05 : FAILURE in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0198f4abb7644104b4050cd572dccd37 : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22e9fd305f13403e8098e10366f96ddd : FAILURE in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/759bec1d07be4f5fb0dacd813958ddb5 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ca8ead74a7a24765b72ac6e576985520 : FAILURE in 2h 15m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/157d020d58f040c8900d78763ce49374 : FAILURE in 1h 02m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5e3acdbc314b4ff9a2df1d96f6ab9a26 : FAILURE in 53m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e1b9830c6cf454289540fb17419542e : FAILURE in 57m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ddde7f1deae84a7182eb437eecf76205 : SUCCESS in 1h 02m 51s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/da3d3fda5eb144de9c77fcff7bdcf0bd : FAILURE in 2h 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/12d9fe421383460a9c84c80c2ea95c7b : FAILURE in 2h 12m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d27ddd510fc4169b281c8711d67d038 : FAILURE in 2h 20m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2dc240ee9483481ca65761c46ba4f343 : FAILURE in 2h 31m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9f3e12eef4754d05af90ca274a82bce7 : FAILURE in 2h 22m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d0172b16b0034968b782bcd6b1c4855f : FAILURE in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80e2cdf2d360124bb68f7b62f6674e0420206344","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-30 16:45:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f42350828a4aa42f25d3d7275a300b4690633a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 19:32:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/99d289c07be24f23b3277422df4830d2 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de84b03058314c1baf73abb6753f8150 : SUCCESS in 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b352fbbe5b9b4420ab35a658d4fe001b : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ce5fa55ec1452eb81d8faffb9f83f8 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d7b9dc30ef45c38447448d340dbe48 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d17ce761247f42a4b28cf3dc28c8fc81 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4ad9536b054eae9754290a07432989 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1268578d3b447cebb9bd1f9f6538e34 : SUCCESS in 5m 29s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3580443884c145289c96b8043bbe7ff2 : SUCCESS in 8m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5c0132f7fcb04290be9f6a969367b97f : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9914f5a774fb44879817b66eb4971e90 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/615f6f3273ee483181383871be505ee5 : SUCCESS in 1h 14m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5028a35cde2949988d02f0d1da6a5499 : FAILURE in 2h 28m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6090006a59a42bf991406786ccb6509 : FAILURE in 1h 01m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/32b7c34fe1be49428771264630ad4a59 : FAILURE in 47m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe8f191294884398b4abc0059dd522cc : FAILURE in 1h 46m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/211762fd469b4b8e81fd5d4b67b4e155 : SUCCESS in 1h 04m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fdb04c8bd25d4113946dc8af12993864 : FAILURE in 2h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5f6c8052a6d4bc1992936582b9d34ad : TIMED_OUT in 2h 32m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/094e0f62f5ea4f22898cb40729ad920b : FAILURE in 2h 13m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1d3accfec5b34eac9b3eed2deca4bf40 : FAILURE in 2h 11m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9ba7c1e67773483baf4b5762153cea68 : FAILURE in 2h 11m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/1b2f24cdbcfd413693f8233825981c61 : FAILURE in 36m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2000c0f3e90ed05944700e7c8f3bf805f86e31df","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-31 02:53:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f3c9ea4f02da599fb229db6e60510eaddbdb3f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 05:28:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae1c1a9f771e445c99075745c9044027 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f474d3215629400e8ffe47e3b8eb2bc0 : SUCCESS in 10m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6cefb38cff481195d796bd8504087b : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e7aa348f0934fb2b0596ea55bfa1a46 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0829496eb40488c88d1afcc52614671 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec3255e6df3b4e45abb999eac9f0f6df : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae908e5bce72488dbb8640bfbd4b66b5 : SUCCESS in 27m 20s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/542a03488a16439d89922b9c26430853 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c5c98d874bda4d4d9741d2096efd9173 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e5b19f5f4fa84859941bced84719ca29 : SUCCESS in 4m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/419439e3c53848ee953f55c3ab53ad18 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/510324b7319e43d48752bfc0fd35413d : SUCCESS in 1h 47m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/747f56d3a8fb4241bc3d01b081113478 : FAILURE in 2h 14m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/afc8c3a2a39147bbbdea5157cc91d1bf : FAILURE in 1h 02m 13s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/516b411956024f718a9f514212c7c796 : FAILURE in 52m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de1ed70b3d864d0493a573d41bc2e85c : FAILURE in 53m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a209a92790264610904ce6b5b6588b91 : SUCCESS in 58m 25s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbd325452f6348d4af1d8ec7f5c3dd6b : FAILURE in 26m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c67219f45b43428898148e7c28bacf87 : TIMED_OUT in 2h 31m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/506585dfea384f74af3e2f9f9ba479c6 : FAILURE in 2h 12m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/38e2918bc0e8413f9f5b2098eba9392c : FAILURE in 2h 06m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b4cb02887fb0423a905b64afe142662a : FAILURE in 2h 12m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dc3d745ed97447308985400f73e2f1bb : FAILURE in 30m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d04e87e7d2af1ac1d16cedc3629bc94ed507093","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-31 09:07:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8f426064061260522d2f42726300b0d7700ec5e3","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2020-10-31 09:10:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b8ed88e87c65bd05a5e0dd3b7ab757302ee1b2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-31 11:00:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1422a14c461349299ad5353a272e9574 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df3eedb808b4cb3876be662eb7d96ca : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92f1a8aa7d474db68f5dfd2ce1fe8f03 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdfc3f8bbc644621ac955fb732abe516 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4040d3f7ff264b93ade04a8bc5357be7 : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f918c77b9f9e41809e5fb67be942d80e : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4cf85cda76d4099b6cec3b2b89ea425 : SUCCESS in 11m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e47600faf1d4f848fbea2f7b7bd0693 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/79d1340cf1d1429b9e5f16386929fbe2 : SUCCESS in 7m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6402152ba6c546c0a0ae6dbf032cbcfc : SUCCESS in 4m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c6512e3ba7f436daece8cf010cb357d : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e05c3585eaed456b8ce8bfb5b0f3399b : SUCCESS in 1h 22m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/631365e6d2dd4edbbdbea3ecf57c9bfd : FAILURE in 1h 27m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/751bd4cf7c024884b60cf95655186bec : FAILURE in 1h 10m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/64b251c06df846b2a73204c900729f44 : FAILURE in 32m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a1599d95ccd4463fa1689672785fa086 : FAILURE in 34m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6635d72229b6480299b4d42197d549d6 : SUCCESS in 59m 44s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/13499637486b4fdfbe08f1fa01cce498 : FAILURE in 1h 47m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/80734daeb1634859909ae95e39f9efee : FAILURE in 1h 46m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/65b1920d2095486789ed8ef66fef7430 : FAILURE in 1h 10m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/794eaa8050564d6398bf4d23b166e03a : FAILURE in 1h 10m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0b224755b03c4d3193223c3432955cb4 : FAILURE in 1h 02m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d08966cd45a64f2a9f97731b96cee61a : FAILURE in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ec7e4c1a30ce968dabf769974b36e158db0b4f8c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-29 22:24:47.000000000","message":"Removed reviewer Brian Haley.","accounts_in_message":[],"_revision_number":5},{"id":"79ee8752977261c92b0e647e13b94a57c2cc01c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 16:46:59.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa741addac3ccd5b3f098f67d456a937ce7a085d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 00:01:52.000000000","message":"Abandoned\n\nUnfortunately a duplicate of 798146","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8addd38724d33b91855fe17e971c4d9b679fd1dd","revisions":{"239c6ab0e9ccbed3f48371f389eda11fb90893ba":{"kind":"REWORK","_number":1,"created":"2020-10-30 13:24:43.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/52/760552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/760552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/760552/1"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:05:29.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:11:51.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nDepends-On: https://review.opendev.org/#/c/760160\nChange-Id: I8020ed9299b4dd39f943ce4768bcea44ef64c0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/239c6ab0e9ccbed3f48371f389eda11fb90893ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/239c6ab0e9ccbed3f48371f389eda11fb90893ba"}]},"branch":"refs/heads/master"},"ef1d0e694722bd299ef5be94787adfb887ff6eee":{"kind":"REWORK","_number":2,"created":"2020-10-30 13:28:09.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/52/760552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/760552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/760552/2"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:05:29.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:19:38.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nDepends-On: https://review.opendev.org/#/c/760160\nChange-Id: I8020ed9299b4dd39f943ce4768bcea44ef64c0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef1d0e694722bd299ef5be94787adfb887ff6eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef1d0e694722bd299ef5be94787adfb887ff6eee"}]},"branch":"refs/heads/master"},"04587d0f4176be7788078b2ca5c55d95aaa81d46":{"kind":"REWORK","_number":3,"created":"2020-10-30 16:45:14.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/52/760552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/760552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/760552/3"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:05:29.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 04:20:17.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nDepends-On: https://review.opendev.org/#/c/760160\nChange-Id: I8020ed9299b4dd39f943ce4768bcea44ef64c0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04587d0f4176be7788078b2ca5c55d95aaa81d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/04587d0f4176be7788078b2ca5c55d95aaa81d46"}]},"branch":"refs/heads/master"},"0f920aa1fe524cc03d3ea5bdcb01c4e6ab399d23":{"kind":"REWORK","_number":4,"created":"2020-10-31 09:07:30.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/52/760552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/760552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/760552/4"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:05:29.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 14:16:04.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nDepends-On: https://review.opendev.org/#/c/760160\nChange-Id: I8020ed9299b4dd39f943ce4768bcea44ef64c0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f920aa1fe524cc03d3ea5bdcb01c4e6ab399d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f920aa1fe524cc03d3ea5bdcb01c4e6ab399d23"}]},"branch":"refs/heads/master"},"8addd38724d33b91855fe17e971c4d9b679fd1dd":{"kind":"REWORK","_number":5,"created":"2020-10-31 09:10:46.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/52/760552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/760552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/760552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/760552/5"}}},"commit":{"parents":[{"commit":"7f26a5d5b496826705b5178873130fcd24a67b29","subject":"Merge \"Stop install unnecessary packages when run tox -ereleasenotes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7f26a5d5b496826705b5178873130fcd24a67b29"}]}],"author":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 01:05:29.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2_yewu@cmss.chinamobile.com","date":"2020-10-31 14:19:11.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nDepends-On: https://review.opendev.org/#/c/760160\nChange-Id: I8020ed9299b4dd39f943ce4768bcea44ef64c0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8addd38724d33b91855fe17e971c4d9b679fd1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8addd38724d33b91855fe17e971c4d9b679fd1dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
