)]}'
{"id":"openstack%2Foctavia~760711","triplet_id":"openstack%2Foctavia~master~I82d482bba3e4c40913583fb26f692824c56e6ac1","project":"openstack/octavia","branch":"master","topic":"tox_constraints_file","hashtags":[],"change_id":"I82d482bba3e4c40913583fb26f692824c56e6ac1","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-11-01 08:35:08.000000000","updated":"2021-06-24 10:00:12.000000000","submitted":"2021-06-24 09:58:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760711-tox_constraints_file","meta_rev_id":"fd274e784421c4e110883ea858be6b4c27759b9a","_number":760711,"virtual_id_number":760711,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-24 09:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-06-02 12:54:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-06-24 10:00:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":1,"date":"2021-06-24 10:00:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-04-23 20:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-24 10:00:12.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2021-06-24 06:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-06-02 12:54:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-06-02 12:54:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-01 10:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-03 14:26:37.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-11-29 12:40:42.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"},{"updated":"2021-01-21 07:49:32.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-04-23 20:44:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-06-02 12:54:10.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-06-24 06:19:05.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"3325c55d768f2bf9f4765c412e5ce1455a33c9b5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-01 08:35:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bad52c52ca2b10d4af2d202855e7bc0333ffad25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-01 10:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/629d5dc6e0df48ee9336316d6637fc2c : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc89e10878b4ccd848d78edf9483761 : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1202013165d94757825602ae3604c3c5 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e3b857cc7141c1887ae24a11ab7112 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/749a0c5d17104ec5b6878b716f1cae19 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7cd9a6166c4e439388ad2cf04d95a9 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af031dd2d70842b58baac008dec1aa3f : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faa821c2d70e4c1fa54db84e28301cec : SUCCESS in 4m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/02a63598bda9404c86a1ea9c9f3f87a2 : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/549d3d8d1fb049e7aae32825a519dc74 : SUCCESS in 6m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f65c67f8b91240b5b4862224f0324578 : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2938d23a30c64001ae7723d927691e59 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a145dbea8f244bdd8159102b0b94b1ab : SUCCESS in 1h 05m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6b55ec4ec11547e58c4e1a3599fa97f5 : SUCCESS in 1h 53m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8559b33818134fccbb2ffb92426e436d : SUCCESS in 36m 10s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0150eb342ee54ba39b602ea2b4546d3c : SUCCESS in 40m 54s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13d192d69c6b4db5bb243f2bdda8c889 : SUCCESS in 39m 19s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a2a5aac348b241f7bd70892dfbb012c5 : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e45dda83a5ba4cc0818e7b9714b80541 : SUCCESS in 2h 06m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f95613334904c55bf27ae67c048b27f : SUCCESS in 1h 38m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6cbc8476d62441179eee8349405dc42a : SUCCESS in 1h 29m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/60d5dbebefba4e02951b9afac0b0fc12 : SUCCESS in 1h 26m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b3cbb5cfc66144868def7ad3d6067306 : FAILURE in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d6314f85fa8074e614b6e73d6a04f950e20304d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-03 14:26:37.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks the same as what I did for the ovn provider driver recently.","accounts_in_message":[],"_revision_number":1},{"id":"a1cccfb3409922825c603b4df261ea73c74319ef","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-29 12:40:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b4e9d56874a787f24279635c331aa316c6dabce","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-01-21 07:49:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e155aa86db923b982fe694e58da0d183457ad022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:42:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc425e7efb9da942c1cd511f4e6903fc027ccecc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-04-23 20:44:49.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"227dc89d5df166e73a576179c74ad18caa14d138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 23:22:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9314048777e847eda8ec6594101f0bff : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a2adaec8f4b41c2a464eddb81d785ef : SUCCESS in 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b05a2a10082c4c3cbb2a2b2bac3a7035 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4b26a23cf9400bb53ae9af857fc759 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/103acecb186d46fe8ca60686f3c32c02 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3fa3d17dafb4025b8644861e39f2b99 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4512e2931704ac797d0bc26f0fb1983 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a7e6ccd76e5423ca7feca18c6ba5c1b : SUCCESS in 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dde34937751748c48ca1bd51a092376f : SUCCESS in 5m 26s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ba76778d56fb47f78c5d75031e87b7fb : SUCCESS in 7m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/17247a00f5b44916b82ebd348d66f2e9 : SUCCESS in 9m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c1c79bd544604538a7babc4874177c9a : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb2d8d8e24a0484e933a10f5527cb46a : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bfbe5ee862fc4cb4b3540c2e80a7c285 : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0b19bfeace0b4120812dcf62e3f2df95 : SUCCESS in 1h 44m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/935b8b833e7f4a1a8f5830f9be9660a0 : SUCCESS in 43m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/063df56fd1d7433e96f7520366feb0f6 : SUCCESS in 31m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e12cbd2c1bc44aac86b8aefc0e95c8c0 : SUCCESS in 1h 06m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6367dc9bf2bf4b09a06617ec377fc50a : SUCCESS in 1h 58m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7631ac3ece6349c2826cde950f7c52f2 : FAILURE in 48m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7ca2ba16009c4fc28e233f36a4f797e4 : SUCCESS in 1h 49m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/6ae61b99d0fe4911a074c3fc817b73f3 : SUCCESS in 1h 49m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/44f048a8403046b1968b89c32d3ac63a : SUCCESS in 1h 35m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2d7579fbb002490eb9188e4ca39eb9b5 : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95aa8cb11a269e4cd0908bc4dfa5cb287be36545","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-06-02 12:54:10.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"263d894f507baf7d6cee64b6133371935006bfe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 12:54:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60d0d62771e4bcebde636053a33a988a0bc435fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 15:26:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97a5bd46ea284c6bb3ab565ce7b85224 : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b46a6f9531294a0386a675811b5b15f8 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2758b54170429fa5604f58112685eb : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84a805ec57e74c7ea207758405974898 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e8c1eed8abc4381b6ff24544c8468a3 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9115e94d91ae49a895be5cf17a822552 : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222a48178af94ba99ed75a8b9537b784 : SUCCESS in 7m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ebb073346e9240a489ef9a80c4dbe040 : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef30769dca6f432ab0016e55bad72213 : SUCCESS in 8m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d792fc849da141d6b5111438d2beb242 : SUCCESS in 1h 22m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a971923581874c378a448486e3de3307 : FAILURE in 2h 29m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/018719a2cf964b7d819ac1b842805865 : SUCCESS in 36m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f9a7dd44a7e247498bef136b2bf14aae : SUCCESS in 46m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4fed69e4bb224592ba59956a0de35add : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":2},{"id":"005c1abcd939e393a527a8814cd3b298188ea716","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-03 01:02:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"62755e51530f547b3eac4583cae7d619dfc95d83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 03:38:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4d3eec1d670b40bda22ada029f629990 : SUCCESS in 9m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61daf73f81d24d61a8ee72100fafe005 : SUCCESS in 9m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fcf7461bbf74d57a6b65a4ca18e67fa : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85ca633eab54088b09fdb2d483c43e6 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aff43f2ce45a40d584fd44de13626f94 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bd48e99b7dc4be6a4e2d0b2754b5a5d : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f69f6efc23dd450297b05b0908a22081 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512bb470cbf844ea851999c16da6e89c : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/578cfdb1dd684ca69e6961cb0fbefa94 : SUCCESS in 5m 31s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/47ea6c6de97e49d292ec94cc7b9d2cb0 : SUCCESS in 5m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2aff74f453ff466895094ab0a1235da1 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d63eb456e46445a9b2175f0a3188b5de : SUCCESS in 5m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79720df7e21c4c1c8655bf4eefb25b64 : SUCCESS in 7m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2407884aa7db44f1a9a1c8686974c6e4 : SUCCESS in 1h 27m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1176b00c22eb44ba9a8e5dbfe7db06bf : TIMED_OUT in 2h 34m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fdcaad8f95994bce89e4e0fdab8f85e9 : SUCCESS in 1h 01m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/36e50ec5d17b4cacbbc7af6dfcacbb1f : SUCCESS in 33m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/46095cb00a204ce496d8f8d81629a60f : FAILURE in 2h 18m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b95d64f002ff43cb9e8b54889999acb4 : FAILURE in 14m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6bcf7de35e5a464f81f3b1897fd36dd9 : FAILURE in 2h 17m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d1967640ec94343bffca466395b6d07 : FAILURE in 2h 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3b7de66822c743b790bbe2563bc619a0 : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2c921f7a936a4b3f80ef156eaf21835a : FAILURE in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eb5f855eb6942162179481e1eee122c7617c9ea7","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-03 06:24:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a41a1078e23fad890f54ae70ad76a2e0fdfabb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 09:07:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e337d285cca403496242e7260beb189 : SUCCESS in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff2773fca6e64b37aa39472c66efc9e6 : SUCCESS in 12m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/965b4d858695448d920f038541e711e3 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ef7c53296b4e7f89fc251d726606f5 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98ddc31ab9144da8d8815a99642cf5b : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ec8f39bb1744fc935fab52b2156075 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42e0cb509fd344d69f5dcb5ba0e4b452 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977424940c2d4eb39392e99507a2d6ff : SUCCESS in 14m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa9dad2926540ee9749ca35f2f77146 : SUCCESS in 7m 58s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4040f294ae904199a7d0ab6c2ad7b00a : SUCCESS in 8m 45s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f3ebbad6c4bc4be3a7b50e568121a69b : SUCCESS in 10m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c52453c005454703ada3c041b6435c8f : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a6703a9f56824bb7971128b8622d9048 : SUCCESS in 10m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a5f501fd221344b58185d6a4e2123e22 : SUCCESS in 1h 33m 36s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1af8689b66be4332b5e7dd3a0e0447aa : TIMED_OUT in 2h 34m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aefb4aca4a2c4e349e0feddc33b011e7 : SUCCESS in 37m 41s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/27a2a40b866f48a792aa8bb62d9c55bb : SUCCESS in 34m 34s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dbc70aae588b48d4853c7bbe51051f58 : TIMED_OUT in 2h 33m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e10ae3f09d748c7b463676cbab4b836 : FAILURE in 18m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9666bbee67684c7da5aeaf9e8341526b : TIMED_OUT in 2h 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5a05c6aecbd54a7b905eaa2e98d9b3a8 : FAILURE in 2h 17m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fe78cc9558db455db0b7ac358a2185cb : FAILURE in 2h 06m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/85f5c02f5f6f42d6a113e7323dce96a8 : FAILURE in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa19c793ccba3253a327176688298614ab8286f6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-07 05:59:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"52f916000ed69c780594ad5e99f995acf36edcfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 07:55:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/41c0fa152b134378abe2282608b92793 : SUCCESS in 8m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e17c37633df74e78a6e5eb26675ba97f : SUCCESS in 10m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3672263ab0064dee9de727924878623f : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc892853a2149f0b895d261459ba968 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9437b52c2204621a4191e6c4e68b950 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92dcad6e80c849e8b6a5022e1160ff6f : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/933f6be98bd04b0aa09bb7738fae2b11 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f28ee1d7e4d4d0fa12d85e1a3f2a3bb : SUCCESS in 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/175e06b2464d455eb42be05f1fbf0c85 : SUCCESS in 6m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3223222d311a4084939a564b472f1106 : SUCCESS in 7m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b094283f2fab4cfea6890e33ed2d19d0 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a6b0624698f74a88ae3da930eeb18483 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fcf515df1a345d99489ab1800482141 : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5acfd7babe1743c18dfe8538cd8c44da : SUCCESS in 1h 43m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06b227c2142443619760889e83c2b593 : FAILURE in 1h 27m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ce7494bfcb940a79b12a8068f92dbf0 : SUCCESS in 38m 52s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7a8923e499242f78a10fec7597ae5a6 : SUCCESS in 40m 04s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/036d3715f71d40b7afd5211bda866e93 : SUCCESS in 1h 26m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b90932dd5f3a492db78a3344eded60af : FAILURE in 16m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be7e50d330774c3fab994568142b5969 : SUCCESS in 1h 31m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/abc9ef1c81754c57879c67dd1360d5b1 : SUCCESS in 1h 42m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7928596fa26e47a686294e7a03a9ca6e : SUCCESS in 1h 21m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6f0427eb9df943d694f37f2a86f2e74c : FAILURE in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7962b43a270deb86d81403c3217c09eb02844d2","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-08 10:45:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e63b81f8c404200c9ce8679911e9c8af7cbdbeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 12:57:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d8a960d2c1b44366a51eaf13c792779a : FAILURE in 5m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1625aa11e6e141a68743d9beb821a714 : SUCCESS in 11m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eff169ae25e4fcab76210f6b8d0cb8c : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1160ebec41ae4367a0f4b7df7056fd77 : FAILURE in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e28766804dd4a8e876a43a4b91d94c7 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2c774e310e54a01b4a9aafe3411ee73 : FAILURE in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b57415c171468990af459aaec5cf00 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1b328515bc4977825b4516fcaf4d7d : FAILURE in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdead047549a40e4aea8dfe97bdc73bf : SUCCESS in 7m 06s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07d40b6b2b5c495c8349aa614be75fda : FAILURE in 3m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2b54a166b72444cb8db7f896d1eea2b6 : FAILURE in 3m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f98622e4aea54847b616dcf02858a237 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c0d98b351564401b4051cdee60b3606 : FAILURE in 3m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53d7f7ca1bf941e480322a75475efba8 : SUCCESS in 1h 34m 16s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/47926f1b0d534ee791f8a74a66dcc320 : SUCCESS in 1h 46m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dcbf29011cd64e20a3d5fb87041840e9 : SUCCESS in 42m 18s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb55ad56cdac4e04ad32a8404292b019 : SUCCESS in 43m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1abb87fc33a145a78de6e2571069ab5a : SUCCESS in 1h 48m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1269834088e14e28a1ed795b6a9b69a2 : FAILURE in 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/04be1b24f3b641be845ad596def58b4d : SUCCESS in 1h 52m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/977664e308a94f908cef8b49e4ed3ef0 : SUCCESS in 2h 08m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3fdaf3c1400f47989a92ea947aef5268 : SUCCESS in 1h 48m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/149db485045f452c894351fedf7a246e : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61b16057673bd889e9f9518d9a4a507d534d38d8","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-23 11:55:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d157fb1baf881b9ebcf6b398f3c36b14cdac271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:02:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ed3385dbee345578e2f5bdbbb5dae3a : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cde9adbfd6644cab7b5cd52c22b8ae8 : SUCCESS in 12m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4baf0ea0a9604f7c99ddfffd8bc0ec8f : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b37244a22054a77a23b12fe57caa7fc : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26fec9011f2f45a790a4e8689fde4052 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1e25253b6ff4679a40173fc661eb7b0 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9c787371d7a4496ae740219d7c16c1f : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae939c52c814565a0d8238bebef5162 : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a0cba545a684a168b02e35fa393918c : SUCCESS in 5m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7678c472556744f5b56e6e74615fd810 : SUCCESS in 7m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/492e607bb2c24ea69fcd443b39554649 : SUCCESS in 11m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b266f93b8adc48ec86483920511b7c9c : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/524a729907fe41ff843e0e31d55f012d : SUCCESS in 11m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca10d62f6fa345618f2709ac1dc6e8df : SUCCESS in 2h 05m 24s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/391cfa211ea041739cf0f18fd2979cf4 : SUCCESS in 1h 52m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d3b629b4cfe3472fb1cf363f73cbb15f : SUCCESS in 46m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8207663d78e47739c3c69cf3daf962c : SUCCESS in 46m 14s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7027a0f4db55420d85db6f35dc55f51e : SUCCESS in 1h 46m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f435ed4b90444d41b0024dfdd43b3a6f : FAILURE in 19m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc15cf904f484580b7e690ff6c454e2a : SUCCESS in 1h 53m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/30cd2e5bd4a44f14a6cb760d33ac3634 : SUCCESS in 2h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/619e5b4b775c423ba14540838a668f6f : SUCCESS in 1h 45m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ac496031fcdb4ee48b2193cc668243ec : FAILURE in 10m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c8406b5d98ab0a807b630b237bc581eff5ca82d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 14:03:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f71dfc21541810a262e55dc551b66366499b5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 15:37:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/257a5413ee1d462a82e4840b85b4de87 : SUCCESS in 6m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d23c6f97b8e84489953340df8f51a3f2 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4f2281b8d148699e44f34effbec23d : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f3a29a904f4447a4db8539af02cd0c : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12673bdbcfb64bc9aa71412319876254 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d99691083d432ea942774aa96d2c50 : SUCCESS in 15m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2c30f6dc0cc4c169f12680568087250 : SUCCESS in 5m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5618c13260aa4bc686f8e47730c4de45 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7cf37c95fef48108763d8f54c0b0934 : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b702ae5b8c18410bb14f1cde75818db5 : CANCELED in 1h 32m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6ddfb3befd645d9a274cf4ace92db74 : FAILURE in 1h 33m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38fd13eeebb640d196b0f234ccae493d : SUCCESS in 47m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c29c28370134d96a3638bb19ade0c07 : SUCCESS in 46m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f0e9cebfd074154b5748edad1226902 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6963bee07362e9cb4d89e8b44983569823fb43ce","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-06-24 06:19:05.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"532438c4d4fc4f0fbb9966f8fd0218182668da42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 08:11:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1db9a1439be64032b4f458e95c9e9805 : SUCCESS in 30m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76201e95f19a4cbca40d8c261c78220c : SUCCESS in 22m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1868f6af84814f30bfc1a0da61fdaa5b : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abf751fbbdb4fc383648eea88553833 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5312f2d9430242e9aac9c42cc905ddb3 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/656efe24bf4b4fcda1c7bbfd8a602709 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f391ce1dcc704608b12107b8cb465848 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4293ad892fe4f03a7b0e2bb152cf079 : SUCCESS in 14m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eee4d3f5f9141528101d45407fb6f73 : SUCCESS in 7m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/45a8bbecc925483c8b0872637667ac4a : SUCCESS in 8m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6e286a7a7a1243eda66808b174a76529 : SUCCESS in 12m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b0a74c9ce55b4a159dd5e9c7278aa4ac : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/796104b653db4d9fb25aa4e871a000cb : SUCCESS in 10m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/73c0b60006d4451990a07979ddaee0cf : SUCCESS in 1h 39m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5a4efdfca5f9424786bcd171856eb49a : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0e0c873dacf4a25bad62c2cda9695c2 : SUCCESS in 35m 16s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a5f479fc60f4e9fad2f3c1e9d9e54bc : SUCCESS in 39m 37s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d493f453bddd4bc5b00475242db5b1b9 : SUCCESS in 1h 24m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/12b7cee6bc9d4325b441834ebe716a8b : FAILURE in 17m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8d2e2413703f4254a5579daef467a270 : SUCCESS in 1h 42m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f83c50a0912b4e1f84c0bd2a4c11fc92 : SUCCESS in 1h 49m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ed901e97f5e74f74a5708e1d48f50559 : SUCCESS in 1h 20m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/583a63f61fb248ac82c80a077d8b8be9 : FAILURE in 17m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"063c37e4e419129a4b02b75aba2aa7478bb4dab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 08:11:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4e464436509d076c55938412b527822ce756db92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:58:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0bb93bbe6550485580a0ba8d11c4a704 : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/840a9bfbf850455eaa8d33d112de7498 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd05e13af014183bd097f1993bc02ec : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c790791c79cb41d0ac5ebe833a93a183 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d12b40d04a49b3b7bf1570e0b039a4 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cfba8560da4deca93db12bc332832d : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b820e6d28f464abf991d2127804c8ace : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da12de2b683f4012b6fac7db8912bdd0 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97f267f4804844b4a841f9a201485059 : SUCCESS in 10m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/38662155ce57412ea6adda70d9ead6df : SUCCESS in 1h 44m 06s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b4a090c63fc04ef6832fde41ea48f3e2 : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8f4c40f192274871895f10fd963ac323 : SUCCESS in 38m 49s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c32e9cfe3e140beb38b1cd19e015a7b : SUCCESS in 36m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e2ff0a717464ec08d59e7c52db5f974 : SUCCESS in 46m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7a2e2c9a6271c6720f1265ad948fd54bee7345e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:58:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1c52d7b343a1e727ebc52fbbe4e58bcf64e9575b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 10:00:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/500409ec98be4a41a644bee81b9c6933 : SUCCESS in 1m 15s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a0791bfce4248c490725f54abe039f9 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a180f89977724f07ab6b83926d539cb6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56f5987867154ac698d69cc4496f124f599af90b","revisions":{"ce3f513e5371a5e30429fa565659ec360657ad33":{"kind":"REWORK","_number":1,"created":"2020-11-01 08:35:08.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/11/760711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/760711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/760711/1"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-01 08:34:10.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-01 08:34:10.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: I82d482bba3e4c40913583fb26f692824c56e6ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3f513e5371a5e30429fa565659ec360657ad33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce3f513e5371a5e30429fa565659ec360657ad33"}]},"branch":"refs/heads/master"},"56f5987867154ac698d69cc4496f124f599af90b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-23 20:42:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/11/760711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/11/760711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/11/760711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/11/760711/2"}}},"commit":{"parents":[{"commit":"beccbb17e2be3272d6acf86f06b5da4da3afd694","subject":"Gate-Fix: Replace Octavia hacking O349 with H216","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beccbb17e2be3272d6acf86f06b5da4da3afd694"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-01 08:34:10.000000000","tz":480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-04-23 20:42:43.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: I82d482bba3e4c40913583fb26f692824c56e6ac1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f5987867154ac698d69cc4496f124f599af90b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56f5987867154ac698d69cc4496f124f599af90b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
