)]}'
{"id":"openstack%2Foctavia~760737","triplet_id":"openstack%2Foctavia~master~Id354d9e30b6fa6f4b72116a659d16a2fb8030c87","project":"openstack/octavia","branch":"master","topic":"assertItemsEqual","hashtags":[],"change_id":"Id354d9e30b6fa6f4b72116a659d16a2fb8030c87","subject":"Replace assertItemsEqual with assertCountEqual","status":"MERGED","created":"2020-11-02 01:00:07.000000000","updated":"2020-11-17 03:28:20.000000000","submitted":"2020-11-17 03:27:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760737-1605583624606-195fdec3","meta_rev_id":"556a09b50f30bf58a9fd70e070eed8b661f7f676","_number":760737,"virtual_id_number":760737,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 03:27:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"date":"2020-11-17 01:00:42.000000000","_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-11-03 08:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2020-11-02 06:48:48.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-02 07:57:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-11-03 08:28:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 06:48:48.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-11-02 07:57:10.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-11-03 08:28:26.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-17 03:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9387c0e2da65b06ad836e9100e4f1a32162d4c0","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-02 01:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"341249e1e6b4fdf8ea82232eb507af0fa45d8a9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 01:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed84d6bd6e0943fa8d6c470dbe7e23ed : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59325bdc36004693874ddf08b60fbf53 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0335de6d954e5f965a213292a0186b : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf92b98bc1a421185964eac3934bf04 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5365bc81c194ec6ab515d3d9905ce03 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9776af46b19b417eac3d87f7088ab92c : SUCCESS in 11m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dffb52aff7f24fb8b4d5bc0835f3a59a : SUCCESS in 4m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/de0af5c3171e4c878c89c2941b36624a : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8499a3e515743a5b597322cd3426f36 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"dc27f609b3b91ac854931a4def9735e445516e72","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-02 06:48:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"562c34c6912e96122b80bcee94f2ab78b2bc4903","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-02 07:57:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28ac7c8b1794a57816abf1d6ff1d552d5e7c48f4","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-03 08:28:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"754ebc4523bc602ab7befc077e53a584aa6a64b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 08:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b589c51c6feb2c54f1a405b7494482f37a29a6fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:28:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17b81512fbf0423e92eb3bfbedaa86fe : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768c7337bb2b4215b77abe3c1166edab : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d47be90746f4f2aaade4915ef515689 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c818ee1b3df4b6296a71f35fbaaeee1 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c8bdea053f447283557f69171c777b : SUCCESS in 10m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88f43fececf14de6a4ac13880f73b51d : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2808350573534147b335560ce9b88fec : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fd49b3ffb84948de9044af5455c95625 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71e828c9de3340f09592744c6be46c75 : SUCCESS in 39m 51s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/e0a8b96211b0471b9184b4aaee1f49fc : SUCCESS in 39m 39s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3798e3ca7be6459389b543e2618c8502 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a681404c574e4dabb5e9e1806cc60ddc : FAILURE in 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"19187b6af2e8e3f05c52b2aa8800ba30d2e66995","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-03 10:02:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbe05f8fe59a841df088c6734b99aa16bec66aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:35:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edca7fe9e9d046899872088a13f33021 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c404c9afbab4322bc026cfc265fdcde : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35214d2cc4e4966a43a879bfb56ab7b : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73e7bce590214472a2eaf2fbf9531a2c : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/664b9ad038254b588a178b3d6afcb982 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b81c000159e341649cc65eebe5d95086 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7676160010f4988811799b87c006b94 : SUCCESS in 11m 27s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cfd7f138163483da6663860b424a4e5 : SUCCESS in 5m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0037d76ae14943aeb79286baa6023485 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13a4b873ae3547a183cbe6d3b82b8f4f : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5306c44c7a9fd7816810de74e3ccc00161811b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 12:35:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76d313c841208e15263271f6243d26f5641f036f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 14:31:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e142a676b7345158ce0992360928ad8 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/693ad7ac8ed14b4fa8f567289bcc4b26 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8af98aee45445039762ff7200b83e9a : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf63a5a94c1e469593ba493d29a332ff : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42497e2f1e584e3b9a66a019e2258b21 : SUCCESS in 10m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eefd50fba00141eca5f77e5affebe013 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0fdf4ffd6b17480ca7f90c85e59b0958 : CANCELED in 1h 06m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0e34e77bac44a0589c96d336b3d49de : CANCELED in 1h 06m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/954808e4c2b1458aa045b2b08d8bd0c3 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4fa8777972874f3fb6aa61b3bfe447ea : SUCCESS in 42m 35s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d4f725717a6f47ccab83b2ee683cdc77 : SUCCESS in 44m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6bf44de52593478f8d5e05348218f9b4 : FAILURE in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4df3b48ca2bf82674f31c399d9ff1b95b958ee87","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-06 10:41:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a5f218ba73dde1e53c7d0fe2fc440bc57018bfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7916ab98c060463a8fa8fce2e452fdcf : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3335ec7b48aa4b7daed880a0319a51a8 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4170c644c4364da6b178eb3d60c6dad2 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/228dbbabfd59481bac1c46349ea166d2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0de945a456546e9a37c5f3184bafa37 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1db6a9095e840cb9d36660810af445d : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23cb5ae7cdb94c629561989deea2f8d7 : SUCCESS in 17m 15s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c7fd1bf121a64dd9a3e2ec14128a30da : SUCCESS in 10m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ed2b12333f9d4f5d96d496b907f59488 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f784856705d448e58a488fc0dc536c52 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d8d17ecdbc4c8a66934fe6a08852221b52efa069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:06:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3833b6fa30683ec6a086c6fd7575efff136891de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 11:41:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8909f0eac87b49de8811a4e7d3519997 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43df6662d28d439d82730463fa1f768c : CANCELED\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47cbdb9476b84e4cbc78005e96dd1ee3 : CANCELED\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0c821f4d632400a87c4b89f7d05f894 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69797d9fc54b4f8d8c98af8e693dbada : CANCELED\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4779f6bbbead43e9b84a171a08fc1541 : ERROR Failed to update project None in 1s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0af95904f88484da6eedcb6e5ca841f : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2d236530c462475f90bdb4628a26227d : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4be09bbb350143e5a651ec594c1b881c : CANCELED\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/86a44a474cec43f6a62a4b27fd3e1272 : CANCELED\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ca8f82e50b645ef8d0ba1903a799edb : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5f88d5ad7a44cb4a59a3760fc2b3a80 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"e1c155834723119315d4cec4bad46ffa2bf284f9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-09 00:48:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9708f5b943c46a988203b32bce9b7e599457d4ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 01:03:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45797973419b4f6093a75267b576b003 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa899146a98845069060cdef56c29a5b : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e8c4e414b44de1a5e4f7dbba67caec : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/412067c4e8354591a00d4926e5940941 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cabe9f2d284438c8bec166158d4569c : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c5021b3936c42ababf934dd235e3be6 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0e153ecee14abb8570449e45c12ced : SUCCESS in 13m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a1c174b9c1494054b2c1b949d5f3a7f9 : SUCCESS in 4m 25s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/03387f8b870a420db1e9426d32ba6ffd : SUCCESS in 7m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8753864b460b447db0a061d04d7bbfde : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9133709b5eea452df50d7f4fd42bbf82e92d3a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 01:03:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06deea33a5f6f33b78c0eddbef61541c02caa926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 01:56:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b7abb7bfe24bb4b92b5bd82f843d3b : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aa81cdc2eed4987b0a908f0265f22a7 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cddfaaf514c54837a5769a1760450afe : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/366ade7ab5204cdfb49ccce505f76903 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98fc59999c7f4c2f8a2e82e0ffae5e54 : SUCCESS in 18m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/811d53c82d7144438920fcbaa7202773 : SUCCESS in 6m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de77ba037cfa48b8a044001eedd9cad1 : CANCELED\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95fe6fb05b0c4e2e8da66dbdb54f8f26 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5479aadb848c4a9d86613d9ee9175713 : SUCCESS in 39m 34s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/0233ce8da67f44eab0c024cf2043d61b : SUCCESS in 47m 05s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ab2abdbf5034fcb96ef8ee1d8d95d7e : SUCCESS in 40m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/54a9e185f90e4446856b05489ce236b1 : FAILURE in 50m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0a791913340d29ca3726a4996980a305d82a3c5e","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-16 05:34:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ffb2aac512749d6aaddd70c7135cdf66b3dceca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 05:52:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6d38aa931841eda39e43d2c2dd369c : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/263cfe43786e4f87a5838727dae87706 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ec256efb7a417e97d18126a49a4029 : FAILURE in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0015033b4a3f4888af64a013f362199b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfc28be178924fc29e49e999750053ab : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c74894db2804e8f87ba43e2f964d3d8 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142923b243be465ba5c85a3507cd0951 : SUCCESS in 17m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d1c6f19a794d4f09bd5691d1cbc108fd : SUCCESS in 5m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/762090d4a04d464ab13a1c91d468879a : SUCCESS in 7m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/68724f06d1364111a3b286b9d6e7f539 : SUCCESS in 8m 27s\n\nWarning:\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py","accounts_in_message":[],"_revision_number":1},{"id":"cfd5e2ef4943bb3d11bcb999fcc961d2e3c40218","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-17 01:00:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1d26e07c5f040899fe7be00bce1cc8100ac83f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 01:26:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f80f3c31e684334959a83a07dc5f757 : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1925f336888c49e7825baae038308992 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ff08e3a9904693b9b91c78f6212410 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5df9c2460df4790af3f6a031d2a0f9b : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0126dd8a69b443e81d3a17a88467de9 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ac8531e24464231a38a393a8f3dc7ca : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e128f0d3e044df90a6066a16907afb : SUCCESS in 17m 00s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6c837700c0d34bb3b33ec6a8c17433c6 : SUCCESS in 5m 15s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f09168f4e3fd414e9b91979d437d65a6 : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9808f3a78ff4d47aef04682f8e321d4 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"827bcf0a2293ecac56eb381ffe98c71cd676e3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 01:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f6840e573cbf5c136793f5b52ad09c759dfce2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 03:27:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ea04bbcbbd74dd6b2d9422a5bcc6151 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6909ab048041ef987918d861306ea7 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4c6fd8433c64d20bc9edc60f3f4f03c : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/972be626235346c1823eee0ccee055b1 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b96639f76064f7ea9bd477cd5adda7d : SUCCESS in 17m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46a0d3d48fab42e49c30f318bd56a5f9 : SUCCESS in 7m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e128abc9e394d95908e8a65322f4b9f : SUCCESS in 1h 25m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8c4a1563b6945bba62c858b644f70f6 : SUCCESS in 1h 58m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/76b16ccc3b054f21b3593c43eec63ab6 : SUCCESS in 50m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/845e3a693e05488280b2c2a01585e750 : SUCCESS in 39m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/296f7ef75c49403ea2c5d1cb8462b210 : SUCCESS in 49m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d926f151ee7145eb8aa04358d086a543 : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d54e79ac4b5e04461d0397f441fa1167cb82c00b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 03:27:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a724e7ce10a909645c211e92474c8d911801ab86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 03:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d1e1e2e3e84ec98a67e3335028c067 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd4abac1d6482deab27e4ecd24e9a9d5f5a78cc3","revisions":{"fd4abac1d6482deab27e4ecd24e9a9d5f5a78cc3":{"kind":"REWORK","_number":1,"created":"2020-11-02 01:00:07.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/37/760737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/37/760737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/760737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/760737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/37/760737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/37/760737/1"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-02 00:59:39.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-02 00:59:39.000000000","tz":480},"subject":"Replace assertItemsEqual with assertCountEqual","message":"Replace assertItemsEqual with assertCountEqual\n\nassertItemsEqual was removed from Python\u0027s unittest.TestCase in\nPython 3.3 [1][2]. We have been able to use them since then, because\ntesttools required unittest2, which still included it. With testtools\nremoving Python 2.7 support [3][4], we will lose support for\nassertItemsEqual, so we should switch to use assertCountEqual.\n\n[1] - https://bugs.python.org/issue17866\n[2] - https://hg.python.org/cpython/rev/d9921cb6e3cd\n[3] - testing-cabal/testtools#286\n[4] - testing-cabal/testtools#277\n\nChange-Id: Id354d9e30b6fa6f4b72116a659d16a2fb8030c87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd4abac1d6482deab27e4ecd24e9a9d5f5a78cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd4abac1d6482deab27e4ecd24e9a9d5f5a78cc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
