)]}'
{"id":"openstack%2Foctavia~761220","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I43b79c4c203ec5c7ecf617c449805fa963dd2db8","project":"openstack/octavia","branch":"stable/victoria","hashtags":[],"change_id":"I43b79c4c203ec5c7ecf617c449805fa963dd2db8","subject":"Add octavia-tempest-plugin to devstack plugin list","status":"ABANDONED","created":"2020-11-03 19:26:23.000000000","updated":"2020-11-18 19:37:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b056296ad518b040905ea6a529f4f9bdb1a06f67","_number":761220,"virtual_id_number":761220,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-18 19:25:44.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-03 21:51:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 21:51:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 19:25:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"a19b60c3d3836fdaee2d08d2e2ff21163867062d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-03 19:26:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee8d92936f0cad8c6d2525fa85664fcbfcc84283","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-03 19:26:32.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 911009955b9265308e677203ca8868f9dab6b352","accounts_in_message":[],"_revision_number":1},{"id":"6dcf5ccb8829e474a59fc5fb76d006946db5b858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 21:51:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27529ea10045462ba59664bcaeaedf90 : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4993fb65719422a98103db412d7825b : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e73e0b7ad9491ba81d469b9da30413 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bb2bd44f7ce4a1a90695d7c512b4393 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93537d924eb743088cd232416e14ef0f : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a815f808714dddbba6c059ec751150 : SUCCESS in 11m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a5d165d2f8534585b22847e19c24b386 : SUCCESS in 4m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/78473ee2753c4243894aecd92d7e98b3 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/625be4087493464eb089228d6eacaac3 : SUCCESS in 4m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d839f049dd642b0bfceade976699ea1 : SUCCESS in 6m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b4ee1b6dac4245479da0b2baefef13d7 : SUCCESS in 1h 19m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/974b558440244df49550ff8e38043ef8 : FAILURE in 2h 04m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e258d905f0224f518ecb642a927c2515 : SUCCESS in 46m 04s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/5b57a88245504b8fbfe58b203357716a : SUCCESS in 40m 02s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a075918fc2ff4f978a2645631983e1f8 : SUCCESS in 40m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4164e31775324d588e02cbfdf519b25b : FAILURE in 52m 46s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7164e76863424fd4a74b886f10e7f889 : SUCCESS in 2h 11m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/31b53be70047453da7ef312103688ee1 : FAILURE in 1h 51m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6001de2ef354997bcec183f6b433ba1 : SUCCESS in 1h 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/63dc69b5e5f24bdab89fcd3c56bec7da : SUCCESS in 2h 17m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7277c88730d24c1fb0578a5f0c5069dc : SUCCESS in 1h 55m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/589670a3ac104abc8cdf2f79a8324787 : SUCCESS in 2h 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"997bb240e2cef525ec2409f846beb38964dda0aa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 16:09:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"385d3e72cce44efe91a6fe0c3b8c1d33ef537e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 18:30:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e08ddafb86b445b6b85ac5333b123fae : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9be9a52618b84f679764bd000e65e793 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8feb41c19c743d3a003f86e4e903e92 : FAILURE in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efb676efaabe459e94f99f9c378f2ef0 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af2dd21b57e74de2a3626814bf8c74b4 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342ec2e0a23e48069a53f8b8e0fef8f3 : SUCCESS in 12m 13s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b71bdf03a9964f81bbcd7a99ecf14e37 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dc3029b840ea47b3a34a30728122200f : SUCCESS in 7m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e1d1a7c1e4dc4857a28a2a2069e2414c : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/989349b4ffbf47f388a4eec84d572d0d : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a36feea41c8845fe9c02d772c5909fd3 : SUCCESS in 1h 48m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f39a7430beae478fbe8268957fda086b : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d118c74dff0d41d184351964e674dae7 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9812f33054794d859c83fc3c9330f5df : SUCCESS in 45m 30s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/389ed8f9192c4d1cb7b6382c45bc46a9 : SUCCESS in 1h 28m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35af065acda64d9bbb24b24ef7695163 : SUCCESS in 52m 07s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/97de82e244e246628a74c88dd64305a5 : SUCCESS in 1h 52m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/849c57f6d93140c286e062af95254ec0 : FAILURE in 2h 15m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b07e9b3d4b34421f8b98f651c496551e : SUCCESS in 1h 41m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/73ce17baff32490ea42913e27dfdebc7 : SUCCESS in 1h 51m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6c6d6bc8bc92461aaa817eaf50056ef9 : SUCCESS in 2h 07m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d678c7cbab6849bcbd48660ad47ca9ae : SUCCESS in 1h 51m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/cmd/house_keeping.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/server.py\n  Comments left for invalid file octavia/amphorae/drivers/haproxy/rest_api_driver.py\n  Comments left for invalid file octavia/controller/worker/v1/controller_worker.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/amphora_driver_tasks.py\n  Comments left for invalid file octavia/controller/worker/v1/tasks/database_tasks.py\n  Comments left for invalid file octavia/controller/worker/v2/controller_worker.py\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_provider.py\n  Comments left for invalid file octavia/tests/functional/api/v2/test_availability_zones.py","accounts_in_message":[],"_revision_number":1},{"id":"26a50b1bb94bc103e557905ae713199ac708d3ca","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 19:25:44.000000000","message":"Patch Set 1:\n\nThis was merged into https://review.opendev.org/#/c/762906/ so can be abandoned.","accounts_in_message":[],"_revision_number":1},{"id":"b056296ad518b040905ea6a529f4f9bdb1a06f67","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-18 19:37:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e5378e802c5d25074ec63f1b6113bcc6e4b5bec","revisions":{"2e5378e802c5d25074ec63f1b6113bcc6e4b5bec":{"kind":"REWORK","_number":1,"created":"2020-11-03 19:26:23.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/20/761220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/761220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/761220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/761220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/761220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/761220/1"}}},"commit":{"parents":[{"commit":"b28743c22ac166e1920697837dad23acfad4effe","subject":"Merge \"Remove unnecessary joinedload\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b28743c22ac166e1920697837dad23acfad4effe"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-11-03 10:06:25.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-11-03 19:26:23.000000000","tz":0},"subject":"Add octavia-tempest-plugin to devstack plugin list","message":"Add octavia-tempest-plugin to devstack plugin list\n\nPatch [1] deleted the test server binary file from the\noctavia-tempest-plugin Git repository. The test server will now be\ncompiled at devstack test-config phase via a hook in the\noctavia-tempest-plugin devstack plugin.\n\nThe dsvm-scenario jobs, defined in the octavia-tempest-plugin Git\nrepository, were updated to include the octavia-tempest-plugin devstack\nplugin in the devstack_plugins list. The same needs to be done in the\noctavia-grenade CI job defined in the octavia Git repository.\n\n[1] https://review.opendev.org/#/c/758778/\n\nChange-Id: I43b79c4c203ec5c7ecf617c449805fa963dd2db8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e5378e802c5d25074ec63f1b6113bcc6e4b5bec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e5378e802c5d25074ec63f1b6113bcc6e4b5bec"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
