)]}'
{"id":"openstack%2Foctavia~761283","triplet_id":"openstack%2Foctavia~master~Ib8549d71b7adb95d0dcb36903c1744190056adae","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"Ib8549d71b7adb95d0dcb36903c1744190056adae","subject":"Add default value for enabled column in l7rule table","status":"MERGED","created":"2020-11-04 07:06:27.000000000","updated":"2021-02-01 21:07:25.000000000","submitted":"2021-02-01 21:05:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761283","meta_rev_id":"41fec43e31c61b0872268b7f35e6ff3f48e89303","_number":761283,"virtual_id_number":761283,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-01 21:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":2,"date":"2021-01-29 23:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":1,"date":"2021-01-28 21:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-30 21:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-30 21:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-04 09:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-05 12:58:50.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-12-01 14:29:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2020-12-25 14:33:59.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-12-26 04:25:28.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-01-04 20:31:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-28 10:35:27.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2021-01-30 21:29:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"df653ce6baec315af87cb57e768ad091c8243a85","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-04 07:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55a6711aa06759279fb74d01a4aac9e086397b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 09:43:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e434a4f7cc46dea91cece826e90981 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a6e939defe64b69851ec5b99cd27f6b : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18528695d52449f3960739cdc726f154 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2895e72afde44ffb113059161555a80 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bf75928105249a1b8557b24b867224b : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73fe0edc6b274388862c610fccd468bb : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e1f2114f8245a68b96f2694dd5463e : SUCCESS in 18m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/007e68ee14af4070b0c45d4e452d6159 : SUCCESS in 5m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3ae57b6d6bb24a8fbe227a321972bd8c : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd99fa8693374041b9a1634d09046bbb : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1ba65fd12b464259bc5d2fe18b948278 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c04a16425944677bddf28be5d9886d7 : SUCCESS in 1h 41m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0129122974d54fb8940e942a37b5673a : FAILURE in 1h 56m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e252bab4219043f288d7d4cfe512a803 : SUCCESS in 42m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1c1de2b9937944d084bf42eee56531a3 : SUCCESS in 42m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d532468cd2fa4e6081f5ae0a904eca2f : SUCCESS in 44m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0bcb687ce57a4a8ead9dfc963facfece : FAILURE in 57m 38s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/02d37accf1d34495810e79fea9fdd116 : SUCCESS in 2h 22m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7535b8465f364ea9ac52e5a84acd8565 : FAILURE in 2h 17m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92271c2634784bb680ae224faa4bb9fd : SUCCESS in 1h 29m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4db2fa6991b04368abd7c2c244711131 : SUCCESS in 1h 43m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/730f4b9fac114de5aec848131a2ab450 : SUCCESS in 1h 46m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c4f74581248a41f38785f5b9c9b22884 : FAILURE in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a4756181c1d37a13b46ae9f46111467fad07b86","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-05 10:08:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a16b740f2f6710697eb57ec8a53b31829280417b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 12:36:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dd4997850c74f0e920fd466309ce54f : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec04b7aa0aa64ec0bdfe0c81ae7233ef : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0003746202b045cba65b48a48f6f5d24 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92eccf385e224b84ba895041cff928ad : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b8da943464f41dcb8f21e61705286c5 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c969ce2c071249c1b952c9751c4e13dd : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d1e6456b564569b577955ce3be4f0a : SUCCESS in 18m 42s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a1f753372bf4ddca89a42b2415ae86e : SUCCESS in 5m 36s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5a1a7f175d9f4e95a2c0ddca4cddd0c6 : SUCCESS in 7m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee73b98eff824feb98d08665ecf049d3 : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/648da626a018415e80c12ac0c45ae86d : SUCCESS in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6072c6bd6a4a4e36bc5e46496c14d6c7 : SUCCESS in 1h 23m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c5b6ad51cc0468f97b4e500d789f180 : SUCCESS in 1h 37m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5aa796817604905b421cc0a140c3d1e : SUCCESS in 44m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/88c0284b935c4291a774ac42d070cdaf : SUCCESS in 43m 33s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4d49d4f91d834acab6d1d2de08a38a9f : SUCCESS in 39m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/368cf56c6b084a389e69f253bb3221e7 : FAILURE in 1h 28m 43s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8c4447faea034b019d77547431b620c8 : SUCCESS in 1h 45m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b03718870434c3e92e0138c632ea3d0 : FAILURE in 1h 59m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/911db07f27ac4622b6a844aa8aa43c95 : SUCCESS in 2h 18m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c46d089f16ae43ab896927f407eab831 : SUCCESS in 2h 05m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5187b7a7ec614de7b8d6aa7b9381fddd : SUCCESS in 2h 14m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/621f7d3dd9c14f2a96555ff03dbb0433 : FAILURE in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffb5a7b1bef8b3513fb268fb3c3a10d7e1741ae2","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-05 12:58:50.000000000","message":"Patch Set 1: Code-Review-1\n\nThis will fix issue only if migration d85ca7258d21 haven\u0027t been applied, so it brings inconsistency in database scheme. You need to add new migration with update this value as well.","accounts_in_message":[],"_revision_number":1},{"id":"7f7c4844f0182b6b78f81e5c67fa390ae414d2f0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-09 09:53:23.000000000","message":"Patch Set 1:\n\n\u003e This will fix issue only if migration d85ca7258d21 haven\u0027t been\n \u003e applied, so it brings inconsistency in database scheme. You need to\n \u003e add new migration with update this value as well.\n\nyes, but after d85eca7258d21 migration, the api has added the enabled value, so this only effects the rules that migrate from the ealier octavia branch.","accounts_in_message":[],"_revision_number":1},{"id":"b127b82a5cc79f994cc229bb6c5dbd3c2c4c1782","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-12-01 14:29:57.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e This will fix issue only if migration d85ca7258d21 haven\u0027t been\n\u003e  \u003e applied, so it brings inconsistency in database scheme. You need to\n\u003e  \u003e add new migration with update this value as well.\n\u003e \n\u003e yes, but after d85eca7258d21 migration, the api has added the enabled value, so this only effects the rules that migrate from the ealier octavia branch.\n\nI would agree with Ann that you need a new migration to fix this, otherwise this wouldn\u0027t take effect if this one has already happened.","accounts_in_message":[],"_revision_number":1},{"id":"0403510f60bc8bfd4db4abfaeaa7b0f5a57640b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-03 07:27:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2773daf616fd452c7300e5573fcfd356bb11b6b0","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-03 07:33:40.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e This will fix issue only if migration d85ca7258d21 haven\u0027t been applied, so it brings inconsistency in database scheme. You need to add new migration with update this value as well.\n\ni add a new migration to fix the inconsistency for enable scheme, please review again. thanks","accounts_in_message":[],"_revision_number":2},{"id":"344c4e3bc2323b0c11f325278d563edbef6df48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 11:34:58.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/619ffc8d925f4c3f980af04dbf4c4b65 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fa5c7f448674af98944ea5d50d33437 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a772cd609e64f6298c8ded5d591146d : FAILURE in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32b1e0d4b2d4e5c8cefbb6533fe3590 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a9d3b09258e4b8b9fe969ce97a4ef84 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3b489dc0fc24ef98075e5687208b7ff : SUCCESS in 13m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4c1163220d428e9271feffec3fc6df : SUCCESS in 12m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f361bfc6784f4e6091ef9d6b08c7e763 : SUCCESS in 4m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/581f499a4b704e0aadd947fcfd96ae41 : SUCCESS in 13m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6d66c21888cf47e19800eb16ec59a5ca : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/059dd954ea1741ce8d26419170777b64 : SUCCESS in 15m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2ac2d41140e4fd3b61ad3a00e0f4599 : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89a7ea2bab2a4c11a1d489b5cbbef0a7 : FAILURE in 17m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80f19f144afc4288b39e83a5d6dab1c3 : FAILURE in 17m 55s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d40a5857c0fc43dd85a2807eb4bc7245 : FAILURE in 23m 19s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e97e381ae8274157901c9603c5c8aa23 : SUCCESS in 45m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/25cdddb794694efea2fe8917703c41a2 : FAILURE in 16m 08s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96475d4c436749c4ac9890f28cfdb52b : FAILURE in 18m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66932f4da993469695fecb41e0b9d342 : FAILURE in 12m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6ad4b0b7af7461db89ed12b08b706d6 : SUCCESS in 1h 39m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8cb0579ec14f47459f75ff7e3be91d32 : SUCCESS in 2h 03m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/200ffa55beb241ba9e5751b48e26902a : FAILURE in 2h 26m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/9974c161a9524d0da273b59ef95cb647 : FAILURE in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c51971969831e1596197d1c66b652d40d77fd237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-04 07:26:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7874435abf2c133654bacaa78533a160e83e4e34","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 07:55:01.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nThanks! Just one nit and we are ready to go.","accounts_in_message":[],"_revision_number":3},{"id":"bf30c63e09f374d58cecb2f2618d344f41c5a13d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-04 09:28:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"cbdd9ff56244fb9a709779f68f56e7223136a06b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-04 09:31:26.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nafter read the docs, i still don\u0027t understand what\u0027s the behavior of existing_nullable param. could you tell me why we should use this param? thanks very much.","accounts_in_message":[],"_revision_number":4},{"id":"a0bc47a1aeb097c438092ee53aa17362b0961d90","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-12-04 10:14:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"601fbecb045868d561134b1db0a3eef675a87c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 12:28:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77e60d6cea6a40498197d8e6718c0fd4 : SUCCESS in 9m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/297fc2f3e81144b5a82cef2e0039d772 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8770a7f072a40f1ac86e9098404b872 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3830b35c554bb1a9cc5b760c179a85 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59d52b6911274abca90560cfacd5de4c : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2d566cc17344b6a37be85729a65517 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c7ac898d58476ba792bb48de4e9012 : SUCCESS in 11m 25s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/358eb28e134f4ad99fe3c02b9a13513b : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8b10ead113aa4cf78d9ca6aefe6d987a : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6c2fe79b80514484b881741cddb29002 : SUCCESS in 4m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98fa05f7c5c245439b48f864fb73e692 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c6628233d54472d9a05cee3a196dce7 : SUCCESS in 1h 41m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b1948797ed44a09af8b40b9ae6d0ac0 : SUCCESS in 2h 08m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/214cfed0731a4e57a6cd767eeda2fa85 : FAILURE in 2h 04m 01s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/59f1cde325e844e582eecfaf3b2a47aa : SUCCESS in 56m 31s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a2bc5419246843abab7c706aca07dd2d : SUCCESS in 1h 43m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dbe062b244d149bd89ad6057e6da5982 : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4842d51f1c47448b9d93710189cdec98 : SUCCESS in 1h 43m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5de4e3af4fbc42ddbd8c86c5a2218d75 : FAILURE in 2h 06m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6311380cbe6d4ea8b7a0648936cc7662 : SUCCESS in 1h 54m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/43627b6b51fa4d30b45b00e5f0c1384c : SUCCESS in 1h 38m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4836df2e1a4141c1a65669a6a5af5bda : SUCCESS in 1h 45m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/dcc00453b08b45b6b552762dd8ea3a98 : FAILURE in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8618217e0e69cfb1857e71d9226ce410bba294de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-12-25 14:31:24.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"06cddef4f8480ca69d3ca924f5a900fb80c51626","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-25 14:33:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"5f8179fb85e80d17e256bfafd0fc3cbcba584b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 16:26:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/868c6179cff64fc5af6168ed2e85b13b : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a1fca209abb41a5b2a312caef4e4f96 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc6097004a24016a57d34b044f5f6d2 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cbf60d482284cac9d9c2b0ce905b44a : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba85e7b1b594be2918c6a774f708af3 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c2ad8a1851f43059dcab5457c6c3512 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7976e3fcf65b4446a03e3d87de81be75 : SUCCESS in 11m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/633ee0179d034ea5be14f0506ae31f34 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d04b3e3b1f24460bb01efc2d3007d2fb : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f337e38d552249159de4cfee1adcab6c : SUCCESS in 4m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79040e1324de45829fe41003c5e5ca73 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0b215fa59994b92a66b7aa2fb284010 : SUCCESS in 1h 09m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6ee595d94824e8e9f1da2b0f58f897a : SUCCESS in 1h 42m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a1c0529f2c6f4fcf8e13fbb010847d5a : SUCCESS in 40m 23s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8ca3f615e12a454490bf3d2f8a350a2f : SUCCESS in 34m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d48152d79ef4b6f9fca8a6a6c4b1363 : SUCCESS in 39m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0b889b47358d4e62a874e23a7b84ab9e : SUCCESS in 57m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ddc584709f14e67a099e407d6185754 : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/06420269c4394c3d86254b16678516ec : FAILURE in 1h 53m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2283aabb89e848f0ad866aca2e6307c5 : SUCCESS in 1h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/4aa6fdd8bb2646b1bd49292dfd961cfc : SUCCESS in 1h 40m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e3d20962d9e54e129bd697e1396cb976 : SUCCESS in 1h 46m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7e0d23776cfb409b9e5b66a90cdbee0e : FAILURE in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3bd7a53416d4577c8f747f6cd33a1319d291cb1f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-26 04:25:28.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f58903e274ac977c1a585c52d94a880452a86410","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-04 20:31:22.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2e464bd998427a7c6249aa4e0d93b6f904c47d1b","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 10:35:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b209cfffe9fb4faee7d6fb5c8d83b16980a4f6ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:35:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fa175557c63b806ee3ef5a045607653ed6027615","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:13:21.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/585cd2e0880f413298909f7332013303 : CANCELED in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d1f442cfa94bf49d5391c13730701f : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a38bcfd4659454193fedc50fa19ccc0 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe64876e706d4a069deb612be3e656ae : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44e78241d654aefbbd621c65814888e : CANCELED in 7m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/96536bfe877c4d4a9c2574d35c55f166 : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57ad7c76343f4f5caf4c450f6f1080c8 : FAILURE in 6m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ef6014b2ab648789dc8e7ace78669e0 : CANCELED in 2m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d7f66103ec414b2b990a09a8edf0c4be : CANCELED in 6m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/251ff53b2adf49c99f7bd6ec2ba2a76c : CANCELED in 6m 15s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2ee7fa1dca0c461eb59e403602ac775c : CANCELED in 3m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/290d0c1668ab4ce49299b256e477d5ec : CANCELED in 5m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/81f0eb05519d492e9bcca0fb16ebb1bc : CANCELED in 6m 20s","accounts_in_message":[],"_revision_number":5},{"id":"a7801b0eebf3e0d36753be8fff9822fd1163ac41","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 20:20:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3f561cd64cc0742f3c6e6c660d2b1e1a0bff107f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 21:31:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54c58a426bef4928b0dd073be62640a4 : SUCCESS in 8m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de0858a995a0452a82cb20d1553c747d : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bfd6c5587674077a0ab8a2a6ff94098 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3afbd8ce6d450386077be3ef689eaa : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a8966e42b194425b1e7ec00d465a4a0 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/095d5daa802341b7a11b66552a944464 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7db07473761743338e4200aca85a8669 : SUCCESS in 12m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e884a5680f5f4c58aaa311392fdc933a : SUCCESS in 5m 16s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3bc4e9e772d84f42a83e77cca26d53f0 : SUCCESS in 12m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c307399271f49e893e8fd31002c7add : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e76cee309d4447e3b1b0d371df3c856b : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1cb7639f6e8484b8198690c3fdfd981 : FAILURE in 22m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/03e078f325594f8d8fe2a222fbb9d63b : FAILURE in 21m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/377f55545f5f464685282b54018d74e9 : FAILURE in 22m 58s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/da2680b4bb3d46518cb12bc118845c55 : FAILURE in 21m 53s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6909a456b0ce4984b5cab16ab6957566 : FAILURE in 20m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8c9aeb24108545689ee55a6e287d85dd : FAILURE in 44m 16s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94d932e522e94f78b2d5674d8b75a223 : FAILURE in 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2489273a28c74d6ba980d99f6078ef55 : FAILURE in 21m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c7f5812879fd4705a77a263a1da93352 : FAILURE in 56m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b9a2a86669d94530ab94b2c9c2a27933 : FAILURE in 19m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b3bf549ebac34acc85cbd474c8bacfdb : FAILURE in 18m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/74827ed07d2a41aaa967d98267fe2435 : FAILURE in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a0dfdf996ec8a2c6f949fb38bcbe1560d5066d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-28 21:31:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"654f1a0ef65b2d3db34423192969b7f911e5c590","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-01-28 21:44:48.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"55e02bea9cff994520baca36d7568752f59df7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 23:44:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b73384115df643fa800dbedd6d2ee2c4 : SUCCESS in 9m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca4b00b9a83f48dbab9ebc5f399f8fb9 : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57af2b8c24dd415b8139bf12d21e6e4c : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c5aaa5577e4807841a82f6459a4370 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b04cd55286c24bbf93b6cacde3df2ade : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f843fc6f9d744e039b043194b2df06c6 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34f885e2409140048120f6c106e11f7b : SUCCESS in 11m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fbf7528898244d7194c1b8ce93f73542 : SUCCESS in 4m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fa2fe13e87db416db16c9bb25b54f023 : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/053d7dcf24fa42b8b9533316bf113045 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f612864d1c2140308452e328447f77f3 : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5fe14f4f69f64dcc9369be769e430632 : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/62d39b6ab8d24be28a9f07824998732a : SUCCESS in 1h 46m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f890198f52c84e01b0a654ab696bcc81 : SUCCESS in 46m 03s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d51dbf9b5f6443c8b6d89775a4e5e054 : SUCCESS in 41m 51s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ecd7889458d84f86b7a260a7446380a2 : SUCCESS in 38m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c8d6cc7c2e674fb8aea3cf81eaa4b23f : SUCCESS in 1h 03m 27s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17ed52eb205648a190b5ec3dff47a9e4 : SUCCESS in 1h 42m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71433e3baedd4b75b422bed498da88d6 : FAILURE in 2h 01m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb1efb12f9854878961fc2c625f9cdbe : SUCCESS in 2h 01m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/482033e12a27497c868af427f25fc0d8 : FAILURE in 28m 17s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4c29d08a12a14635b763183b30fe1610 : SUCCESS in 1h 47m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e88dc825e9084e3f89613c7c89fd66e7 : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a2b82d2eb545012448ce65d2c904a8f5984ede74","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-01-29 23:00:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"1a678f80c4c4f4b769d932ed3be7b1bdf8f1ff08","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-01-30 21:29:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"d51744784c01a934d4a662326745896130774504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 21:29:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"65f33650e3473503521420a1cb290df5b1b1f903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 23:48:49.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/314c090d858b4d61ad84883bc4652808 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425a55e2efe54c9e8991f61dd315c15a : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42e1f6cc2f1d40628034b55f35b05c0c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4041e176de54c2da2ef36272fa8cdc2 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc3e29ff2534d40873f3d16401a22d2 : SUCCESS in 11m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8871f155f3ae407b9d3d0a05906852ad : SUCCESS in 4m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb7117f6961d4636a97a241acf662d35 : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4f0f4fca1644b259127fc6769579099 : SUCCESS in 1h 21m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce9048ae3373421a8c8380e00ec02864 : FAILURE in 2h 17m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07d38b0a13834bdf8659c85427ca725c : SUCCESS in 34m 35s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/c3c9d8ef9ea6428f8724762bf79d1a78 : SUCCESS in 1h 03m 48s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efb46784ad4e4ec885debb1521b0c834 : SUCCESS in 34m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8a254eb1aca24e2a9bd8d7d3af1f75d8 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":6},{"id":"b6dc643bd8fc1dcfd86a7c83f8efffd75eac3453","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2021-02-01 16:45:34.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a401cd7551605967ff650daa08bc1d3dffbb3fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 19:07:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb68335d217d452b9d3e1627db82f454 : SUCCESS in 10m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5b6f28d97df47e29eee616ca5b99b88 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f15852d40e4910b86c677a3c5a21c6 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a04c07a381349dc9643c05d53f24079 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93080a9aeed44078a10eada4e34b3713 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92cae81539464aab8f63c22daccfd0fa : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca1220f3fd34c919976dcb7982e5672 : SUCCESS in 19m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/72ca2811224e47dc98454612756d7603 : SUCCESS in 7m 40s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/326dca59a51d41e7bd1887af46565b9d : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/75f60a53fb824008b9af2d51b60bc5fb : SUCCESS in 5m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ade95b7a25b48eab720833f7ed76bda : SUCCESS in 8m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/334ab7d6c43d4133a53ca310b294eadc : SUCCESS in 1h 12m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/71940096a03c4c44a85af9bb9675d076 : SUCCESS in 1h 43m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b70cdc2051e446a9995291352d76840 : SUCCESS in 41m 43s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/cfeb517e47b04d5ea16914ffdbe80c48 : SUCCESS in 1h 06m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e9e8050c1944459808c6d68b0eb62d9 : SUCCESS in 38m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7daf7c744524056952a1c873ac33c46 : SUCCESS in 1h 03m 01s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b8b47eb0982d4dba85e0f5e16a16bdc0 : SUCCESS in 2h 02m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3e6d0171c80d463ea01c0464c996dfc2 : FAILURE in 2h 14m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f67af117e274376968b9f788d7d6d99 : SUCCESS in 1h 56m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d5f297645fb04678a2bbf5547cb83c05 : FAILURE in 21m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0d71f1307bd04a51a9e8aad34224f8bd : SUCCESS in 1h 51m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/7b116af7ff464219bc3337458364bf1a : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fc62a3aa052d9469bc4048c5aa37e24625ee268d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 19:07:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"be78fe81ebfd02fdb39b4dc6b199a95ee0121a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 21:05:44.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68116b4a755949529046bb4a6915bbbe : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e56545d77304b898a15511634e215d1 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/666da02bf99342f1b702d69b95892b17 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5044fa54ae6842ccac403af05f9d9c33 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3deb2d3938b463fb35bdc3f8b787194 : SUCCESS in 11m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3e04f19a362c417cbeaf440411908cd7 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b07bbe6aa10c44f4893016e74a9b2fac : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97080db5d73446cd866164230447e136 : SUCCESS in 1h 14m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce682168fb0742fd8faa2cda3b1ea940 : SUCCESS in 1h 55m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5042539e59747f7854b6e0634fdea39 : SUCCESS in 42m 37s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/dd485566cc9a4c56b8c42b46f66bf4a4 : SUCCESS in 44m 11s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/422359fe47c44c2ab48e5a9752b4e5b2 : SUCCESS in 53m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/89d60e53313d4d5b85b82b9e66a6b7b8 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":6},{"id":"2eb372f4d62a21629d149e9fba103c043b9f8484","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 21:05:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"41fec43e31c61b0872268b7f35e6ff3f48e89303","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 21:07:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f11101f0c33402da55e75d177280983 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"53f8c28aa3fadc0feaaad9d073f006dc36a39b7a","revisions":{"3f14ea5643953820a4515824c431d05ec60e7aae":{"kind":"REWORK","_number":1,"created":"2020-11-04 07:06:27.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/83/761283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/1"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 07:06:23.000000000","tz":480},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f14ea5643953820a4515824c431d05ec60e7aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f14ea5643953820a4515824c431d05ec60e7aae"}]},"branch":"refs/heads/master"},"8e5df59f8ad52b3aafcbfbacb30e40d0bf670e2b":{"kind":"REWORK","_number":2,"created":"2020-12-03 07:27:57.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/83/761283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/2"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-12-03 07:26:18.000000000","tz":480},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5df59f8ad52b3aafcbfbacb30e40d0bf670e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8e5df59f8ad52b3aafcbfbacb30e40d0bf670e2b"}]},"branch":"refs/heads/master"},"52345125bbe954155fe03de35c9a00061fe2368d":{"kind":"REWORK","_number":3,"created":"2020-12-04 07:26:43.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/83/761283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/3"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-12-04 07:26:07.000000000","tz":480},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52345125bbe954155fe03de35c9a00061fe2368d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52345125bbe954155fe03de35c9a00061fe2368d"}]},"branch":"refs/heads/master"},"954e99315a4ed3fefe88f3183b762628275e5064":{"kind":"REWORK","_number":4,"created":"2020-12-04 09:28:19.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/83/761283/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/4"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-12-04 09:28:05.000000000","tz":480},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954e99315a4ed3fefe88f3183b762628275e5064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/954e99315a4ed3fefe88f3183b762628275e5064"}]},"branch":"refs/heads/master"},"133c041b31e878f12972143da82041cfa5b58e48":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-12-25 14:31:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/83/761283/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/5"}}},"commit":{"parents":[{"commit":"1a86e454e7be22b36039a4a4bac6a886f523e9d3","subject":"Use \u0027bash\u0027 in the keepalived check script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a86e454e7be22b36039a4a4bac6a886f523e9d3"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-12-25 14:31:24.000000000","tz":0},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/133c041b31e878f12972143da82041cfa5b58e48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/133c041b31e878f12972143da82041cfa5b58e48"}]},"branch":"refs/heads/master","description":"Rebase"},"53f8c28aa3fadc0feaaad9d073f006dc36a39b7a":{"kind":"REWORK","_number":6,"created":"2021-01-28 21:31:43.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/83/761283/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/761283/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/761283/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/761283/6"}}},"commit":{"parents":[{"commit":"d2d5fc80f8274b7330c896b6825909ac04b653e0","subject":"Add ALPN support for TLS-enabled pools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2d5fc80f8274b7330c896b6825909ac04b653e0"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-11-04 06:54:03.000000000","tz":480},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2021-01-28 21:31:21.000000000","tz":-480},"subject":"Add default value for enabled column in l7rule table","message":"Add default value for enabled column in l7rule table\n\nwe don\u0027t set default value for enabled column now,\nand mysql will fill them with 0 value(false)\nif we upgrade from ocata or ealier, all l7rules is disabled.\nso they will not render to haproxy config file in amphora.\nthis will cause all l7rules  stop work after a update.\n\nChange-Id: Ib8549d71b7adb95d0dcb36903c1744190056adae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53f8c28aa3fadc0feaaad9d073f006dc36a39b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/53f8c28aa3fadc0feaaad9d073f006dc36a39b7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
