)]}'
{"id":"openstack%2Foctavia~762565","triplet_id":"openstack%2Foctavia~master~I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","project":"openstack/octavia","branch":"master","topic":"fix-gate","hashtags":[],"change_id":"I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","subject":"Fix pep8 failures in gate","status":"MERGED","created":"2020-11-12 19:00:41.000000000","updated":"2020-11-16 18:54:48.000000000","submitted":"2020-11-16 16:05:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762565-1605542715896-6545178d","meta_rev_id":"ab9405944317530cd4124391b804773cc0131f23","_number":762565,"virtual_id_number":762565,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"date":"2020-11-16 11:45:24.000000000","_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-11-13 14:25:10.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-16 16:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2020-11-13 01:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-11-13 00:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2020-11-13 01:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-13 00:47:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-13 01:35:13.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2020-11-16 11:45:24.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-16 16:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d0051e27f121ccf4c04a940f7798e899f409b5e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 19:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989f0d56c5bf94b2684dfc26e390a6006105441b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-12 19:14:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(10 comments)\n\nThis looks good, just the \"str(e)\" issue.","accounts_in_message":[],"_revision_number":1},{"id":"8c37066109231afc0654c202ffd46fb516ca2af5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-12 20:50:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f5544ec0f6fd8fb66eeed8b9c16c177af32d457","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-12 20:54:21.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"f05642cc4ad6a3c43f3d14a76e44dea77534950e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 00:46:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e486fdc8cb636f83a6d3eb20a4b8da8f6b39338","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 00:47:36.000000000","message":"Patch Set 3: Code-Review+2\n\nMerged in the grenade job fix as this won\u0027t merge unless that is fixed as well.","accounts_in_message":[],"_revision_number":3},{"id":"1ceecf718608b62100f430cb79f6ed35f7240235","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-11-13 01:35:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"703a640087c7ca8de191c2d17f70fc50917d55e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 03:08:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/354f283f402f46c7b5e1604f5f1a27a9 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63100bf33aa04f67a7d62442c5ed0d35 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936f60126935465e8b7dd9245ffa2d0e : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46804ee726c04a489079ebd1f84afe36 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15197b13f84f4f1a8f4476b3dc4df352 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252531ee9fc340739f5bff99fd998480 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ad2757aed84e48a5b30cfe192541fc : SUCCESS in 17m 49s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8cbe414c26a44d0783619471249ec23d : SUCCESS in 4m 53s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9053a6cd4abc4bf7bce4f5186201e5da : SUCCESS in 6m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f622834546804b508476eefa5c19b751 : SUCCESS in 4m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2895f050aa0e4c49af437876f06b28e0 : SUCCESS in 6m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cecea855fa24122b832ebefbcdd87b5 : SUCCESS in 1h 26m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ba7fd286a0743feb441f3a671c6b9fe : SUCCESS in 1h 41m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbe6db044c9b4430bea53b808b054873 : SUCCESS in 48m 42s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f2d6b9ba85324f08bb406c4bd1afb37f : SUCCESS in 44m 56s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f85a59d2786f443286f5d0c9b4b42b56 : SUCCESS in 39m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b560faf0f62242f89309397f314f9f7c : FAILURE in 1h 05m 55s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f401db5d6274e5db1fc5aec6c821c61 : SUCCESS in 1h 53m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3466ae3f7a9a474d9c9ddea12935f01f : FAILURE in 2h 19m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c59cc19083e64f3ebda0e242111a9b26 : SUCCESS in 1h 25m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cc196f9de6cb4f25b15398f9cd256419 : SUCCESS in 2h 07m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/11fd8c4379264018a2b63d1cce212c42 : SUCCESS in 2h 12m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/072825f8791945c18f2f27742ac735c9 : FAILURE in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd0ca2aed9c3e50c00d0c20983a41ae86d79129a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-13 03:08:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1aeb028490add43971ec50ca6f9e1544edfda41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 05:44:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f2ca6f33ac5462185042cb7021793c4 : SUCCESS in 11m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c89efc8ffc484e3d91ce2d667171e479 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f93124c3c9a4e43b9014d0033304547 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/862a142d447d40d98afa6711ae6357cb : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47336dded6c8447aa4f8c3cb11b22b3d : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/293b07b5efa142b0847845a56a991a62 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c948930d289d41c6acf1504b5574669e : SUCCESS in 16m 47s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/deefe10229aa43198d3f9239a5cccf88 : SUCCESS in 7m 35s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/94cd4b1e227f4b76815009b591396a24 : SUCCESS in 6m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f80a54ac20a45e2bc15a72147f303e7 : SUCCESS in 4m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ae1943b0292416996c9b341c8e01222 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ccd61ac3fa84c988959c03730ab24fe : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/caadb68759524ea1bf81213026bf3d5a : SUCCESS in 1h 53m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2dcdcbeb5323463d907f3265fe306609 : SUCCESS in 40m 33s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/d1ba37d33dac46d2b29fc24d7bbf4c5e : SUCCESS in 43m 07s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/957c7e494343436e82f8f6cfd10bcd41 : SUCCESS in 37m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9427f9d47aa4eefb90ab5c8215d80ed : FAILURE in 52m 48s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/774efd4103204aec8a146eae389371f4 : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc05173614414593974e9e15f81c98bb : FAILURE in 2h 22m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/141ba017ec6c4422947dc6931c9b32e3 : FAILURE in 1h 51m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f2dd54e265b14f4d94c1821400487098 : SUCCESS in 2h 13m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/77f2550cfa15487abddf1b87dd42c86f : SUCCESS in 2h 00m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/6bf6a0f55707459499d5192c1c44bee9 : FAILURE in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55272c83fe1e8b489a2aa84cb7ad7f317c8bb47c","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-13 09:28:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5d1cf970d4b22bfd333f0876fa4cb29840942bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 11:55:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc6237452b7f40c39b448c1f83c7ad78 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f711c1778594b21927af36b5617c598 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d78d03463a488691393f758bb3c4da : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afc17dde041b4861bb1c3e32e03c65c7 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/821376eec66848eb9d03650256d57ad7 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb90d501bed492584d92c219de963d4 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0056bdc453e34d73b23b460dfdea5a43 : SUCCESS in 16m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a29bd45092424571b5b097d2d146b7a3 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/d263456bfd644af2b64e474f78df00ff : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5799bb8c68bf4bb2989367f017201632 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b54ccb7bf0b147f19a7f572a76ffed33 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17bead80202342deb78193bec9eb4e6e : SUCCESS in 1h 19m 56s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8a18cb2161244c6b8189dd3cd743ebfb : SUCCESS in 2h 01m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9611835927684082acad902ad001146e : SUCCESS in 46m 36s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4d9c6c9ac5a44a718b120d70038dd0db : SUCCESS in 40m 01s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ba59c49b453e4cb9b7ea2c89dca42a57 : SUCCESS in 43m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c7564cf8831c4309922468547d73f6ef : FAILURE in 57m 06s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6678ca99062348d4a3d73b0775df6962 : SUCCESS in 1h 47m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/73267d1071e84168bb61a65b8765b97d : FAILURE in 2h 18m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60ebdde52ed944829d22954779e15717 : FAILURE in 1h 48m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/78deeccb0b27485890fd9840cef66e78 : SUCCESS in 2h 02m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/99f20a27175040ea894800ba6d46db30 : SUCCESS in 2h 00m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ac5daf2a14b246269734813aab796a3b : FAILURE in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92033c6cb2cb0a92de8e596c95d693bbd190738e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-13 14:25:10.000000000","message":"Patch Set 3:\n\nFrom job-output.txt:\n\nSending command: scp -v -t /dev/shm/test_server.bin\\r\\n/opt/stack/new/tempest/.tox/tempest/lib/python3.6/site-packages/octavia_tempest_plugin/contrib/test_server/test_server.bin: No such file or directory\n\nMaybe that\u0027s what Michael was trying to address with the jobs.yaml addition?","accounts_in_message":[],"_revision_number":3},{"id":"3cf0866b6390863ce6fffc5d31e0a321ef55a3d4","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2020-11-13 17:26:51.000000000","message":"Patch Set 3:\n\nYes.","accounts_in_message":[],"_revision_number":3},{"id":"a3c3f669c6c6112bbe9447b4071b5a61b4bbdffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 19:46:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac3208c6dade4390b9385c4997e135a0 : SUCCESS in 9m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba09b26043ad4c7eae69f490e7e18211 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1796881b7d54bb89f2a1cd16f2b8544 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9100397e95741ec8fddf0a0b23c22c4 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e2da9c715634a02879b8c635997422f : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c780f22bad40f390be84dbc725dcce : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15aab9a55f34432be976f13d7558da0 : SUCCESS in 18m 48s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/db0d70f280c34adb9bdcb9e0ded66601 : SUCCESS in 5m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/c555793afc644cdc9822e315c8440c40 : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/80c6ab33e2ac40789e230ddc8a991f93 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c51dcc2a9a0049d78ab96b9a042dfd1b : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2196d7c9e6894146a2d57a4f709488b2 : SUCCESS in 1h 23m 13s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c976387481c4b03a2259d3923db83e8 : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5b33b6107724daf930bc502fe80de38 : SUCCESS in 1h 04m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/18c157d70ebe4defaaf2416b9e2c3e08 : SUCCESS in 41m 44s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5d20cbd6e8a431997699e327f99474e : SUCCESS in 43m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ff72b41ec6b4a8e96ab601f62b2affe : FAILURE in 50m 36s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dfb394d0f9d448f2b3e35420a4de456d : SUCCESS in 1h 52m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c5963c3b84894b85a5bec423acdb4665 : FAILURE in 2h 11m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/26dd8a1cd40d4f3fb0fe4aadfbce4458 : SUCCESS in 1h 46m 45s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5520a81fb2e340ee818f7af0a9c091f6 : SUCCESS in 1h 43m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4a5207033fc64ce780f44638032001a8 : FAILURE in 2h 01m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/5488c734b2904a1c84d58827f2609faf : FAILURE in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a43d327fb1a803988d65a4ccfb7ea2a20511711d","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-16 11:45:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"503aaecd02caeaa9f555326abe266c849c111679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 14:12:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed5a03a67bf44eff8651ab34c00fff40 : SUCCESS in 8m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b47781a37e404591a0ec59b25347cea0 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4f5e6db539481aa0c503b67c7af9cb : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721c97f89bc943dd92e9d71a271d2355 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220e7c48434e4cdb8c853f448d35973b : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2f50704f7554da09e9574ac63cec4c1 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9648bf4728a34dd9a3f357ab977080ea : SUCCESS in 17m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/1d5fb9ed76fd4ec48003649b60742fd2 : SUCCESS in 5m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13d0a03c92b342ddb3486ec3ae283412 : SUCCESS in 7m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dbe875e2c664c58be5e831186dc1e1c : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ffbccf856ab43ababa74cff0756458c : SUCCESS in 6m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc6fee6f6ca245fe8577f994cb19bcfd : SUCCESS in 1h 16m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/275c60583f6747988edd7f5263d7d787 : SUCCESS in 1h 37m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c131c6743a3e4cd595a16be9eef314fb : SUCCESS in 45m 17s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a375c6d867dc4091b07db342f5b81dfc : SUCCESS in 42m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79fe672a943144819fb4035cb78a4fce : SUCCESS in 42m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/40d9b5b0abc240179e642e7057cedeca : SUCCESS in 1h 00m 31s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dcc9737747ad4486bef3e542aaebdba7 : SUCCESS in 1h 58m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00773e9949f14ec3aa15ab13bb1600ee : FAILURE in 2h 14m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c15d1d3d1feb465b921d574641dd3e69 : SUCCESS in 1h 37m 20s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f977af548093443a837d6df447db83dc : SUCCESS in 1h 37m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/050aebfbbb6a432b91cd88e639620c7c : SUCCESS in 2h 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/341838ab68804818a05524adf81edc4c : FAILURE in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b56d26b2e6fac9ea81dcf5c8eb5e452e376e5a0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 14:12:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3d1da165fed90f7da7dbe9f7ccc03facd64f873b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 16:05:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b2a312f5a99420a9fd7603eacddd3c7 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8363b358dd4d1e90b6b3249fb9bb8c : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c0808b469146dc8a8114cb1baa07f9 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92a5d867957d450e82c71ea11f04a7c6 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b192551af0a14abda314f1db36f7f5ea : SUCCESS in 20m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/919ab575118140648c7356e318bd7bd8 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45b07a2f90ae4c60b74b6bfc49fbbe74 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4fd3e09611f9482690a16ddb7b1a18ff : SUCCESS in 1h 01m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73464c9450f44f9da3c945ce6a88e11d : SUCCESS in 1h 49m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/21a6e227094440fd9877cfba2885fa8e : SUCCESS in 40m 30s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/289671d1056d448a9f8cd3d36de72ef0 : SUCCESS in 37m 46s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1bc15283446b4cbcba32812f01828edc : SUCCESS in 39m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9304f1a996c3481ca5fd986df7c2c999 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":3},{"id":"4f142bbb23514bfb38d7af0e7702873a68250a4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 16:05:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0a81b018f9d1e49afbea71e010529ba409e9e324","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 16:11:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83991b10deeb48beacd4fba5f0c1b204 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3},{"id":"520eeaa1b7b9d5382db26393941d3692fd8fa126","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 18:54:48.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit fc058c3a8367c369ab623083c34cb423f808c52e","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","revisions":{"e5f5808ffb8dec070d73cb1d71ec327cda810873":{"kind":"REWORK","_number":1,"created":"2020-11-12 19:00:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/762565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/762565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/762565/1"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"subject":"Fix pep8 failures in gate","message":"Fix pep8 failures in gate\n\nSome package must have moved forward recently, causing\nthe pep8 job to start failing.  Fix the issues since\nthey\u0027re all valid.\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5f5808ffb8dec070d73cb1d71ec327cda810873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5f5808ffb8dec070d73cb1d71ec327cda810873"}]},"branch":"refs/heads/master"},"f8d0c313026c62244c0fec11d1029f056841afac":{"kind":"REWORK","_number":2,"created":"2020-11-12 20:50:17.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/762565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/762565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/762565/2"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 20:48:13.000000000","tz":-300},"subject":"Fix pep8 failures in gate","message":"Fix pep8 failures in gate\n\nSome package must have moved forward recently, causing\nthe pep8 job to start failing.  Fix the issues since\nthey\u0027re all valid.\n\nAlso changed other messages using an exception string\nto use str(e) to avoid any possible py3 issues.\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8d0c313026c62244c0fec11d1029f056841afac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f8d0c313026c62244c0fec11d1029f056841afac"}]},"branch":"refs/heads/master"},"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd":{"kind":"REWORK","_number":3,"created":"2020-11-13 00:46:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/762565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/762565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/762565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/762565/3"}}},"commit":{"parents":[{"commit":"198980639cbe02820f90783fa04d5239717ed80d","subject":"Merge \"Use run_flow for failover functions in v2 control worker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/198980639cbe02820f90783fa04d5239717ed80d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-13 00:46:36.000000000","tz":-480},"subject":"Fix pep8 failures in gate","message":"Fix pep8 failures in gate\n\nSome package must have moved forward recently, causing\nthe pep8 job to start failing.  Fix the issues since\nthey\u0027re all valid.\n\nAlso changed other messages using an exception string\nto use str(e) to avoid any possible py3 issues.\n\nFixes the grenade job to properly build the test server.\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
