)]}'
{"id":"openstack%2Foctavia~762677","triplet_id":"openstack%2Foctavia~stable%2Ftrain~I254557083715328fa02cf99f03e8c83037b4d640","project":"openstack/octavia","branch":"stable/train","topic":"train-grenade-otp","hashtags":[],"change_id":"I254557083715328fa02cf99f03e8c83037b4d640","subject":"Enable octavia-tempest-plugin for grenade job","status":"MERGED","created":"2020-11-13 15:53:28.000000000","updated":"2020-12-01 19:17:52.000000000","submitted":"2020-12-01 19:14:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762677-train-grenade-otp","meta_rev_id":"b0521332f8596045b48b2cc4f453be6118616492","_number":762677,"virtual_id_number":762677,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 19:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-11-19 08:35:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-01 17:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-11-18 20:28:00.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-01 17:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 16:51:58.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-18 19:45:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 08:35:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"093c70596f21fd85539926ec6401acbd977d753b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-13 15:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1bb60b982c9bd90add73df881604697d786eb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 17:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e144ea1065456abf0b7138122068dc : SUCCESS in 8m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44dea14d1d2d4dfaa66c0d56987e1599 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1060bda8393d4b3a861b1ef602ac83bd : SUCCESS in 6m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3717c50ca93f46c2b8f366bc45151d22 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1daf8e5c1e0a4a15a593220a16f7cca5 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d59ed52aa5444d9bb9aaa27f6efddc : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfcc90933b064ce7b29cb3a1ae0e27c2 : SUCCESS in 13m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f7fb838ff8434e4da1d428d5cdc56fc5 : SUCCESS in 4m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6835e9a2a2ea42b1bf74d9a624cb623a : SUCCESS in 8m 05s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/71283ed6fd204b26b7715ceb89ac7ea5 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71a7e40e2fb74fa1b8233d893596193c : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ebbc403e95d04b8e82fb7a34cc09cd49 : SUCCESS in 1h 16m 47s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/cec5f80299cf4789a65b15be27b0ccca : FAILURE in 14m 47s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b199162e910147a38cfb2466f7f04ac1 : SUCCESS in 1h 44m 31s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/5f027314eeb64ff590443b56369350c0 : FAILURE in 15m 11s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/8482e008c6324321bd763a4485296d8d : FAILURE in 16m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/18b6fe66aa4040a9964b89a20f7eb0da : FAILURE in 26m 55s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/5d091baacbeb4cc98f649e7634a6d999 : FAILURE in 18m 34s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e18f0cd032144aaa2341ffc41050ca4 : FAILURE in 1h 22m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bafe93cde924b009d439cf07a1843a6 : SUCCESS in 44m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/7d9631f79a004a83ac56c6c802bee852 : SUCCESS in 1h 23m 02s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb6c4bdc001d4d0390de7c1ea4f0cfed : SUCCESS in 41m 17s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/98156f0a1f4a440d93ed7d4a3325a5b6 : FAILURE in 23m 52s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/c54daf64b5e247dead775e57ff2ac1b5 : FAILURE in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aeba79305ab9568e7b435b205114c48db3de53cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-13 17:48:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dfd9d4365079c039936f766fc3493ad089ed928","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 19:03:58.000000000","message":"Patch Set 2: Code-Review+2\n\nYep, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"4a9a0915d9ac003ea132b22837b7b8efae16672c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 19:15:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbfb40757a1b4eebbb582ebb8ea8e0d8 : SUCCESS in 9m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2917ce79a00d4c938465d7c909821522 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6ac5ddc0064900a4c4abae4875b3ed : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/940e37338890434890e7c4c703c782a4 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3af608682fba41fd9a88e6c76ff11944 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e3842fb099545b4ba999c9d12a85fab : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f7ece2aa034a188730b88a2513d8db : SUCCESS in 10m 53s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c4641f090f2d4c7cb70e8de10a6aeaca : SUCCESS in 6m 08s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8d4d4d3581a44d8b8c4955a20cd968f9 : SUCCESS in 7m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/535dbe3626d44ed491b226e489e8bccd : FAILURE in 1h 18m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d6f17312aa54aa76229c68bd9c156382c37881a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 19:43:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"353e650762674d74ad071c9ea5af2930d8e99b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 21:11:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a65d68bbea44a318023a10058813a7b : SUCCESS in 9m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce25b5fcfcd0400bb42f1750e77ce60a : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e99af6f0364c4a9c566a1759c7984a : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c7f28cba56b425683eb26bee922495d : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20e7548930dc42a8bfc9d246029d8d3e : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ad5ebe033cd446da88c70dc2f9a3354 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e76d03bc38a449eb9918b387ca322a0 : SUCCESS in 10m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bb45f9f50d54489a981cc52f72121510 : SUCCESS in 5m 10s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/83f3eedfedb24bddb4c00bf95638e5ff : SUCCESS in 7m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9bda6335d7494eb58144aa54ce71a3d9 : FAILURE in 1h 19m 04s","accounts_in_message":[],"_revision_number":3},{"id":"17d6fbb7360c7ba998b9d88581fb2255cf745aa7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 21:59:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"80c461efa293114d973b2ae70fd2250a7917fd07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-13 22:38:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90e3d6b3a01b4c0a91631c5cad42c030 : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/072befb16393446d85380cf565bd2e54 : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce2ea224d474c01a8948b11dd1bccd9 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db21c6bb50a749ceb9a94ba4bcc0757f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4afb87dc1d394cf881fef7993adb1d16 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2d5cd0298d24a81b83ae47eb8a7334e : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6eaeaf882a45fe89a7212d0a98ad82 : SUCCESS in 10m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c6f8981c774c4d51aecd7ee65c77b7c8 : SUCCESS in 4m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/03122aece87b46559561538f0a01f8a1 : SUCCESS in 7m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8119d78526af4656917d5641ea66a9ba : FAILURE in 35m 51s","accounts_in_message":[],"_revision_number":4},{"id":"a7646e4cb2360e8acc21572822ff5e75d03d9592","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-13 22:42:07.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c1e1c3261741523d9eef632747c5d04346d7d27b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 00:05:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3ada026e17049a5a992e2c4a139a6c8 : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce55a08ea610437fb23dde09266b0e38 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9eb3ebcf91547c9b3727a1887fcc5af : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82e2bc3bbe5b4418b5dee68c1f5644b4 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a45212520ab349f4b1199f5c2ed2491e : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0dba92f7ce04ea9a5a3f5f17d0127c3 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b089fb2943cf408c819a528e0359a519 : SUCCESS in 11m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/27b69071b4d74c48b970d74a933c6f1e : SUCCESS in 5m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e1b1453bb40f4b36b0f78d9ad583a7f4 : SUCCESS in 8m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/08ead13a557e4d618b7d82a6071c7b8c : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":4},{"id":"27adb036be6ed70e4b606d653df8cfc7bbdce975","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-14 22:25:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3e2ccfda567de01d2e22662b2d2e65095cf81b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-15 00:09:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f507ca2da7a479c9284cc83604b3015 : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c319ff00dda4433b598f53cbef29dd9 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af26d755d94e40e183e1104947c5c682 : SUCCESS in 6m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a33b1ed9e36445278f3e24b0d78444f6 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/626ff92f5c6f4fa0bdc73f93cf20ca39 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af92a70293114363b3132b671624d88d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cf68274eb54963933bfd313dd8a281 : SUCCESS in 11m 33s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/36ec5d5d00564ba09bfe546193ae550f : SUCCESS in 4m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6070811d9e47419f81741eb4eee2962e : SUCCESS in 8m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/456b960e841f4039887bedee524f48ec : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ad8daefc265416e8baa1c79cf08072b : SUCCESS in 5m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fdbef4947c34c448cba8bdf64c1a97c : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/8089370af36542ec9bba1c9479acbdeb : FAILURE in 12m 49s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95465a6be6844c8da993f216c1571706 : SUCCESS in 1h 37m 14s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/7b78be1066ba4d089e270b8b60480fce : FAILURE in 13m 09s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/12a99cb58f0a464c823b65934c39165d : FAILURE in 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3c9a1580c894e36a69cd378fa333dc8 : FAILURE in 23m 41s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/f18efc3f447b4c49ad8e4bb6eb6f7e6a : FAILURE in 15m 16s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f5bf598e27f641f080885ed31f27b7aa : SUCCESS in 1h 31m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/27733fbeeb5249ddb7777c6f0eb9a6fc : SUCCESS in 37m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/34ba73c4a94a4f0c8187efcb0bef3512 : SUCCESS in 1h 42m 19s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/df34931fbeea4c8e91064cff41814f24 : SUCCESS in 41m 47s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/6179d835973449a29eb2e15a4d3a0a63 : FAILURE in 12m 54s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/5374c341d98d436e9e76ddb0db9f02e5 : FAILURE in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"98bd77f34a81e67c28b2f3bfa8ea4da5f995bd45","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 13:17:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8e3dad5ef31fe275a13711c2ecf87782dc124db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 15:16:29.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ead5033448604329b6236d09c48ba547 : SUCCESS in 8m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/268bca267bf049e29e0d3dde1e92dc9d : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7669490d6147466291f5fdd5014374c4 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b968fe04a24442e90d940ba3ce8957a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/193153946fe6480bb2a3916d06734775 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/782dc5df21184773b80df9b47a58be1f : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05ee0324d1b40b4bdf165beb91077f8 : SUCCESS in 9m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/786a677b2cdd474283da576de8f41743 : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/13ffe6ccedf74e4fa621b10b766558ab : SUCCESS in 11m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3583be25da78415489cbc6c28fe13ab6 : SUCCESS in 9m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f817aa3d4c32485ba73da217af77aef1 : SUCCESS in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c684f95f1c6842dbbd5ce976b109f6da : SUCCESS in 1h 21m 58s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/b2f2b87acfda46ee91edd4f924535754 : FAILURE in 19m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/31284ebacc704483a055ddf0cbe45efb : SUCCESS in 1h 47m 47s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/15b50312251a486a8c94fc2e13c83336 : FAILURE in 15m 05s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/4de892fb519141c380bec222a03b782a : FAILURE in 24m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fb02cf7318b40b3be276bcf5b489892 : FAILURE in 28m 18s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/17d78942a18e45cfbe360fabfac6217b : FAILURE in 32m 53s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63c2d389d2e341cdba74aa59cab66af0 : SUCCESS in 1h 43m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73b3c93fc94a4af3bdfae08a54f77f16 : SUCCESS in 54m 13s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9c550c6d30ba498186013c7a8473a364 : SUCCESS in 1h 54m 23s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52a03b0913e144e5859b9ace7d6f804e : SUCCESS in 42m 42s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/46688c8f8bc34b6285e53e22ddff8438 : FAILURE in 15m 57s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/d6841a3fb554469a8b5b9d6ca8dee094 : FAILURE in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8289fd0b40ee08be22e079c5e7972fd5e769cc48","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 16:19:08.000000000","message":"Patch Set 5:\n\nThe noop-py2-api job had an interesting failure:\n\nObtaining file:///opt/stack/octavia-tempest-plugin\noctavia-tempest-plugin requires Python \u0027\u003e\u003d3.6\u0027 but the running Python is 2.7.17\n\nLooking at the commits we did merge two changes:\n\ncommit bd587263755716f3de4987ae177e97fc28e957c0 (review/jacky06/725758)\nAuthor: jacky06 \u003czhang.min@99cloud.net\u003e\nDate:   Wed May 6 13:21:06 2020 +0800\n\n    Remove all usage of six library\n    \n    Convert all code to not require six library and instead\n    use python 3.x logic.\n\ncommit 47e4589d453ad69d94b004a1e335032481300b85\nAuthor: Brian Haley \u003cbhaley@redhat.com\u003e\nDate:   Tue May 19 15:56:11 2020 -0400\n\n    Drop python 2.7 support and testing\n    \n    OpenStack dropped py2.7 support in Ussuri cycle.\n    \n    octavia-tempest-plugin is ready with python 3 so it\u0027s ok to drop\n    python 2.7 support.\n    \n    Also added requirements test support to tox.ini\n    \n    Modeled on similar neutron-tempest-plugin change.\n\nI know that second guy :(\n\nSo maybe we need to set the py2 job as non-voting on train.","accounts_in_message":[],"_revision_number":5},{"id":"44f8fb45841a19d31f52c758cae33ba7d4f7498d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 16:21:21.000000000","message":"Patch Set 5:\n\nStrangely though, only recent patches are failing on that job, \"older\" ones are not, but probably will on a recheck.","accounts_in_message":[],"_revision_number":5},{"id":"b7e8402b5e81e07b2d5c47dfec73470b72c9079f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 19:39:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"24ec53c4b724bcf31b0857a3b61d4a7ea05b4f1d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-17 20:23:01.000000000","message":"Patch Set 6: Code-Review-1\n\nPython 2 is a supported and tested version in Train. It must not be disabled until we exhaust potential fixes/workarounds options.","accounts_in_message":[],"_revision_number":6},{"id":"ab29a1139d02d2321bf0a9e2f8a34074a6c4aa76","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-17 20:26:05.000000000","message":"Patch Set 6: Code-Review+2\n\nUgh. However py2-scenario is already non-voting. So be it...","accounts_in_message":[],"_revision_number":6},{"id":"4430f8fbf41c5a2a12a56b6af0f82575357aca2d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-17 21:07:35.000000000","message":"Patch Set 6:\n\nCarlos, yeah, I didn\u0027t want to disable it but it\u0027s broken, partly by a change I made.  Just not sure what else we can do here.","accounts_in_message":[],"_revision_number":6},{"id":"fc3eabb7d59c19c08646809253877258a98bf7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 21:46:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c500533cc5c47fb823903bb052d6d47 : SUCCESS in 8m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eb041bcd92b454da9c01c120bacdf47 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfd7283a99d946ed9c8b9816f3c00917 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b26bc64faf564f63a39ca1d74910887f : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8706302e84de4799a99dd13c686cd657 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cc3bf5968bc438c8c348906effff9a8 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83de667a7a574718aecaed2470789a4a : SUCCESS in 11m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/cf3017daab4f421c849f68093fb467b3 : SUCCESS in 5m 00s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0fc4f0e3ea15479e8716ee33c806ae34 : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/63bf5a134f0c405ea3e1ba175fd70354 : SUCCESS in 12m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d21853be31d54057ad1828acd013ce81 : SUCCESS in 15m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b0090a7f7f9f4a76ab035f617eb6c9bf : SUCCESS in 1h 04m 10s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/e96d00e9440442b0b6d9486f037c8ce0 : FAILURE in 15m 24s (non-voting)\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e491d3e060d448d48d82b34a3324f1f2 : SUCCESS in 1h 58m 53s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/067500b38bc04266b14b394ea77286fc : FAILURE in 17m 00s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/f6f659da7f2d40b59a3ac43533508da3 : FAILURE in 12m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a204a8070ae43b7ba01b7b9707aefef : FAILURE in 29m 30s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/b6a5598d9e644d0b9349e3f0a69b95fe : FAILURE in 37m 55s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a041a1d955947b09e0098fcd5c192a7 : SUCCESS in 1h 30m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2dc8e86ecda34f97ba2a31265af873cc : SUCCESS in 42m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/793a3ed7a3ad4a7a93417135edb998e5 : SUCCESS in 2h 02m 11s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d26e80f6036c4b9fb78d14df0f38d0c0 : SUCCESS in 43m 50s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/e915a7c29d5f49c6b35962c2851425d6 : FAILURE in 15m 36s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0b2da3b51bdd46b6bb0ef5d9a6b711c4 : FAILURE in 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0768dbe8d42e932a99681bf7a590490ed85128ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-18 16:51:58.000000000","message":"Patch Set 6: Code-Review-1\n\nWhat about pinning the py2 job to the last py2 version of the octavia-tempest-plugin?","accounts_in_message":[],"_revision_number":6},{"id":"db27b55144a6602ccc4c8cc7042235ece8ff8a11","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 17:20:52.000000000","message":"Patch Set 6: Workflow-1\n\nMichael - let me see if I can fix this with a tweak the the devstack plugin code in tempest.","accounts_in_message":[],"_revision_number":6},{"id":"edc891432d628147f892fb4d968a7146e3f8ffdb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 17:29:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ffcecee34df5c55ae6606c4c43a784890288e123","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 17:29:59.000000000","message":"Patch Set 7: Workflow-1\n\n-W until we see the noop-py2-api job pass","accounts_in_message":[],"_revision_number":7},{"id":"21fcad345f02b29f8731695f17b3534136325a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 17:41:15.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"9dd75b29267fc464f31d14178ac624afdf93f3c0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 17:41:30.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"52bedae11e618323347e58d8f8267e3e50f3338a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 19:45:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1df62bd76ebc4f8bb6c8ded0af420a4c : SUCCESS in 9m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73845608b8f14f759edf634699e397bb : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a5b0f6881447eab6bd01afeaac35d2 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cfb2d18586843d49ceaf7b30fa7a805 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35792103af7c425e9bd9355e0feff34b : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e289368226cd4e3d99ad1743c1242e7b : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e2e74be757492b873537947abdd2a7 : SUCCESS in 10m 34s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6505480343df4163a16e853d4f4fac7e : SUCCESS in 5m 17s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/7f28cacf54b241eaa44d4b5e28af9a0a : SUCCESS in 7m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2ba03362d3ce4873afd797685431ef77 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da4b8a01f7d84cae9a752d8006e757d2 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d7597394a1e14ccb8344483803e433ec : SUCCESS in 1h 30m 29s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/1877346b36c040859192b89ffecf9c06 : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/247264a246de488db8d87bc1b7dfb4b8 : SUCCESS in 1h 23m 23s\n- octavia-v2-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/299b7ba82280423cbfc8783b8afb2aee : SUCCESS in 1h 26m 15s (non-voting)\n- octavia-v2-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/0155c7aa298f47f08c9c1e9493ce75eb : FAILURE in 17m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3cd2933333164b1caf34084dc7b9f8fc : FAILURE in 31m 29s (non-voting)\n- octavia-v2-dsvm-py2-scenario-two-node https://zuul.opendev.org/t/openstack/build/c351eafc80184dcc81f3b728ac9dd881 : FAILURE in 49m 43s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e107ddb378a349fea14975ba856289af : SUCCESS in 1h 36m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/668a4ff1c1bc484fa0f388d96cf5862d : SUCCESS in 42m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/314496f05f1f4dd8a938c31ba77efdff : FAILURE in 1h 57m 14s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-scenario https://zuul.opendev.org/t/openstack/build/de277284bd204dc2b7d693432661c366 : SUCCESS in 37m 15s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario https://zuul.opendev.org/t/openstack/build/17b191e3de4d4cb9a599878a2d743382 : SUCCESS in 42m 09s (non-voting)\n- octavia-v2-act-stdby-iptables-dsvm-py2-scenario-centos-7 https://zuul.opendev.org/t/openstack/build/34709d41b59b48ed8f48b3155fa9bb17 : FAILURE in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0e69351df957fb249aaae53441cfd0c978e61313","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-18 20:28:00.000000000","message":"Patch Set 7: -Workflow\n\nLooks like it worked, just need to get a +zuul on the tempest change","accounts_in_message":[],"_revision_number":7},{"id":"28b2322a6f47dc40b8cb9d4919b06a3bb46ad959","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-19 08:35:53.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2d4ca3a204e7d339d4b6a4b019635b6238fdbca0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-12-01 17:10:19.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"b0aaf202b674887598eceabea8b0f8bee7c7558b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 17:10:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"95ceb0a69fbf695ccd72760166960020d10adef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:14:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de58904d85174ee49bf3915d92ce9acf : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62366de4f44b4e6987cff272ec00d4b0 : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e68a490f96534e3289762156ae6ad398 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2e60167099047379f4120ba7c23f904 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69210453b5a64e329e8d1ed3e6637cc6 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/803e204c9bc54a0fbc549d3800d522df : SUCCESS in 12m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2f6d9fbb75b54347a4f1a916f9b748ae : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c755a819bd7343ad9d7bd4ec02f501c2 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b3e81d93abf472b83f836844f38b515 : SUCCESS in 1h 04m 23s\n- octavia-v2-dsvm-noop-py2-api https://zuul.opendev.org/t/openstack/build/24d58017b3f84d51a898b44d1b87491b : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8ea30f7745de48969e43b62fb5f83549 : SUCCESS in 1h 57m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66b8bc70ca9a49de84fb728887638c0c : SUCCESS in 1h 39m 32s","accounts_in_message":[],"_revision_number":7},{"id":"54638c0a624bac50833ef6ddaecb8786ca6714dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:14:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"b0521332f8596045b48b2cc4f453be6118616492","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 19:17:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b8428d7f90460c9e236cd0b5d1a845 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe","revisions":{"a84d57e1b5f0ce2a7e07f0aad49f6b5b3ac23547":{"kind":"REWORK","_number":1,"created":"2020-11-13 15:53:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/762677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/1"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nThis is a hack, but have to start somewhere\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a84d57e1b5f0ce2a7e07f0aad49f6b5b3ac23547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a84d57e1b5f0ce2a7e07f0aad49f6b5b3ac23547"}]},"branch":"refs/heads/stable/train"},"88ddd15dd8e48d041d26ab1d0f904098c4bff959":{"kind":"REWORK","_number":2,"created":"2020-11-13 17:48:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/762677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/2"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 17:48:43.000000000","tz":-300},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nThis is a hack, but have to start somewhere\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88ddd15dd8e48d041d26ab1d0f904098c4bff959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88ddd15dd8e48d041d26ab1d0f904098c4bff959"}]},"branch":"refs/heads/stable/train"},"d2507daa81f21c5cb696dd8df101b71396c60357":{"kind":"REWORK","_number":3,"created":"2020-11-13 19:43:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/762677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/3"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-13 19:43:36.000000000","tz":0},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nThis is a hack, but have to start somewhere\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2507daa81f21c5cb696dd8df101b71396c60357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d2507daa81f21c5cb696dd8df101b71396c60357"}]},"branch":"refs/heads/stable/train"},"e1a96ff1fe4c97a41cc538d18f6dcab5b176a5ed":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-13 21:59:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/762677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/4"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-11-13 21:59:03.000000000","tz":0},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nThis is a hack, but have to start somewhere\n\nDepends-On: https://review.opendev.org/762716\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1a96ff1fe4c97a41cc538d18f6dcab5b176a5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1a96ff1fe4c97a41cc538d18f6dcab5b176a5ed"}]},"branch":"refs/heads/stable/train"},"0357abbdf4f31bc6111c02de00a642cbb85aa91c":{"kind":"REWORK","_number":5,"created":"2020-11-14 22:25:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/762677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/5"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-14 22:24:42.000000000","tz":-300},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nRequired to get the test server binary installed.\n\nDepends-On: https://review.opendev.org/762716\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0357abbdf4f31bc6111c02de00a642cbb85aa91c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0357abbdf4f31bc6111c02de00a642cbb85aa91c"}]},"branch":"refs/heads/stable/train"},"b01602b816e21fe9d418a248d0540fd58390c0e2":{"kind":"REWORK","_number":6,"created":"2020-11-17 19:39:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/762677/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/6"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-17 19:38:02.000000000","tz":-300},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nRequired to get the test server binary installed.\n\nHad to change octavia-v2-dsvm-noop-py2-api to non-voting\nas py2 jobs will no longer run with tempest.\n\nDepends-On: https://review.opendev.org/762716\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b01602b816e21fe9d418a248d0540fd58390c0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b01602b816e21fe9d418a248d0540fd58390c0e2"}]},"branch":"refs/heads/stable/train"},"811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe":{"kind":"REWORK","_number":7,"created":"2020-11-18 17:29:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/77/762677/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/762677/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/762677/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/762677/7"}}},"commit":{"parents":[{"commit":"874ca308955263c37dda9857dcd9f16995334e90","subject":"Merge \"Remove haproxy_check_script for UDP-only LBs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ca308955263c37dda9857dcd9f16995334e90"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-13 15:52:37.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-18 17:22:44.000000000","tz":-300},"subject":"Enable octavia-tempest-plugin for grenade job","message":"Enable octavia-tempest-plugin for grenade job\n\nRequired to get the test server binary installed.\n\nDepends-On: https://review.opendev.org/763228\n\nChange-Id: I254557083715328fa02cf99f03e8c83037b4d640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/811a6c83e3f9c0f23838a09ab4cf9c8ab8574abe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
