)]}'
{"id":"openstack%2Foctavia~762906","triplet_id":"openstack%2Foctavia~stable%2Fvictoria~I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","project":"openstack/octavia","branch":"stable/victoria","topic":"fix-gate-stable/victoria","hashtags":[],"change_id":"I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44","subject":"Fix pep8 failures in gate","status":"MERGED","created":"2020-11-16 18:54:48.000000000","updated":"2020-11-17 09:38:54.000000000","submitted":"2020-11-17 09:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762906-1605605627064-96dc3111","meta_rev_id":"f3ea11e57bfcdf8b0ed9645366ac6408b1bd2468","_number":762906,"virtual_id_number":762906,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"date":"2020-11-16 21:36:08.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-17 09:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":2,"date":"2020-11-17 07:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-11-16 21:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"all":[{"value":1,"date":"2020-11-17 07:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 21:25:14.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-11-17 07:27:32.000000000","updated_by":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"reviewer":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"state":"REVIEWER"},{"updated":"2020-11-17 09:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a97a19332d8522a427c32817a1aa910827a07bba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 18:54:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b67e5fb90a35c90ab6fe4ac552ee0a4df94a8357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 21:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/690a73ab829a45dca3a00364da9cc29b : SUCCESS in 9m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3c05e35714e41f9a42100113b23b489 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4412181d6743a8bc6376f5e275cd34 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66441f2c013c4955b4b783e9fe1204c1 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a16d708eeb994484969647326b5d4485 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be493fa3f7114419bd7b8dce269ae73f : SUCCESS in 11m 19s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/6ce7220b4f734152bf07b52450fddb41 : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/afe81b9e7a724971a97edb932f062da0 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b009f205a0c14387b78689f30944fb2c : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dbdf88a257324d359cbb5c0524ac58a6 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/17db0cdcca104c10b8c820d8baffe896 : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5d7c808d264148719371c2705baf06b7 : SUCCESS in 1h 44m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dec2a6c5491d4a9eb43bdbc618c4e91a : SUCCESS in 40m 29s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/42dfc378283248f39942a51d2a1bfb53 : SUCCESS in 38m 47s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2fbc731791c7498d8ba479add7be2102 : SUCCESS in 38m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c454c520250b4ed9951470e754459c5f : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4363ce6ee4dc42358646583f48c1368a : SUCCESS in 2h 02m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/581da4e81b4247178acb25f2b8cc89f5 : FAILURE in 2h 08m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3141982736a046d1ad9216d40ef4d340 : SUCCESS in 1h 41m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2f784760a76e4f458620aba74bc3f11c : SUCCESS in 2h 00m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6fdd7443fe864351bf21abc021527ce3 : SUCCESS in 1h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/01ccaef923104eac81f39a52c5169aee : SUCCESS in 1h 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92f54f771db3d72519b72d81937dd3881d9c9718","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-11-16 21:25:14.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"75e1dd34f1b416dda87102c47a3aeb9a4331aeb1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-11-16 21:36:08.000000000","message":"Patch Set 1:\n\nLooks like a need to hand cherry-pick this to ussuri too, same failure.","accounts_in_message":[],"_revision_number":1},{"id":"f8baba9818441b4c3945255957e2ef49a79f2606","author":{"_account_id":7249,"name":"Ann Taraday","email":"akamyshnikova@mirantis.com","username":"AKamyshnikova"},"date":"2020-11-17 07:27:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"79b4d5084d1ab813f95c05719188e7216e4edc75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 07:27:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bca15f7db7a7a61d26e5566530133dec66e0b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0aa80dd1bd94c7e95bb98a222b2bab2 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de14de0156f43ccbd261429810f062a : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71409bd62ac6425ba645255e778ec49a : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766f01cc7f0c49098b120cce752e30d0 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d4d4732cb84675807563f0cd1d8517 : SUCCESS in 9m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/797cc7044cf6428dbaf35351208ce23c : SUCCESS in 4m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90ea3968cdba4ee9ba1b744b25f84682 : SUCCESS in 6m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/18437869bf9f403886976a43bfbeac2b : SUCCESS in 1h 55m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3be5046217d6496caac76cc8bb931cd3 : SUCCESS in 1h 54m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/183e1c8c7bc74492834b30597477b1cf : SUCCESS in 39m 38s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/14763b9b981f43ab940f4a616bf990fe : SUCCESS in 37m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2678cb90a1a24606a35bdefcbeec3ce3 : SUCCESS in 42m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/419daa0e18db445783e08393e42219ba : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4d29b59b1fcced39e5f47a7fc2c088d66d0d5163","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:33:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c80c3728ea522f35e989dce0d57be516307b355","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:38:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dc05ae1b7e04e5ab34b43f19208cac4 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc058c3a8367c369ab623083c34cb423f808c52e","revisions":{"fc058c3a8367c369ab623083c34cb423f808c52e":{"kind":"REWORK","_number":1,"created":"2020-11-16 18:54:48.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/06/762906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/762906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/762906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/762906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/762906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/762906/1"}}},"commit":{"parents":[{"commit":"b28743c22ac166e1920697837dad23acfad4effe","subject":"Merge \"Remove unnecessary joinedload\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b28743c22ac166e1920697837dad23acfad4effe"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-11-12 18:58:58.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2020-11-16 18:54:48.000000000","tz":0},"subject":"Fix pep8 failures in gate","message":"Fix pep8 failures in gate\n\nSome package must have moved forward recently, causing\nthe pep8 job to start failing.  Fix the issues since\nthey\u0027re all valid.\n\nAlso changed other messages using an exception string\nto use str(e) to avoid any possible py3 issues.\n\nFixes the grenade job to properly build the test server.\n\nChange-Id: I5c98c6d41cc4bce907e09dcb20ede3a9e7268b44\n(cherry picked from commit 57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc058c3a8367c369ab623083c34cb423f808c52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc058c3a8367c369ab623083c34cb423f808c52e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
