)]}'
{"id":"openstack%2Foctavia~762915","triplet_id":"openstack%2Foctavia~master~I2e86378e20190e0ba987a255f18d675895197541","project":"openstack/octavia","branch":"master","topic":"sysinit-removal","hashtags":[],"change_id":"I2e86378e20190e0ba987a255f18d675895197541","subject":"Remove upstart and sysvinit support","status":"ABANDONED","created":"2020-11-16 20:27:38.000000000","updated":"2024-10-24 07:14:59.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"483f88ed8bd799fd6665f56ddfb5486f47f212bf","_number":762915,"virtual_id_number":762915,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-15 09:55:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-03-15 08:18:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2021-03-18 22:07:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2021-03-15 08:18:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 13:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 01:37:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-01-04 09:32:18.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"b2251b863a393c26f39f2ad656acae2df151c95e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-16 20:27:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2263fe9855db6aedf253527bfc0840a23daf19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 22:51:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4860afaab714a4ea324a7a7a54b91df : SUCCESS in 9m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41f4f8a22fed4d74bba2a1ea31dee97a : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f909b7680fb5418f93991fb3c67f9aee : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9995e6c0e28b49f0a5e8a014dfa83e20 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ead31947f3487a912055d3a9ef1398 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23f2b5ff6a444355addb54db52e23e73 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67da339727d547b1828125548efae2fd : SUCCESS in 16m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7474ded529349299cc94ef60ac13887 : SUCCESS in 6m 22s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d12c8096c76e43d5b1a618befb362f31 : SUCCESS in 6m 52s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/32273e95797d46f998c184d7e79c77f5 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b1ab0e3af974747935bad37f0f04bc5 : FAILURE in 4m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7ee84a75ec3442e81a0c27e49eb6afa : SUCCESS in 8m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24a8bd3c595245a2bf6d90f8e83aebb2 : SUCCESS in 1h 23m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/154e60aa4f744286874f229c904b3e45 : SUCCESS in 1h 58m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2fc2c07322c4f68aecf03841e8f168c : SUCCESS in 49m 22s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/7c61037e833b48eb8390aea54f23a494 : SUCCESS in 45m 36s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fcd7c339da7a4040880a29b9b4b8ab4f : SUCCESS in 47m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b93779f1c0d747fd873eefe5bc9b6cd5 : SUCCESS in 1h 05m 23s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c65b9091c0144e2b86c2839c1f54b1c : SUCCESS in 2h 07m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a627a7d2423945679576840fee9c80a3 : FAILURE in 2h 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a952a88ae24b44bd80f764aa13e6d555 : SUCCESS in 2h 03m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9425bcaa19ad410a92aba89c83ea8dba : FAILURE in 2h 01m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d4c600260c7d4155a38d0b252914f7d3 : FAILURE in 2h 05m 45s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/66438a2c93524793969df6bd02d64838 : FAILURE in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6744cb3370257add20866e36b09793803b4a2902","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-17 10:33:42.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"588c558f7c771b36b39aa4b990f00745116cc7bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 13:05:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c31b0dd5a09a481abf5621ef02d2b652 : SUCCESS in 9m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c40d1927763a4b30afa41ab466240f2d : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1338840294cf47959ff87bc1fcde5f96 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c09f4410fb5c45d091fee81a6722854f : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d0cdf0245514cb9bbcb38e7d271fa5d : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35df5c08d7a94e289fb83c2e8e028490 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94286c9902dc49b1ba36a1035e8d04d8 : SUCCESS in 19m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5587784ced34abfa9b6fd41b1827731 : SUCCESS in 5m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3eb99546c7cc4ef8a7fa12b2929f4827 : SUCCESS in 5m 13s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/143c2284f6584ef5a073b01f6b9b556b : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d75124b5f632464ab372623533c0d530 : SUCCESS in 5m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf37fc7dec244fd58509a13a66bdce3e : SUCCESS in 7m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/885fcfd941824da3a8f21bc6350a1061 : SUCCESS in 1h 18m 44s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0a85e44813c641589e49eabcfda1a69c : SUCCESS in 1h 59m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/513bd7574acb4ae9a403423456c364be : SUCCESS in 47m 44s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/9f36ccfa1f504095ae47864335df06fd : SUCCESS in 1h 00m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bb8fc63c576b4fec9e9e9611750e1b99 : SUCCESS in 38m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c426612f4d484692934f0f217a0d808f : SUCCESS in 1h 11m 59s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/10cfca1449f446a0b8a89e232ff88afa : SUCCESS in 1h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e256c10a449f4c1088144ba5bf3240dc : FAILURE in 1h 50m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/edafef4fd3c940c3a4a64329056240c1 : FAILURE in 2h 03m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/2acb760a8a5b4ec09741340e94e5426b : SUCCESS in 1h 46m 21s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7abf6d8c1f644332bfa10c8a57da6ac7 : SUCCESS in 2h 15m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/d3f3710208ef479ebf1cad55663ff44b : FAILURE in 46m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8ffe91a61c16432aace8c7b0478bdac9f8c4876","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-04 09:32:18.000000000","message":"Patch Set 2: Code-Review-1\n\n- doc/source/contributor/guides/dev-quick-start.rst: still recommends using upstart.\n- etc/octavia.conf: option \"use_upstart\" can be deleted (deprecated 4 years ago)\n- octavia/amphorae/backends/agent/agent_jinja_cfg.py: \"use_upstart\" in method \"build_agent_config\" can be deleted\n- octavia/amphorae/backends/agent/templates/amphora_agent_conf.template: ditto, delete \"use_upstart\"\n- octavia/common/config.py: options \"respawn_count\" and \"respawn_interval\" are upstart-specific. Can be deleted. Also, option \"use_upstart\" can be deleted.\n- octavia/amphorae/backends/agent/api_server/templates/keepalived.upstart.j2: file can be deleted\n- octavia/amphorae/backends/agent/api_server/templates/keepalived.sysvinit.j2: file can be deleted\n- octavia/amphorae/backends/agent/api_server/templates/sysvinit.conf.j2: file can be deleted\n- octavia/amphorae/backends/agent/api_server/templates/upstart.conf.j2: file can be deleted","accounts_in_message":[],"_revision_number":2},{"id":"1c4faa856255fd312775c4b1ca4cde9f175a2863","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-01-04 09:33:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"812726d48e27d808cea6106fde6b0636a2d84add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-15 07:20:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f2ab4c429c0e9c5944d412148a39692fc8709742","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-15 07:21:49.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e - doc/source/contributor/guides/dev-quick-start.rst: still recommends using upstart.\n\u003e - etc/octavia.conf: option \"use_upstart\" can be deleted (deprecated 4 years ago)\n\u003e - octavia/amphorae/backends/agent/agent_jinja_cfg.py: \"use_upstart\" in method \"build_agent_config\" can be deleted\n\u003e - octavia/amphorae/backends/agent/templates/amphora_agent_conf.template: ditto, delete \"use_upstart\"\n\u003e - octavia/common/config.py: options \"respawn_count\" and \"respawn_interval\" are upstart-specific. Can be deleted. Also, option \"use_upstart\" can be deleted.\n\u003e - octavia/amphorae/backends/agent/api_server/templates/keepalived.upstart.j2: file can be deleted\n\u003e - octavia/amphorae/backends/agent/api_server/templates/keepalived.sysvinit.j2: file can be deleted\n\u003e - octavia/amphorae/backends/agent/api_server/templates/sysvinit.conf.j2: file can be deleted\n\u003e - octavia/amphorae/backends/agent/api_server/templates/upstart.conf.j2: file can be deleted\n\ndone most of the items in this list.\nI added deprecations for respawn_count and respawn_interval, I don\u0027t know if they can be removed without notice","accounts_in_message":[],"_revision_number":2},{"id":"78c82ac4347781ec5d0bd9910e085ae2ce67e6f4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-15 07:22:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0852f3eef911e5bcd36e07b7177498b208eb3e0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-15 08:18:37.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2\n\n\u003e I added deprecations for respawn_count and respawn_interval, I don\u0027t know if they can be removed without notice\n\nIMO they can be removed because they were specific to upstart which support is being ripped out altogether in this change. My understanding of deprecation is to express intent to remove/stop supporting something in a near future while the setting is still applied when set meanwhile. I think that is not the case here, hence the two configuration options could be removed. This is not a blocker from my side though, hence CR+2.\n\nThank you, Greg!","accounts_in_message":[],"_revision_number":3},{"id":"0adb5f66f3a940d276a1d150c15798d5116490d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:55:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7f10578c544454f9520171bd821ac69 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f794d748c0d4430ab0d25e3e167376ce : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3937ecec7d43485c85309f67b10a9ba3 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24a36d7d2cf341ee9047c379ea7451a4 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c1587f40d3349d79aa5b05f35bc9d6a : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7051ae58e71e45979128e11379662694 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea4d5e77612467ebe656eb4175bb2fd : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a53df72fa3d846cc83d8dfca5995b810 : SUCCESS in 6m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3c7deb057c6c4123952cc83a0a5da748 : SUCCESS in 7m 09s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a3269fd31c254e29bbf13940259af990 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5ae9d59bb1f94b93a5224e5cd46b4b9f : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73474b9c0e194a08a3722512021ce6dd : SUCCESS in 9m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78af10f7c6aa478cae29532fbd82e6a9 : SUCCESS in 1h 27m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32a5d7721e2d45639e19aba889cd82d4 : SUCCESS in 2h 14m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d8240562c634b09bfd94136d27c1787 : SUCCESS in 57m 41s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/2f92cea28264486386912a73b7f2ed0f : SUCCESS in 46m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5755248ecc0348baa0a69fe8fa25237c : SUCCESS in 55m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7d3aef427b74c1aa63cc620d9d1b6d0 : SUCCESS in 1h 05m 42s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8705ce49081a4cfd87723f9fe1aa9a15 : SUCCESS in 2h 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/40c0a6e64d104694bea84b06dcda5dd0 : FAILURE in 56m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2d5627c664534aa79fccaa88e380c73d : SUCCESS in 2h 04m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/86a66bb54cd14dda8b3af944434b37d8 : SUCCESS in 2h 29m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/d3fbcbf5d6a44150a1a28856ccc73b0a : FAILURE in 2h 20m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/b3ca1ed87e184e62aed0541b6c51be98 : FAILURE in 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e3b1b169ef936cff74604726cada17461f9304cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-03-18 22:07:50.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"483f88ed8bd799fd6665f56ddfb5486f47f212bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-24 07:14:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9e281b718b7aa0443ef5057474eee61228a689ea","revisions":{"e19895b386ea4663a138ee82ba797778b37868cf":{"kind":"REWORK","_number":1,"created":"2020-11-16 20:27:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/15/762915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/762915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/762915/1"}}},"commit":{"parents":[{"commit":"182ea3ce5ac1c66c3cb909035bdf6ed518d78ecf","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/182ea3ce5ac1c66c3cb909035bdf6ed518d78ecf"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-16 17:24:33.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-16 20:19:25.000000000","tz":60},"subject":"Remove upstart and sysvinit support","message":"Remove upstart and sysvinit support\n\nStory 2008350\nTask 41262\n\nChange-Id: I2e86378e20190e0ba987a255f18d675895197541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19895b386ea4663a138ee82ba797778b37868cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e19895b386ea4663a138ee82ba797778b37868cf"}]},"branch":"refs/heads/master"},"187b624ed53b12fcb2ce58db69065d1a97166801":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-17 10:33:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/15/762915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/762915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/762915/2"}}},"commit":{"parents":[{"commit":"489b8989cc44b260df061fa42b3a56e3693adf3c","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/489b8989cc44b260df061fa42b3a56e3693adf3c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-16 17:24:33.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-17 10:32:32.000000000","tz":60},"subject":"Remove upstart and sysvinit support","message":"Remove upstart and sysvinit support\n\nStory 2008350\nTask 41262\n\nChange-Id: I2e86378e20190e0ba987a255f18d675895197541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/187b624ed53b12fcb2ce58db69065d1a97166801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/187b624ed53b12fcb2ce58db69065d1a97166801"}]},"branch":"refs/heads/master"},"9e281b718b7aa0443ef5057474eee61228a689ea":{"kind":"REWORK","_number":3,"created":"2021-03-15 07:20:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/15/762915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/15/762915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/15/762915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/15/762915/3"}}},"commit":{"parents":[{"commit":"a9ad9e9bbc6c42c20a670150e14bcfe55d7ede26","subject":"Add generic network interface management in the amphora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a9ad9e9bbc6c42c20a670150e14bcfe55d7ede26"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-16 17:24:33.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-03-15 06:33:48.000000000","tz":60},"subject":"Remove upstart and sysvinit support","message":"Remove upstart and sysvinit support\n\nStory 2008357\nTask 41262\n\nChange-Id: I2e86378e20190e0ba987a255f18d675895197541\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e281b718b7aa0443ef5057474eee61228a689ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e281b718b7aa0443ef5057474eee61228a689ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
