)]}'
{"id":"openstack%2Foctavia~762916","triplet_id":"openstack%2Foctavia~master~I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","project":"openstack/octavia","branch":"master","topic":"redhat-cronie","hashtags":[],"change_id":"I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3","subject":"Fix missing cronie package in RHEL-based image builds","status":"MERGED","created":"2020-11-16 20:40:06.000000000","updated":"2020-11-27 11:29:01.000000000","submitted":"2020-11-27 11:25:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762916-redhat-cronie","meta_rev_id":"26aa01fa84d1cdf9ad501a41bf0655111b648083","_number":762916,"virtual_id_number":762916,"owner":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-23 14:14:24.000000000","_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 11:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":2,"date":"2020-11-27 07:56:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-11-20 09:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":1,"date":"2020-11-27 07:56:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"value":2,"date":"2020-11-27 07:56:10.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-11-20 09:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-20 01:15:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-20 09:04:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2020-11-27 07:56:10.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"7f3f6ff4fccc1e1886d0b6bff41e40d476395b0e","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-16 20:40:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ad9267a34e201f62e1e0099af6da1fc1c649f79","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-16 20:43:19.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5105feda181d0e49168a2629eca9cbb64db5b134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 23:12:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c35fde75c2645e6bf560914b9163f0f : SUCCESS in 8m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e84eeac4520648058a6f6f8a4b5d4121 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171b36bafc26458baeced17b66af6af3 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7300444890d941098bf31a914462e095 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19ddc6d3dda478cb3ceb132874a22bb : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f16001ea72341b2b7cb969b02df75a3 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7add01670982432a801508ffa4aa14ae : SUCCESS in 19m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/89660dc698d549bb868ddbfb84f93198 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/694426bb3e194732aecadf6e523e7e09 : SUCCESS in 7m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2e61d01ac40d4d91af95a27b4559e114 : SUCCESS in 4m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/397ccff9d820464e954f16b08b8c284d : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aad055358a484665a3e4b11c659d1833 : SUCCESS in 1h 16m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a415e56774cd4a519cefa1d5c9b8d0f4 : SUCCESS in 1h 40m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/09b8fefcb54a41baa4ef4ab5db89e8b4 : SUCCESS in 1h 01m 45s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/1b535d93db594d42ad3956eeb4b18049 : SUCCESS in 50m 38s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6187e4f07b94b98b7cdf6d7f6501ae6 : SUCCESS in 55m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b838e6bc5b834b4b9e6767901e34800e : SUCCESS in 1h 22m 45s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/449a2281cdbf4812af917b0f0077afe9 : SUCCESS in 2h 01m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00d6ab7307f94929b17a6ff590c2509e : FAILURE in 2h 25m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bdb2cc02154e4803a5474331720edd11 : SUCCESS in 1h 41m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/905b1d0b5c714ec993c7786521cc3284 : SUCCESS in 1h 56m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e0c09e56f28e4f70bc47bffa63d5e1ca : SUCCESS in 2h 10m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/bcb92340d9384531bdbea7fb80dfb344 : FAILURE in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54c7afe3b4014d469cd4f22394b7951eba8bafa0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-19 14:00:38.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2\n\nmaybe it requires a release note there","accounts_in_message":[],"_revision_number":2},{"id":"3afd154f9bd7553a5ec55f256c2ff97df989e793","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-19 22:50:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e127d8337e048bdf8e90b22251655d9d4368d66","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-19 22:57:23.000000000","message":"Patch Set 3:\n\nFull disclosure: when I built the image last time from the master branch, cronie _was_ included in the final image. It was probably included in the base image because it wasn\u0027t explicitly installed in the log output. But there was a point in time when it wasn\u0027t in the final image so I think it\u0027s still a good idea to explicitly declare it in pkg-map.","accounts_in_message":[],"_revision_number":3},{"id":"d94232a5b13e3d4cdd450dae8781b20371ee9d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-20 01:15:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e442d85523d14e74a2e43b29ca5d2fab : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68941cab63a74f1593d8ac3f2b8aca77 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/512ed4094fb64950ae2570f5c6101e58 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fef8b25d19e491aa61b24f2628c259b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f788a12b6c4b06b19accff86fe7eb1 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2118ba2ce8e1465abc9ca1a514c58ab5 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39f77b93fb345cf8052863b6a676b9f : SUCCESS in 18m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a387d88b7acf49028ad32ee66695de4b : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/12804f1aa125419f870f55f87964d2d9 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8e8d962c847f49a5a1a0a1bbcd2e6ba2 : SUCCESS in 8m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d90387c1c70e4d1eb1631b967624de25 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a122cc3618384aa9b24ee84a8af1000a : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f75d0afff7894790a286112fa87099cf : SUCCESS in 1h 15m 50s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9563e49ae9b445aca776baf9f203452d : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e27b2ef3b84c4db1aaaea2f2015788d6 : SUCCESS in 56m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/f72e840d1aca46dda911674e2289ad27 : SUCCESS in 56m 26s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c461618c0b6340c686f5bc1e3caf9b8e : SUCCESS in 38m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a27b5a630da14c5c87c188c75e5c88a0 : FAILURE in 51m 22s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/12f7acb4d5ba4bbbb40e3eadab7cd1b3 : SUCCESS in 1h 54m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/25eaa6f456434d1c8d715360db0f6609 : FAILURE in 2h 07m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5e4fb5e8e305455f857b370e79a16e05 : SUCCESS in 1h 37m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e5ce14cdcea642fe9790dd23aead9c25 : SUCCESS in 2h 16m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c62297aa1a9846a584e8d49934642ed9 : FAILURE in 2h 01m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/c1c2945896a54943a42695951fd0cd86 : FAILURE in 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7072d52d9522db0f45db28e76946efffe2c0d0d0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-20 09:04:55.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2\n\n\u003e Full disclosure: when I built the image last time from the master\n \u003e branch, cronie _was_ included in the final image. It was probably\n \u003e included in the base image because it wasn\u0027t explicitly installed\n \u003e in the log output. But there was a point in time when it wasn\u0027t in\n \u003e the final image so I think it\u0027s still a good idea to explicitly\n \u003e declare it in pkg-map.\n\nUsing master branch, I don\u0027t have cronie in my Centos amphorae (different dib releases?).\nSo adding it explicitly looks better and it can be backported on stable branches.","accounts_in_message":[],"_revision_number":3},{"id":"89d3ab4ca565c0643d5cfd4d6a277c750e7a7067","author":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"date":"2020-11-23 14:14:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"26e68478329e7271be297942ecf3012ee939d381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 16:51:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da6d357006634585ab93ce2dbabf7c9d : SUCCESS in 8m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c9cea7cbdef4d14a3f9e71d760120f2 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c4f3732da74f6da645f6b71bdabac5 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a775eae59634f24a115b3321c4f9b7d : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/292a7e4b327b4367876b7e02c2e87b8a : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e30444dcdf341de95c9b0edc31d3389 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda6b90ebab1453bb81bca199a57dcf7 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6cc4f9547e34d5cbd7c80879747bdcd : SUCCESS in 4m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2cf44fd2ac2840a19325484ce175785c : SUCCESS in 5m 23s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ceb4b1629cd44300b77e3d4188cf112e : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/03fcec8b7d63483b8d1401bdf575bdd6 : SUCCESS in 5m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb751919eb35476c8d14e3ee029677e2 : SUCCESS in 8m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c49fb4a494db4014ba6da455e0bf4f05 : SUCCESS in 1h 40m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/88048ab380c24d5192ae0ea9235f46bc : SUCCESS in 2h 19m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/51bbbdb9f7c44cea8799592af731528f : SUCCESS in 43m 05s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/4d4188d4daf24cd1812653e27e05fc8b : SUCCESS in 37m 04s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3415e01767e9479492d531c14a0d1172 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f98ba83fff3a48db8390fc33d2b33295 : SUCCESS in 1h 01m 12s\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f71359c7af2a49e8b98d8a7ac8f5ab37 : SUCCESS in 2h 10m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6825c285cb546e9a0588b9f6b43c3aa : FAILURE in 2h 09m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e7a4361f7cc34161a025ced4252ed655 : SUCCESS in 2h 01m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/05507aa41e294c2b86e127d0525c1f15 : SUCCESS in 1h 40m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/baaed10178e143cc8f52d774d2c2649b : SUCCESS in 1h 46m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/fd51fe5fa891456dad52c4ea488cc6c9 : FAILURE in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"29cfef9785a4dc55df52f1a2f84311d5b95d41f7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-11-27 07:56:10.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e9661ac02c8149f1bf9e9273c2fb6e573166011f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 07:56:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7d002e5947ad19f2c51ae68944c34f00cfcb55f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:25:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8de6a70f90544d4bd0502c326999f77 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a221b7a91d47d4ac22b1ca9e0d4fce : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5bb327f7d004ecdb6fdd1097b5c7798 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae51a9df2f343c9a71966ad578374c6 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab2088789ef4ed4887c269caefd85b8 : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1cc637ffcc449b68248100511ede7f3 : SUCCESS in 5m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c33be034027d4e4dad5f2c17699a1e27 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98d3bcc5907344e2aa13a14626ad437a : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/53cc9a3fc63941ecaa0cf764b419a4bc : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a3b0196ab3f438db3f4f969abcf5070 : SUCCESS in 2h 18m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18bc7c1d2c274c55a36667691227e1ee : SUCCESS in 53m 50s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/961582a1d6514de48d9bc66d47f31549 : SUCCESS in 40m 03s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/44fcbbc32502437bb965bf4ec2575710 : SUCCESS in 40m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e0833a2c84fe48cbbc32a72426421e1e : SUCCESS in 53m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0232989cb567fa1c86953c1118ba1fb0023ed5e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:25:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"26aa01fa84d1cdf9ad501a41bf0655111b648083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 11:29:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f9f45f37f8d4bc6806097346a5aa117 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb8a8320c81b48c7aa87d3d1d5d940c8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6aa23358eb2f4df412e010fd1fbb9e35ee2e1522","revisions":{"aa32e6cad2f8f8cd669990df9fdb15465bb02a69":{"kind":"REWORK","_number":1,"created":"2020-11-16 20:40:06.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/16/762916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/762916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/762916/1"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 03:11:41.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 20:36:42.000000000","tz":-300},"subject":"Fix missing cronie package in RHEL-based image builds","message":"Fix missing cronie package in RHEL-based image builds\n\nIn RHEL-based distros, cron is provided by the cronie package. This package is\nmissing from the amphora built images.\n\nThis patch maps the cron package to cronie for distros in the Red Hat family.\n\nChange-Id: I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3\nStory: 2008351\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa32e6cad2f8f8cd669990df9fdb15465bb02a69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa32e6cad2f8f8cd669990df9fdb15465bb02a69"}]},"branch":"refs/heads/master"},"54c8c117b5433c80c06fc20d8f694384ffc801d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-16 20:43:19.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/16/762916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/762916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/762916/2"}}},"commit":{"parents":[{"commit":"57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd","subject":"Fix pep8 failures in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57e7bd17bfa37c4972ae07adcc4e1b9a5535bbcd"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 03:11:41.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 20:42:53.000000000","tz":-300},"subject":"Fix missing cronie package in RHEL-based image builds","message":"Fix missing cronie package in RHEL-based image builds\n\nIn RHEL-based distros, cron is provided by the cronie package. This package is\nmissing from the amphora built images.\n\nThis patch maps the cron package to cronie for distros in the Red Hat family.\n\nChange-Id: I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3\nStory: 2008351\nTask: 41253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c8c117b5433c80c06fc20d8f694384ffc801d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c8c117b5433c80c06fc20d8f694384ffc801d2"}]},"branch":"refs/heads/master"},"6aa23358eb2f4df412e010fd1fbb9e35ee2e1522":{"kind":"REWORK","_number":3,"created":"2020-11-19 22:50:58.000000000","uploader":{"_account_id":27330,"name":"Martin Chlumsky","email":"martin.chlumsky@gmail.com","username":"mchlumsky"},"ref":"refs/changes/16/762916/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/16/762916/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/16/762916/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/16/762916/3"}}},"commit":{"parents":[{"commit":"f96b0986cda925d471073620047346c70b73927a","subject":"Merge \"Replace assertItemsEqual with assertCountEqual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f96b0986cda925d471073620047346c70b73927a"}]}],"author":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-16 03:11:41.000000000","tz":-300},"committer":{"name":"Martin Chlumsky","email":"martin.chlumsky@ubisoft.com","date":"2020-11-19 22:48:50.000000000","tz":-300},"subject":"Fix missing cronie package in RHEL-based image builds","message":"Fix missing cronie package in RHEL-based image builds\n\nIn RHEL-based distros, cron is provided by the cronie package. This package is\nmissing from the amphora built images.\n\nThis patch maps the cron package to cronie for distros in the Red Hat family.\n\nChange-Id: I6fb7889de90e8e5354c1e6ce2c95ec0199ce90e3\nStory: 2008351\nTask: 41253\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6aa23358eb2f4df412e010fd1fbb9e35ee2e1522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6aa23358eb2f4df412e010fd1fbb9e35ee2e1522"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
